)]}'
{"id":"openstack%2Fkolla~724117","triplet_id":"openstack%2Fkolla~stable%2Frocky~I6200ed527244313107666753257f52d4b9305dd3","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1871600","hashtags":[],"change_id":"I6200ed527244313107666753257f52d4b9305dd3","subject":"Fix multiple issues","status":"MERGED","created":"2020-04-28 18:29:07.000000000","updated":"2020-04-29 16:06:13.000000000","submitted":"2020-04-29 16:06:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724117-1588176373314-0ef6f594","meta_rev_id":"181a81757e08cf97103e63124d5516f218bbbdda","_number":724117,"virtual_id_number":724117,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 16:06:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-04-29 15:06:10.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-29 15:07:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-04-29 15:07:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 15:06:10.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-29 15:07:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-29 16:06:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9d91b61fc268a447d894133b0dbacb3aef3cd91","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-28 18:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0294017bcaddbc257f1084ea35ab417b9b611e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 20:39:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/df4c31362bbe41829f0aeb0114a13b54 : SUCCESS in 48m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/df5045ce97b94d14a59b0d95d7d3f93b : FAILURE in 47m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6af1247e149949759d72a163fc5a26fb : SUCCESS in 1h 13m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/af2653e3bfd146cebcc6f82dd7ddc705 : SUCCESS in 1h 06m 21s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/8947213e1e974e64a8cf9f2fe6378785 : SUCCESS in 1h 03m 26s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/86489a18797b45f4b80d0aff815f3396 : FAILURE in 44m 37s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/948731f1ed0b432b9d36c04e9be36eb5 : SUCCESS in 1h 17m 21s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/58aa44d98f284851b36b9b53cae56372 : SUCCESS in 1h 07m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35a9e909130f4542a75e84229a15e7ac : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8165fd7adb436d961cebe44b703853 : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a298bb75fb44bad981bcf52adf3f610 : SUCCESS in 7m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/02dc39ccca754c6dab8d92b117c696fd : SUCCESS in 7m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e0c97b89792a47d6b235eda3a4d72dc0 : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71d448ff9ad54b9caaa7a733be46eaca : FAILURE in 2h 06m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e03696b2288545a8ac073497a67b792c : FAILURE in 2h 07m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d3f6e94fb964ee1af5b8a85ef873736 : FAILURE in 29m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cdff16b93d70466ab5f1d82d13a6c796 : FAILURE in 1h 00m 20s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"563d8f29833b70f999ea1594242488410b3c8d0e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 21:46:41.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c36d47de8afe36d83611685490d39956d14f520a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 23:39:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/053fbcac9f0144e0bbd02a2e04020029 : SUCCESS in 39m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5fd9f791dbc9420fa2aba6ce246506b8 : FAILURE in 47m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/517aebefac644a7d87804204809552de : SUCCESS in 1h 17m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/66c76cc2c0d649ad8e75f546d5cc476f : SUCCESS in 1h 14m 39s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/5ec12207fc9e4aae8cf1ed713e27e7fb : SUCCESS in 1h 08m 23s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c5ac21119eb54dd4baddd74d5a9b2321 : SUCCESS in 1h 26m 52s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a9e09f5d16a44bb5a4e1f009d8034dff : SUCCESS in 1h 08m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8990bc9cc99a4e33a2d6ae15cf708f83 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef66c52f97f14b8892bfdc938c03c330 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcb061bf50f845b9849d78747fadcaf9 : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/886250f4a7f24475bb87f09422f6b462 : SUCCESS in 5m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/81a4b5ed2ffc4ccf9e6b80c88bf203d2 : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cde7ffc4708d4e409419a022c1206bb1 : FAILURE in 1h 15m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/421a6f98b164430f9ab339651c0f4366 : FAILURE in 36m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36280b4230804542a9807ca850a2d3e7 : FAILURE in 29m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/690993e9cc4f42669c13d43e72c731eb : FAILURE in 1h 31m 52s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"d3e0f0909eececd298768bce0f79ab81b33af8b5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-29 08:16:59.000000000","message":"Patch Set 1:\n\nhttps://github.com/cf-natali/ntplib/issues/1","accounts_in_message":[],"_revision_number":1},{"id":"4f2284ef0e752cdc74b1fc3a58b128b8d649e14d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-29 08:41:22.000000000","message":"Patch Set 1:\n\nrecheck\n\nntplib for py2 uploaded by author","accounts_in_message":[],"_revision_number":1},{"id":"1c6522b0b4f272e25c86a2e8f37bc4419631f05a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 10:18:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/539b4804c0584ebab8e1ba4723065b07 : SUCCESS in 56m 48s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bb208b59e1384c9ea9af7c0f5847b22d : SUCCESS in 48m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/293a1a4c0ca441b6b059a290f6328b9c : SUCCESS in 1h 12m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0ad9fca9ef004fd59c1b5ca2ea914191 : SUCCESS in 1h 35m 29s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a63f108ecacd4d2580b92a5a7808f108 : SUCCESS in 1h 08m 08s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/25db3469531d4310af01d4641b5c59e1 : SUCCESS in 50m 13s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e762fce142824055a84f5d0f712d8cdd : SUCCESS in 1h 16m 56s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4662191c14a74deb9afe2220dd10a910 : SUCCESS in 1h 09m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30f37d64731d4b84916e3d8778316d8e : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa92774639a442e926aa28200b095b0 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bbc9cad9990d4a99bbcb51e014e318ad : SUCCESS in 6m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8570909bf90348eb8748310c87d5d296 : SUCCESS in 4m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/67e12739834c4bc5b61c1f31f42ac06b : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3bb77a9e06b14ebdbbf60c89d8cc207d : FAILURE in 1h 12m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d45d1305049948de81e7418bc53060bb : FAILURE in 28m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc1724e04b47471194d864a9bb8a8df7 : FAILURE in 35m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3283f778889f45ca9c3d939237f9ec46 : FAILURE in 1h 17m 26s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"1bef7e99b43255ba91151a389c460eb82a0deb22","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-29 11:30:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a2de590c70d3969eb7e08fc2b2eac621089d6c33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:31:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c2957234f2a44943b616cd582fafd533 : SUCCESS in 38m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6a477d67e56749bd8c3e3102d44d9289 : SUCCESS in 47m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/60d9b30622d54d8c8925845cd606e876 : SUCCESS in 1h 14m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/571ff44ac0ee433b89bbeea791c3a321 : SUCCESS in 1h 08m 50s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/dc450425ab8e4ae8bd1a7ed1eabc479d : SUCCESS in 47m 25s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5be9f6055d9b42d7a4a1ae4c834140d0 : SUCCESS in 57m 10s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/9a48ff9660844eda86d36c30a03273f1 : SUCCESS in 1h 23m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b53b3132611c48b1a08c97ce0b0d4ed9 : SUCCESS in 1h 06m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/204a5e3744da4a7c91033ad13b3fc174 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4060c2be379f49e2ad5aaae9b3954738 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e5f29f3ecc34830a7e13d2aa8386e4f : SUCCESS in 5m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/67e5e92c75c54701877814d9d55740c5 : SUCCESS in 4m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a6c25f8b069241a594c0256b418d8fce : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99001b563cf44bda87d5ff1a4a44e372 : SUCCESS in 1h 26m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/166795d7931f4caa8ffcab7e67bae340 : SUCCESS in 28m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a5f6d42ce2d74f46b25995153650554d : SUCCESS in 57m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0dd267599f6a40148bf100540e68c31a : SUCCESS in 1h 56m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"8c604cf41afe0a79da38108d1bf9c91c616ce782","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-29 15:06:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"99c044b6e5a1fe973219865e82e64384d655c1b3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-29 15:07:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ed13e3649e35ab6ec94da2ee6bb1cd741b311c89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:08:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2112f374163d0fec6974d6b7684185657d09edb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:06:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ad287628cdc4210ca5bf5b34edbe32ad909bf2ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:06:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/02d8afcdcd0a43ed89a6427dd7250555 : SUCCESS in 52m 44s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d50cb986cb0e4472b96b02e4dfab7724 : SUCCESS in 53m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dc8234befc247598d2abee3a776b57b : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c960ff341824fe4810b2e712026dc99 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15f3b5dc24f44b28b6467bbf11e30920 : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c99a261c5a0e445e8c8714c439d41997 : SUCCESS in 5m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/68ff330d506c47cfa5c486703b497238 : SUCCESS in 4m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c69bd82dcb6e489084847c3c8e88e5d8 : SUCCESS in 31m 57s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"03212577ef09c5d1a613b6277c5d48d6ca4c03c0","revisions":{"938f63b33f44dab4b87968a71de3b30c76a4ca11":{"kind":"REWORK","_number":1,"created":"2020-04-28 18:29:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/17/724117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/724117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/724117/1"}}},"commit":{"parents":[{"commit":"c8774a560c07baabce5cd7288e927ba075e18f1c","subject":"Make deploy jobs voting again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8774a560c07baabce5cd7288e927ba075e18f1c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-28 12:53:25.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-28 18:27:05.000000000","tz":60},"subject":"Fix multiple issues","message":"Fix multiple issues\n\nMultiple issues combined to pass CI\n\n1. drop tripleo-common-tempest-plugin\n\nThis is not needed plugin without any branches or releases. And wants Python 3\nwhile we have Py2 for CentOS 7.\n\nOriginal Change-Id: I9cab91b82941ea125b16d19e68722ce41a1f11f2\n14:41 \u003c hrw\u003e is  tripleo-common-tempest-plugin your stuff?\n14:43 \u003c@EmilienM\u003e yes and it does nothing. I thought we removed that?\n14:43 \u003c@EmilienM\u003e apparently it\u0027s still here https://github.com/openstack/tripleo-common-tempest-plugin\n14:43 \u003c@EmilienM\u003e we should burn it, weshay|ruck ^\n14:44 \u003c hrw\u003e EmilienM: also fine to drop from train?\n14:44 \u003c@EmilienM\u003e burn it everywhere you can\n14:44 \u003c hrw\u003e THANKS\n14:44 \u003c@EmilienM\u003e it has never been useful\n14:44 \u003c@EmilienM\u003e and people who started it left\n14:48 \u003c@weshay|ruck\u003e EmilienM, yup\n\n2. Sensu: Remove ceph and process-checks plugins from client\n\nThe sensu image became unbuildable, failing to install the ceph plugin:\n\n    Unable to resolve dependencies: sensu-plugins-ceph requires english (\u003d 0.6.3)\n\nThis has happened because the author of the \u0027english\u0027 gem has seen fit\nto publish a new 0.7.0 release, and \u0027yank\u0027 [1] all previous releases\nfrom rubygems.\n\nThis change removes the ceph and process-checks plugins, which both\nsuffer from a dependency on the yanked english 0.6.3.\n\n[1] https://rubygems.org/gems/english/versions/0.6.3\n\nOriginal Change-Id: I9d34413fa10b5a2920aca60b08141bfaddcc7415\nCloses-Bug: #1871600\n(cherry picked from commit 818dc4d6ff12a3119be072e0c68e2d2c8da9b799)\n(cherry picked from commit 2d1017be3fd658011fc5a86bd38b7265da306ea3)\n\n3. Use freezer queens releases\n\nThere are no freezer releases in Rocky, so these packages stayed on\nmaster. This is now python 3 only, so fails to build. Pin to Queens\n6.0.0 releases.\n\nChange-Id: I6200ed527244313107666753257f52d4b9305dd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/938f63b33f44dab4b87968a71de3b30c76a4ca11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/938f63b33f44dab4b87968a71de3b30c76a4ca11"}]},"branch":"refs/heads/stable/rocky"},"03212577ef09c5d1a613b6277c5d48d6ca4c03c0":{"kind":"REWORK","_number":2,"created":"2020-04-29 11:30:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/17/724117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/724117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/724117/2"}}},"commit":{"parents":[{"commit":"c8774a560c07baabce5cd7288e927ba075e18f1c","subject":"Make deploy jobs voting again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8774a560c07baabce5cd7288e927ba075e18f1c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-28 12:53:25.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-29 11:24:40.000000000","tz":60},"subject":"Fix multiple issues","message":"Fix multiple issues\n\nMultiple issues combined to pass CI\n\n1. drop tripleo-common-tempest-plugin\n\nThis is not needed plugin without any branches or releases. And wants Python 3\nwhile we have Py2 for CentOS 7.\n\nOriginal Change-Id: I9cab91b82941ea125b16d19e68722ce41a1f11f2\n14:41 \u003c hrw\u003e is  tripleo-common-tempest-plugin your stuff?\n14:43 \u003c@EmilienM\u003e yes and it does nothing. I thought we removed that?\n14:43 \u003c@EmilienM\u003e apparently it\u0027s still here https://github.com/openstack/tripleo-common-tempest-plugin\n14:43 \u003c@EmilienM\u003e we should burn it, weshay|ruck ^\n14:44 \u003c hrw\u003e EmilienM: also fine to drop from train?\n14:44 \u003c@EmilienM\u003e burn it everywhere you can\n14:44 \u003c hrw\u003e THANKS\n14:44 \u003c@EmilienM\u003e it has never been useful\n14:44 \u003c@EmilienM\u003e and people who started it left\n14:48 \u003c@weshay|ruck\u003e EmilienM, yup\n\n2. Sensu: Remove ceph and process-checks plugins from client\n\nThe sensu image became unbuildable, failing to install the ceph plugin:\n\n    Unable to resolve dependencies: sensu-plugins-ceph requires english (\u003d 0.6.3)\n\nThis has happened because the author of the \u0027english\u0027 gem has seen fit\nto publish a new 0.7.0 release, and \u0027yank\u0027 [1] all previous releases\nfrom rubygems.\n\nThis change removes the ceph and process-checks plugins, which both\nsuffer from a dependency on the yanked english 0.6.3.\n\n[1] https://rubygems.org/gems/english/versions/0.6.3\n\nOriginal Change-Id: I9d34413fa10b5a2920aca60b08141bfaddcc7415\nCloses-Bug: #1871600\n(cherry picked from commit 818dc4d6ff12a3119be072e0c68e2d2c8da9b799)\n(cherry picked from commit 2d1017be3fd658011fc5a86bd38b7265da306ea3)\n\n3. Use freezer queens releases\n\nThere are no freezer releases in Rocky, so these packages stayed on\nmaster. This is now python 3 only, so fails to build. Pin to Queens\n6.0.0 releases.\n\n4. Pin elasticsearch gem to \u003c7.6.0 in fluentd\n\nSince the release of the elasticsearch 7.6.0 gem, the fluentd image\nfails to build on Ubuntu due to a transitive dependency on faraday.\n\nERROR:  Error installing fluent-plugin-elasticsearch:\n    faraday requires Ruby version \u003e\u003d 2.3.\n\nPin elasticsearch to \u003c7.6.0.\n\nChange-Id: I6200ed527244313107666753257f52d4b9305dd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03212577ef09c5d1a613b6277c5d48d6ca4c03c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03212577ef09c5d1a613b6277c5d48d6ca4c03c0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
