)]}'
{"id":"openstack%2Fkolla~724617","triplet_id":"openstack%2Fkolla~master~Ic81fb425fa54953f1d6c12063049d60658deb4ed","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ic81fb425fa54953f1d6c12063049d60658deb4ed","subject":"debian: binary target is x86 only","status":"MERGED","created":"2020-04-30 09:12:18.000000000","updated":"2020-05-04 15:31:58.000000000","submitted":"2020-05-04 15:29:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724617-1588606193859-71cafbf4","meta_rev_id":"5ab8a2d3c07334760c010202c4fe490ab1bceaa1","_number":724617,"virtual_id_number":724617,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-05-04 13:58:18.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 15:29:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-04 11:00:33.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-05-01 07:43:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-05-04 14:05:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2020-05-04 14:05:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-05-04 09:47:04.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-05-04 09:47:09.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 09:27:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-04 09:47:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-05-04 09:47:09.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-05-04 11:00:33.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-04 14:05:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-05-04 15:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d16b96dfa89ddbab136bc378fa4c09aab5972e6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-30 09:12:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18883cde16e3aeba33b29523c3254685b7ce2582","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 09:27:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"670f4a6277d3b500c20821de20babae8ca879541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 10:43:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1bae009c96884b10b729a5f4e84fa416 : SUCCESS in 25m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3ffe186f835e4cf0b16940d8cd18b5b7 : SUCCESS in 30m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ab30e443ef0e424f8f27389d6ee90bd4 : SUCCESS in 1h 18m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ef251d4fb260400ab9a369386e3fedf0 : SUCCESS in 1h 15m 20s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/35a81f13ff7d4f26a4c86125bf8a99bc : SUCCESS in 1h 26m 15s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e7ef0e9ac0234bbdb580ef21dff62a92 : SUCCESS in 46m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cd963fd707fc4a12b6de356b2b465e15 : SUCCESS in 23m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8c53fa14b9bb484c8f1f6fadb0232592 : SUCCESS in 21m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4b576adcac694125bc7d5fcdc6d38836 : SUCCESS in 1h 03m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c7ea4c2d2bf4cc5b922832b9aae3404 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39e33cd0fc224379854a1a2420528fac : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ff1967503247ce902dd5310f8ba9b1 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9ec26481b594fdab07e4529239f7ab8 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87efccb10c424b1ebbb850ea8e179888 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f7ef9dfe624c4d8bc6a611b3b74bbf : SUCCESS in 7m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/743fba45063845b295159e5024415b57 : SUCCESS in 23m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/35a46c7a5d944a10940e21953c90c728 : SUCCESS in 23m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/314dd28d044147918992d4f1030ed399 : SUCCESS in 56m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6419073b1d41440d81e76594da1ac613 : SUCCESS in 1h 02m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a735fec5bd9640e7bfbf3bda1790ad82 : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0427868eeadedffda62d22faa9ae870039747b0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 11:22:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a45c9c9b75d7405e9670e3bac13271b2 : FAILURE in 7m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/842dfd5431a042daa2597d7202592ad2 : RETRY_LIMIT in 2m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec05f29bcb8f4633b791138e91170747 : SUCCESS in 1h 50m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e5d1007120074276ac63b50d7e7825f8 : RETRY_LIMIT in 2m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b41860c124c299eb18827f739811189aa2807c60","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-30 11:30:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b73d6deffef841bc8ca50d156f7f2961 : SUCCESS in 1h 05m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a933df2fb3e5d46a1fe8f817b598313c05b1253d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-01 07:43:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ce1c03c31a5fad02d9629a04978707509a9adc15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 09:23:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e290d29053194738827028291f4ce963 : FAILURE in 9m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/51aea4a7dcf9454193096dc72f791c0f : SUCCESS in 46m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ee9a35ab0b54f7eb358fc8156721f07 : SUCCESS in 1h 39m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fa7e7fe49a344c54bea8ed8ea77337ef : SUCCESS in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"428053bdcb3928f5b8dc9a890830b729750d1524","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 10:17:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/364c24b9a5fb46e7a91e608ff5144ebf : SUCCESS in 2h 11m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b7a39725d3854ecda6c6c1f2ced98c74 : SUCCESS in 2h 11m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d72a4d7c70f94d1e86e833ca26899d9b : TIMED_OUT in 2h 07m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b852525966084e76aea296269e91ddbe : TIMED_OUT in 2h 04m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/22a06a50134c4ee5b9fceec6c514d34e : SUCCESS in 2h 32m 58s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/292704a43aac4b409a551c71de29f233 : SUCCESS in 1h 07m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8bdb454ce65c4ced8549ca54c648f2b1 : SUCCESS in 58m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3bcc808ef7a84af08cac0c0cfb6dffd1 : SUCCESS in 1h 04m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f02a27a66e2c4c9bbaef00e111f02058 : SUCCESS in 1h 29m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad3baafc4ea54fb190df255b719d3744 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b94fa62fd40470f80c8b72eca5d04dd : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6738c0a58154a6ca5ff2a5d1fee4214 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f84efb787f924896997cb64981db36fa : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b5367c3445c4118a532e641caa17cd1 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3589bd523c441f6a051b91c83a131c6 : SUCCESS in 7m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce5943aad1244abdbb6bd6b1bf9fa91c : SUCCESS in 2h 10m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e58095ec389a48b685529ac58b7d89ad : SUCCESS in 57m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f3bfdc71d734bd5a1eb0d1530963059 : TIMED_OUT in 2h 06m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca0d8fd068df4aec8796343e92c5517c : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6dab2948432e4ac8a133ffbd35732b6d : SUCCESS in 1h 33m 49s","accounts_in_message":[],"_revision_number":2},{"id":"565d0ab3d84e93ee517062038ffa0a8fdbed7722","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-01 10:30:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/82c1e74b47064f2b80f3f7add9360cf6 : SUCCESS in 30m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c1a15561fac15041362cfa91da2edaa483e5c1c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-01 19:42:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8304459a15d0e19a813c428597820a711112fe1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 21:45:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b2830b705dcb40b9970209fde6cef4d0 : SUCCESS in 36m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4ccaac873f504005aba7eca8d488d017 : SUCCESS in 27m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d84bf695966a49ceb860e880988d1342 : SUCCESS in 1h 17m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6a176afa2f5f4c1db256450c6b4dd618 : SUCCESS in 1h 09m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/df5c56a79e3f4772b8d689ce1813da1a : SUCCESS in 1h 25m 36s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/730c050be1b74861b2089664d382d535 : SUCCESS in 1h 13m 05s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b3aa3862b9924c56a959f49bd894b044 : SUCCESS in 22m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9a60ebe41a1d42fe9f72bc4625039456 : SUCCESS in 26m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ada4d77c73d440e495d4552e4d866a0f : SUCCESS in 1h 07m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b82114ea7464430ea3cd11292589c48f : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8804d80a3da14ca4b6fae5e849cd9ec8 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e93e5dc430a4892a4b8f7196c0fd802 : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15d4a65c9b9445fe8eb270eddb92b82c : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ba0d99a05784aad8c99924fd6cc13ea : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2d94c78ade40f0b432e82d035dbf77 : POST_FAILURE in 10m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e4796eae28d4997bf19832999038aae : SUCCESS in 1h 01m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dccd79bb2ef54c158a1a37f4ceef03fb : SUCCESS in 25m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d47073bcb32a48ca8db76380513666fb : SUCCESS in 1h 05m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b79652448fe4622a49663440ffabc89 : SUCCESS in 1h 07m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eb8258505701441192eeb2cd3134fa96 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4a5e1d9df74c512b08ddb1475ac8f1d0b7c17eca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 22:07:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f66147062a1240e396d943b01a67afc0 : FAILURE in 9m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc4e27a6e5964db0917be23513c56402 : SUCCESS in 50m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/359ae2f6ac3f46d189d8da7d2f1ceece : SUCCESS in 1h 43m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4023a5a771d240f797ec0dcf2ab85293 : SUCCESS in 47m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"25400311dd65e9dca24dec637b4a214981cd61f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-01 22:30:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/26e7fe4fa7bf4a79b307f594c1669065 : SUCCESS in 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4c3580ee487dac987387209642d1e1b40dd8963","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-04 09:46:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"300ef03409da1e32f994d1946ad2196048dc3a04","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-04 11:00:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/4a73d8de42f041e49a89d6ba10fb363d : SUCCESS in 28m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24f73ef9ddae34a1a65c594e673cce4653b5b761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:18:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/52e3d12ee5634ce88f84566689bf03ef : SUCCESS in 38m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/863c15cac6d1438aa9cace7b9d08e81d : SUCCESS in 48m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f9ab2a9ba0374cfeb5f1db2208306634 : SUCCESS in 1h 17m 22s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8865536177b9406786e80810cf246211 : SUCCESS in 1h 21m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a2303512853b489e865085498a42ad43 : SUCCESS in 1h 19m 40s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6cd1c382119f418da8ed8a12911d844d : SUCCESS in 41m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/26f639c5728e4145b09f396f8b1d5501 : SUCCESS in 27m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c157e101c6384fbaa75f5543a8d47246 : SUCCESS in 19m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/66d1a47c653540a1a828d94a2479970f : SUCCESS in 1h 03m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1540a986ae024c91b7493599b2187524 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65c079e5399844a0b442ba7cf97922a9 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fed202e57f49471da005cb4a2232fb9f : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b8490fc223b4edd8afbeaf5569db20b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24dcb43177134783b6cc15155956a964 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b942b29f76c4eb6ab3dda21bd9e599b : SUCCESS in 5m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ccdb0315f8794af4b5340d24883eb7ee : SUCCESS in 47m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4af900f150e46e589387d93374a6f26 : SUCCESS in 24m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d55db9707f94a78b697a8de6e347300 : SUCCESS in 1h 00m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b7852abf0fb4543921081b9a824c03f : SUCCESS in 1h 05m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70d23704c16a4de4adf0beebf4707f5d : SUCCESS in 1h 09m 22s","accounts_in_message":[],"_revision_number":2},{"id":"40ec594d919f8dbb2034a1def7677d85406f518a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:43:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5aef86d52a5a46f58c755c0b1433a90a : FAILURE in 7m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/19ab9abecab445649b70642a4594b593 : SUCCESS in 51m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/49c5b0ea792744ef89f17052f4786d28 : SUCCESS in 1h 50m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/64011874424c4774a7cabb24e7035431 : SUCCESS in 51m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"62ec5bdf0a583fa435fe5fad8dfa3f83c7df1e2c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-04 13:47:29.000000000","message":"Patch Set 2:\n\nwhy are not we using \u0027unbuildable\u0027 as for everything else? :-)","accounts_in_message":[],"_revision_number":2},{"id":"869c4b40bd5a10b538316560ab37388a35d38791","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-04 13:58:18.000000000","message":"Patch Set 2:\n\n\u003e why are not we using \u0027unbuildable\u0027 as for everything else? :-)\n\nBecause if we add \"base\" then whatever you try to build will just pass and quit. And this is this level of \"no support\".","accounts_in_message":[],"_revision_number":2},{"id":"308075dd0eab6b6718cbc0dfbbde3cc99223fd08","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-04 14:05:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthanks, hrw","accounts_in_message":[],"_revision_number":2},{"id":"43e75f66c4b85ebf933057c06a91ebb5e5c5e60f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 14:05:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5211cf6840824e5309570c86e3fe6a0a32182f38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 15:29:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1b229b5cfdd128116b765d4c1d10e9eee2765664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 15:29:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/97002a80767b48459727c0562812fa0a : SUCCESS in 52m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0a688ba6c5714efea65a8c9334cba11e : SUCCESS in 1h 16m 33s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fa8e285f48b44d14b4c8c45104c52079 : SUCCESS in 1h 22m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a88440a1da27484f954d168c979ac4fe : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc8a166ca85442e0aa654f3f5a08e9e5 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7859ef4181ee4f438ac9e35c9058943e : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a8c27e77fd4441783825f2b7897b55e : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8654574dc014ad89ea2f72bd7015d90 : SUCCESS in 6m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b23c2eac88143ba800584ffb1ee1704 : SUCCESS in 30m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a8166cbf7d4440f9e9c447f79921f4e : SUCCESS in 1h 03m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b11eac9db3ca452189db04ed681f253f : SUCCESS in 1h 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5ab8a2d3c07334760c010202c4fe490ab1bceaa1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 15:31:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ec3d09e98534d65a678c74e0374833b : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fe6fd8dc5d1809f8ce91f677f6affd4e21268551","revisions":{"4605f854149ddfcc05f9cfc02de7eaac37b16f1c":{"kind":"REWORK","_number":1,"created":"2020-04-30 09:12:18.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/17/724617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/724617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/724617/1"}}},"commit":{"parents":[{"commit":"62c8dc95bd79a1b72b8633712847d0ba44c89126","subject":"Debian: use Ussuri binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62c8dc95bd79a1b72b8633712847d0ba44c89126"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-30 09:11:20.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-30 09:12:04.000000000","tz":120},"subject":"debian: binary target is x86 only","message":"debian: binary target is x86 only\n\nOther architectures run will be aborted with proper message.\n\nChange-Id: Ic81fb425fa54953f1d6c12063049d60658deb4ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4605f854149ddfcc05f9cfc02de7eaac37b16f1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4605f854149ddfcc05f9cfc02de7eaac37b16f1c"}]},"branch":"refs/heads/master"},"fe6fd8dc5d1809f8ce91f677f6affd4e21268551":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-01 07:43:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/17/724617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/724617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/724617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/724617/2"}}},"commit":{"parents":[{"commit":"8abfa0cc75be0100793d3c805e9c871d65a4d341","subject":"CI: make sure that virtualenv is available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8abfa0cc75be0100793d3c805e9c871d65a4d341"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-30 09:11:20.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-01 07:43:17.000000000","tz":0},"subject":"debian: binary target is x86 only","message":"debian: binary target is x86 only\n\nOther architectures run will be aborted with proper message.\n\nChange-Id: Ic81fb425fa54953f1d6c12063049d60658deb4ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe6fd8dc5d1809f8ce91f677f6affd4e21268551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe6fd8dc5d1809f8ce91f677f6affd4e21268551"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
