)]}'
{"id":"openstack%2Fkolla~724630","triplet_id":"openstack%2Fkolla~master~I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591","project":"openstack/kolla","branch":"master","topic":"bug/1879265","attention_set":{},"removed_from_attention_set":{"22997":{"account":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"last_update":"2022-01-26 23:33:03.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-01-26 23:33:03.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-24 10:26:47.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-26 19:48:38.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591","subject":"Use distro provided GRUB efi","status":"MERGED","created":"2020-04-30 10:08:24.000000000","updated":"2022-01-26 23:34:45.000000000","submitted":"2022-01-26 23:33:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"724630-bug/1879265","meta_rev_id":"0b94dcdae189f36e29cd795ecb1d9c0f9d06fed3","_number":724630,"virtual_id_number":724630,"owner":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":0,"_account_id":26598,"name":"zhangfei gao","email":"zhangfei.gao@linaro.org","username":"zhangfei.gao"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-26 23:33:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2022-01-26 23:34:45.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":1,"date":"2022-01-26 23:34:45.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":0,"_account_id":26598,"name":"zhangfei gao","email":"zhangfei.gao@linaro.org","username":"zhangfei.gao"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-01-26 23:34:45.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-01-26 19:48:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":0,"_account_id":26598,"name":"zhangfei gao","email":"zhangfei.gao@linaro.org","username":"zhangfei.gao"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-01-26 19:48:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":0,"_account_id":26598,"name":"zhangfei gao","email":"zhangfei.gao@linaro.org","username":"zhangfei.gao"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"value":0,"_account_id":26598,"name":"zhangfei gao","email":"zhangfei.gao@linaro.org","username":"zhangfei.gao"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2022-01-26 19:48:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":26598,"name":"zhangfei gao","email":"zhangfei.gao@linaro.org","username":"zhangfei.gao"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 10:13:09.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2020-04-30 10:23:05.000000000","updated_by":{"_account_id":26598,"name":"zhangfei gao","email":"zhangfei.gao@linaro.org","username":"zhangfei.gao"},"reviewer":{"_account_id":26598,"name":"zhangfei gao","email":"zhangfei.gao@linaro.org","username":"zhangfei.gao"},"state":"REVIEWER"},{"updated":"2020-04-30 12:06:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-19 08:03:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-05-19 08:08:46.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-05-19 08:09:34.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-05-20 07:55:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-20 09:05:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-16 11:16:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-01-21 06:06:09.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2022-01-25 08:28:32.000000000","updated_by":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"reviewer":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"state":"REVIEWER"}],"messages":[{"id":"f047280fc82a4e9aa56ec0311440bce074ef0611","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-04-30 10:08:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"108b8ad400b89252b61340f1497f4d2a2d3d3797","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-30 10:23:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc4ae3a1ab1038c2986c38b20b2d59d9bbd33d22","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-04-30 10:56:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac4c1606294aaffa34ecf0235e0782d779555335","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-30 11:12:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"724e25ebb4c04a696d9e518f742592310a5c3894","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-30 11:39:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/15cb41c213f0463b88223246b8b40359 : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab90cd240a49bb0066e04dc5d54166a0785a364c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 12:06:04.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"31636539d43e34e8ea98dee6f8235cf54390a485","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 12:44:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e90bf33dfc64ec2aff8c6b09fec53fa : FAILURE in 8m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2d97d2e386c446ca74da4bd9bfb7d70 : RETRY_LIMIT in 2m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b66459c24a964d12878ab4a66fbe5b8d : SUCCESS in 1h 43m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ecee14dae38465dbde758de932672b2 : RETRY_LIMIT in 2m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bd9f7b7eb6810874c0a7e189da642379933b934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 13:20:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8dca28fb865649538fd003d822f9dc53 : SUCCESS in 1h 13m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/36c12c52b5cc4b6da43d6e8bf64ab63c : SUCCESS in 37m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/06e2a32bfb0948339c006e91bb7331a8 : SUCCESS in 1h 31m 39s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3a03074e94cc4cc689a3610506a8359b : TIMED_OUT in 2h 03m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c3c00ec035884b38829973fde708eb88 : SUCCESS in 1h 16m 54s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/507d2b09308b4c50ad0708cccfa391bb : FAILURE in 1h 50m 24s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/ed5efedf95814b079be1bfadc83c730b : TIMED_OUT in 2h 03m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b2df103263e440f5aaf18b56e86440f8 : SUCCESS in 19m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a1029c9c8fa248b7b76588bb57ff56da : SUCCESS in 1h 39m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2771d8d7305348bba9e308e074d2780e : SUCCESS in 2h 03m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95607f8e18ce43289f8800e2a8484d37 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31dd5ef1275240eeb023a42c7d68f630 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa5af27aa44c406192a9605a529c308e : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c18d656ad3834412a7a637a7a3a030f8 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c08dbbc5f7d048f69bd990d08bbc2fbd : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9833cc99874d6f96cd0341d6d35d33 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49b23043ac5344ffa93b1549db043e0d : SUCCESS in 2h 00m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e3f0c4018274403833885dfc3984810 : SUCCESS in 23m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7466c2676984ce9a69ec88ef5b9d22b : TIMED_OUT in 2h 05m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f63219dae174ec3813129e6b192c386 : TIMED_OUT in 2h 05m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5df608b02294466ba067aaa31e923c90 : SUCCESS in 2h 21m 54s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7f1f8d8cee2b4b4ba890448f3fccb9d4 : SUCCESS in 1h 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eabfedd1a196517f701134ddf9e2f9a39e2bc4c5","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-04-30 14:55:16.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d09ab731535fff5a891bc0b674d1d5c83118bc9f","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-05-06 02:40:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f38d6cbdf4fdf004fc23da8464004d3608367307","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-06 18:11:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\n-1 just to reflect I left a comment that I want an answer on :-)","accounts_in_message":[],"_revision_number":2},{"id":"faf346560d7de71653d591df33f6de92f519e804","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-05-08 01:28:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"12d11d98591845791bc9d25188ab1654aab03eeb","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-05-15 08:38:30.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"65b0cc787ae865862d6a22fad7ebb5cb91144b1e","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-05-19 07:57:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b154519d6f13bf0f3cea593fbd848c50f8dc7860","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-19 08:03:52.000000000","message":"Patch Set 3: Code-Review+1\n\nlgtm, but I\u0027m no expert :-)","accounts_in_message":[],"_revision_number":3},{"id":"fb270f07a106a92107d9dd2a388f4aed400ebbe4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-19 08:08:46.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good","accounts_in_message":[],"_revision_number":3},{"id":"4e3065fca40c9eaf7a3f616c2321746450b1bc87","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-05-19 08:09:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1fefef6480d82950331fa115d603cf671fd5b6ec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-19 08:27:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/103049d068eb4b66b86756283427d927 : SUCCESS in 28m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0a492a04fb9865baacce4ba1694eb5f2aa9ebabf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 09:21:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3a8dcfe10e9b416da53a65974c60ac69 : SUCCESS in 23m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5b9f5d012a934b93b71c7f472d708867 : SUCCESS in 33m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/535db98781fb4ee4bd55426d3556257b : SUCCESS in 1h 14m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/976bf67c028b4f2f809136b9cfe1be56 : SUCCESS in 1h 12m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a715cea0afff4a76b60bf7403cb29700 : SUCCESS in 1h 22m 10s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7f744dacc9b64cfb8f83a0ad265adc29 : SUCCESS in 51m 08s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/f8a9567514fc48c894f7d0fe03912356 : FAILURE in 39m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a9a5bdf76ef5424fa0f2eaffc1410e36 : SUCCESS in 32m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/119ae6589e1c428c9a2e634217e8ca9c : SUCCESS in 24m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f1f7754dc0d442868e60bcfc2fc71af5 : SUCCESS in 1h 07m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f916b5ed0d84b4bb7f4e70b46fbbc9a : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b426b16a094984975e43dfc97ca2fa : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/689e50d6d1aa417ca5774c29ffa6cb56 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c9c9aa9aa344203af919f7d76b87e56 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7595d4062de47cdbdf58e7cbd904bed : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41022df11fb47dcbe91161f6bb6e518 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cebd5322c3a04336a63179ddefd585f1 : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8fb7cc2e571a42768ecc74811d332761 : SUCCESS in 33m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3deb783d943b478e87d0b9fad8ee2be3 : SUCCESS in 27m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa215fa2a8bf4b5886b0f9421d24d36c : SUCCESS in 1h 11m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/365424a24910472387159667cac1f587 : SUCCESS in 57m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3474f5dc802e418599845d3f36f9a722 : SUCCESS in 1h 10m 05s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/80b19b763766488ebdf63b35c5b92a2e : SUCCESS in 1h 04m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f8244c58001bd2be12c31e843f54e81cb1865dac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 09:47:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7b4263979834a6ea1f82160958fbbb2 : FAILURE in 54m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/54724f651cca4d5abe8185d2dcff5a64 : SUCCESS in 1h 00m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/78759301e9eb48de9177c9cbfd7b3603 : SUCCESS in 1h 44m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f78e482f940a4827a88556b6828ed9c9 : FAILURE in 49m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e396a85af1cf8b6b529ec1f3da895e63d2770fae","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-05-20 07:25:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fff7550ccdbb9eeeff5d9fcf8e7169aa0c57d7e8","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-05-20 07:25:37.000000000","message":"Patch Set 4:\n\nFix job \u0027kolla-ansible-centos8-source-ironic\u0027 failure:\n\"cp: cannot stat \u0027/boot/efi/EFI/centos/shim.efi\u0027: No such file or directory\"\n\nOnly install 64 bit shim and grub. Thus use /boot/efi/EFI/centos/shim*64.efi.","accounts_in_message":[],"_revision_number":4},{"id":"a44307bae1c4fcc718aa35505f39dd8a93785412","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-20 07:55:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b74bc46395c6443893390e0e5f2d619a : SUCCESS in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"061c8f924cae54e5ea1f30459be672eddb83ae5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 09:05:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/468a469fe4524a199e77b4e446938062 : SUCCESS in 27m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fdc409b369e34085881a65a5b2fbda09 : SUCCESS in 1h 00m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4ee19c44551244f084eeb93e6af9474f : SUCCESS in 1h 36m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8bdffb423af949948b3bc282fe3a4c51 : SUCCESS in 1h 32m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/aafcd4cbf8a24ee89ce88628c112e55c : SUCCESS in 1h 32m 41s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e0d265d7616147f69cf3f2d21c6eb59d : SUCCESS in 1h 01m 21s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/772fcdf0a6ab40848e598f3ab517e10e : SUCCESS in 1h 38m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/aa8e5aedcdd2425d8539755dc371b6ad : SUCCESS in 39m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7cc29a70db6f430198abff67bc36ef8f : SUCCESS in 25m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ecaf7054dd214a40be18dcfe2818a1dd : SUCCESS in 1h 12m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9382497f2554935abfe961052403070 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff90ac7b35564e5abd62e1645cf5e6d0 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2eef6ed21984c53bfcf72d16c3a27a6 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4cb63939c4c543fab0a0e857751eaae4 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6da89707f97c4d09afa82dc12bc854c2 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af2a0ec7a244dc79aac44dfc6da5281 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8afdf128ccac48c89c256f067ef2667b : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/31e7dba3c88e41ef9c0e90508f0a0413 : SUCCESS in 1h 28m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ffb4287729647839be4a9efea2cf1d0 : SUCCESS in 1h 03m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb7fe767d456412b9e5b3dd1053c5d80 : SUCCESS in 1h 08m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/12c6f7a6b3844b8994bb8ae9d4e76584 : SUCCESS in 1h 03m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/231e81b14416464494af22cc2168b92d : SUCCESS in 1h 35m 11s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/1b6e1ad9161e499b85529d9d6e2abf1b : SUCCESS in 1h 05m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9676d2fed61ed05a5ae06589254365ca5acc6acc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:14:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6113cdf1501543dc81820a89d3896d4a : SUCCESS in 1h 08m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/00914f4ba15a458ea229adddf7789b6d : SUCCESS in 1h 05m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f490de3c98f3494ea4e8a2a53771b67d : SUCCESS in 2h 05m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/48908bf643b641aa9afaf0d038425750 : FAILURE in 56m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"621a1f20d3f4a2c0d6dc57702ff184f1e2014a96","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-16 11:16:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b7a9d2115b1cfccf11d2177324a321cde82412d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2021-12-17 03:56:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"616e0c88f12d42e381001ecad31b5b4265a17afc","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2021-12-17 03:58:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6505c1fd386a8ae9437fe548c25a10c2bc99d623","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 04:55:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/54370e47dbf24f53ad8bbe385ae9fcbf : SUCCESS in 52m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea2a65776b114dbdbb139a996586582e : SUCCESS in 41m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/85a66ff296e745c797ad602f274b3aa2 : FAILURE in 22m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b47d0a60c13425e8489c12797262aa3 : FAILURE in 3m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ae96c1e09cb483db593970bc957c364 : FAILURE in 3m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/abab9259ed6e4a83ac39fdef071f4121 : SUCCESS in 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c112e3bda833dc691247712e1f38a744ca23687f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 04:57:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4252e8fc26b54a7bbff121a26b0716d6 : SUCCESS in 23m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/89472629ad414ad784c4ca6314183770 : SUCCESS in 39m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2631cba3807c43c6a873d66c18d7b96e : FAILURE in 10m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ddcd91fdb1c64efe96a08d6e96de4d89 : FAILURE in 9m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8ca8238eca0840688f5ee8d174a19d66 : FAILURE in 55m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e9059e864a56426f9dce13160714ccaa : SUCCESS in 30m 59s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0aa1318bc2d94295a13d62ee20ed675c : SUCCESS in 15m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b5031ea4bf374e2895d1449d1af562bd : FAILURE in 7m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9575f1902ec9418ab82e783eb372ae50 : FAILURE in 51m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e04b1a7da730419e835a84a83047babe : FAILURE in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f995a649e23404096f182cb49ffca8c : FAILURE in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a2279eaa9a4451088ccd7308dd736e6 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cbd4a14568d41faa3c6e0c9e04511c8 : FAILURE in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0f96301730446d19a091eafda2279f3 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d07406c3729c4c9e9ddc65508b4b5c5a : SUCCESS in 17m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4528b18f53a49b99b3905cfc21366bc : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/94ca222b27424d3ea55252d95ae418d6 : FAILURE in 5m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f815a6ff6244c318b4644ffad28c78c : FAILURE in 5m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6cf4d7837ee04c4c80db61e727faf8dd : FAILURE in 39m 00s","accounts_in_message":[],"_revision_number":5},{"id":"a7482cd49a4dbb52fa57d35dcae30e22dc135c2b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-21 07:35:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2d3e925974b6652cce8bad06df48457ac97cbd4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 09:06:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/20b40c797b71457599197bd9e70f1781 : SUCCESS in 33m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b5eb21d957c64cb7a4d769e0d3ecd33e : SUCCESS in 31m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7911f65ca2064e35b8676885438827b4 : SUCCESS in 1h 23m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ce22cfe38d064b08a9064ad28ad743a1 : SUCCESS in 1h 15m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/592ef40bf40b4ae4b3e129970fff9e38 : SUCCESS in 1h 29m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/18d4de8c458c4e46b4421f77831c745a : SUCCESS in 29m 18s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a3e6a6da9f2240b493172164fd150414 : FAILURE in 18m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b27663e562084ae6865e17e2ecb4b953 : SUCCESS in 54m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cae4794b2bc848c981f98ffe60695707 : SUCCESS in 1h 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b6dcf3d618e4cbeaa91ea24f6051154 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c7cffb00cb4450fba66414707c9d1a8 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/943865390a144f56b6414a6c427694e1 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8d255055df4a6d88b53bf439a613b2 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73fa135458244f0ab5ecab6002ce03eb : SUCCESS in 3m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/affca72a9aba492bb6c4959fc2b6db3e : SUCCESS in 31m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3cf1658397541e28bd304daeeb8d23a : SUCCESS in 23m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a19be4b9733942ee98b8ae0fbb03841d : SUCCESS in 1h 10m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7fd3850bafc24a488d4a2955761437ec : SUCCESS in 36m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/79bc281b6e9c430fa7d09d70bbe3d159 : SUCCESS in 1h 15m 45s","accounts_in_message":[],"_revision_number":5},{"id":"242c30828ccbdf51db006e1ef31726d0e723db45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 09:29:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/609691793f9b4c94967dbf6664f0c899 : SUCCESS in 1h 11m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb45ae506a1e48e1a02fd7d4e6a12388 : SUCCESS in 1h 17m 18s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7497c70f6e5544b59c4ac30bdb9afe81 : SUCCESS in 1h 43m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5f3e9183c4b407cb12acbf7ecc3606f : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/742b81f5cafb4b2390a1ca55a8a037f0 : SUCCESS in 7m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d00030ba5a9b4ad7ae87dabb921bf533 : SUCCESS in 1h 04m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"82bdc27d4dd528c5c79ba8293f7a90c098c03504","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-21 10:23:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"55ae27b5485512dedd6024b8d4a3b963f1470b3e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 11:46:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/ac9d27045d0e41828d185b5bafb8c3b7 : SUCCESS in 46m 00s\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/3a15c7f8ecb442979b8ca84bb9b5123b : SUCCESS in 1h 16m 42s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/a98bc647ea9a4fdf881174db16ba3e72 : SUCCESS in 45m 31s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d1cb6300f7e44bbad8d80da3cab1d5b : SUCCESS in 35m 39s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/311b887b86ca4e92ac6ad4003d74b2a2 : SUCCESS in 54m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a93df3bf112e0d9471d3cc81b7b003e95f2e2cd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-21 13:28:57.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"a82467b3127b6d1c6cb3c8c7396e656a78a51ac8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-21 13:29:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"342812720659edc80cc3d7fb4b7213b854a60cb0","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 14:58:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/4a5bd343dc23412e96ed7efcc743710f : SUCCESS in 46m 03s\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/631a7c457f3a41f59b5465ed2e69718e : SUCCESS in 1h 20m 24s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/066e6984ce6f4f04b0d2556baed39d0b : SUCCESS in 26m 24s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/1be310251b1b4125b4faf7306f9cd6b9 : SUCCESS in 32m 40s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/128b6f3d25c843d49e794bcce4f09102 : SUCCESS in 41m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e3a4e225e8dad4252cc9e02eef98e9c129bb1d6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 15:00:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bfd7d7695d114c1c9206b7c5da4ed588 : SUCCESS in 36m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5b4dc147f2f0468c8304e36b4aa27dc0 : SUCCESS in 44m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/19053914ac2e4a36a94412f105c61f8b : SUCCESS in 1h 14m 57s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/dc157e225e41471481ec488236bf683f : SUCCESS in 50m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/90205a3a15e04fa0a536f067bd7bcf3a : SUCCESS in 1h 28m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cb28fc21f74143c6bffc063aecc3696c : SUCCESS in 30m 07s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0e54586c64c34c34a81d2d912279ef0e : FAILURE in 15m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/409e28aea1934c52856976c9b59473b6 : SUCCESS in 1h 05m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/01e2a805db3148d99fd7e7d7d4e1c9d7 : SUCCESS in 1h 05m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed0b05b7b124fcbb4658418a5aeef33 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7be626dd74b46d7917b7e5f71238543 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/365c0f2bb6bc4258a44cd5acd7d4a7a1 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b927d221a221449ea1298b9bc10a17ce : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eed90e125caa494e940a39a608f2c5b2 : SUCCESS in 4m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9fb1ad5a5cf344439f32b1be767ca5cf : SUCCESS in 20m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4092d5f0326468ab029133145dd0aff : SUCCESS in 37m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/819a6968792747588169954282fc9155 : SUCCESS in 56m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/baaf7da37d314822aef58ade3cbca3ef : SUCCESS in 41m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9f5ce087c4d41a5b0170e40c506916a : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":6},{"id":"6d05782f479a617b7eb4866baabf02121ef21409","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 15:14:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab6fbaf41e3649f48cc8d628d35e0357 : SUCCESS in 1h 42m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b98c8e18baf6407bbb30a9fe0b3ebb48 : SUCCESS in 1h 33m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2531378815b4a89b21657effdae2fa9 : SUCCESS in 1h 24m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecd8ef74cd234d9d867a761fd6e41da2 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d971d544e5704b109ddb2b97c9e63162 : SUCCESS in 5m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1410fc196e8d4ebea1a0fa1926b1cdee : SUCCESS in 1h 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"141ed53522cbf6dd3d4ff3ac34588c34be305747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-21 16:55:23.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"80956ffa5e916c067dc239a9173f9dc3d0717b52","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-21 16:55:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4997e162915c22411758a44bd6ae1418ca9bfa73","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 18:34:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/f9993808c0294c7ab7ae0eb0783b8546 : SUCCESS in 40m 41s\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/a7bb68a1e86241b89109bbd9a9e3df66 : SUCCESS in 1h 17m 32s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/bf8cbb6eec8242148b10f073b324c8f1 : SUCCESS in 29m 06s\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/301ea3b4fcc74a3c9c965c3ffa4058c1 : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/b9608b4d654f45f09248efb7525a390e : SUCCESS in 29m 24s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/bf9c57eccd6444cd853d611de1fb2002 : SUCCESS in 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b87261137337be3bf10cce94bd4fc33144e1a3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 18:35:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/df579086b6ea4163a37ea22a770d519f : SUCCESS in 43m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/273fda5f73994691a1cb4bd012301faa : SUCCESS in 38m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dac18e6d1f9e44b3a085a79edfd45d02 : SUCCESS in 1h 19m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8cbc99d6fad04fe8adeb628f8c22fdf8 : SUCCESS in 1h 19m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8a1af0188b304d1dab4002c1e58e2c5f : SUCCESS in 1h 35m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/164a0fa6e86c458084fd9cbf293eb582 : SUCCESS in 40m 50s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/31a0bf257f58410d85c610e9bcb96f90 : FAILURE in 16m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/03f5dc0239f64db18b4946b365a247a9 : SUCCESS in 1h 05m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f266f55f07b949c4838ad4b7cf2acfaf : SUCCESS in 1h 07m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cc17a487f504255b041c6881b90327d : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbad9de11e3b4e7fa7618afe946fa116 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/944d149cec734ff99abf650bae4f7dd4 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1341660a0a884f2f96191a628d77e98c : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9daf97862344af0ba601128b9e769c0 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f1f7313862ac4cfe9c4ad2e420f0f20a : SUCCESS in 20m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7ff2d767633491c83f1daccdc623075 : SUCCESS in 36m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a4eadd693124939917e46ebd5ddb6eb : SUCCESS in 50m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/982c471e5a184b12bb6e0e45c537ff14 : SUCCESS in 54m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d11b16f8c2814fad8d7ffa8590406dd5 : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":7},{"id":"0ac5a3f132c5108487c1f4150f8c8da7d5e3b9c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 19:35:04.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f13f5de2c7944ed8b2de13bd92c62f7 : SUCCESS in 57m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0acbf463a55f4b2fb36db317f7c32a34 : SUCCESS in 55m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cfbe5d9fa85c487cb13c3be50041b782 : SUCCESS in 2h 09m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cd70eced25d40278235bc02d40ad685 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f82d13bbbe2d451a8720b12b3752099c : SUCCESS in 10m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3fb38bf4fe714e92b023d443ea642e33 : SUCCESS in 1h 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bda584668da16f22a39fab0fa89ce5dde498dcaa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-22 18:34:17.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"50a9951e56121b6851e4fec12cd3a1bf1ae6732a","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2022-01-21 06:06:09.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"e87360670b3a04df64ef320be745e785b5e849e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-21 08:53:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a58f2cc527a140c62ecbe5b5a26eb8f6d75467a5","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2022-01-21 09:12:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"709e5df477a06d01a66b223f2ecf6a3ba9c3aedf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-24 10:26:47.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4e7cddac13612dc83b654aaef87d9de02458437f","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2022-01-25 08:28:32.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"49a7a3a49f788ecfd74220e58ad17d6ede39312e","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2022-01-26 01:39:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0861a2a69ff6fe884ad15fdf872d6f9b25c0a50e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 03:00:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cc6640befdd646528f6fe65cc80c84a8 : SUCCESS in 31m 56s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e0762df19fc24a76ac950acea3f4e10a : SUCCESS in 46m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/008f4ff6d7924999b9ad05b388ae0d08 : SUCCESS in 1h 17m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0b32dc150e6b43109d8176a7d655ee71 : SUCCESS in 1h 01m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/10ea3970cd334e84a0db8f9e9751bb7d : SUCCESS in 1h 14m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9324276b186147c1b6bc02c0306ee211 : SUCCESS in 33m 20s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/47bdad365e1a4c1e9b4c9b63bed3cdb2 : SUCCESS in 16m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9295300b22042cda3c0bb6383e19dea : SUCCESS in 58m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3bb6b406cf264c45b6be0d8e54262f80 : SUCCESS in 1h 06m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb40e90408184f92806001bd41f4a941 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d801e29c08d84c9bba7f3008710a3734 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b00f48273d904e5ea5d644dfbe057c7d : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ca5a4add594e4cbbad1c650b8e7570 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a831a17db631433084138a5a1003e8ff : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7df5f8d5af4b4be1965a7b824558c1a7 : SUCCESS in 16m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff7b8ee6fa574c5e8d9fa0a7b1522b8e : SUCCESS in 52m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/73a2d1d32bc64fb6a4665c519d19aacd : SUCCESS in 51m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/adc0f578322c43258fcbcd910d3e04a5 : SUCCESS in 50m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fb41e7061124506b0b0bbea84a4b6b6 : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":7},{"id":"d30051cdbade83aa1cb4913e59df26988e2f0189","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 03:33:19.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a5333184e5244cb960d74d48b3c4fdb : SUCCESS in 1h 48m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a61ef57bc09346faa85d42bb57867ce8 : FAILURE in 14m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7cd903d6fc6745be9f50ad5cb09ad592 : FAILURE in 35m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f34a8d14deb48668141afa0a0f35ca0 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2266ece1968b4d238c1e8c735b5fdba5 : SUCCESS in 5m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/85a9c3d604e44e478bfb17cbf2696be9 : SUCCESS in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"67f55493808c68754d7f265e4b0b5a20c931fbc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-26 19:48:34.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"aef21999d801b7d7ee88366e1712ead6cdf77c00","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-26 19:48:38.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"7925904b0604872f770dcd2797ecdd8fdf0b2203","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-26 19:48:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"edd47297aeb92af2b4b9bcdf45db96fdb39ab4a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 21:16:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b7a6942b741b429fbb46aa569b40aac6 : SUCCESS in 26m 23s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c42261cdc6284832a028f4776cc9f724 : SUCCESS in 40m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/43f8c7314f82452595671adafb1e7fff : SUCCESS in 1h 09m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0d7d1d0a81784ce7a5eb92d3bb54d9a3 : SUCCESS in 1h 18m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/67919361f9204da68670aa3ffaaa0143 : SUCCESS in 1h 22m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b7e2c50878dc49d685b40034ee22bc4e : SUCCESS in 28m 50s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/04408fa557e74176b217e0e467829cab : SUCCESS in 17m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/de2026236385471a9172f7d0d5e3c5bf : SUCCESS in 56m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6c61336b91ad4c38bb83d985ce1f4adc : SUCCESS in 1h 05m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82aed710c96d4d8a8ab243a6d87ddef5 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bda5493615124655ae478991ffbe6fdd : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c9973212fd2414287587dbd1052d5db : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec2525420d644cf93a8c6cb22da5db2 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7d2a63d3539430fa956bc084353e3bf : SUCCESS in 3m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d332e6052a6944f7ad13570dd10ab867 : SUCCESS in 19m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d1e7afd7ba24596b18e714915134eb3 : SUCCESS in 26m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e7f70e2c90646bfb3c9b017eb2c5471 : SUCCESS in 48m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bb1e496ecd444f07a8d01d366f8a987c : SUCCESS in 1h 03m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d634139a1f0447c96007f68751e3b85 : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":8},{"id":"9dc8e3fbe9e2a9c862ff667191dd92baec2df053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 21:17:34.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"33e5bd0d8be81b607cf4b0c76a7f44b69ac72422","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 21:34:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6c669775337476ebcfe542762c28cf7 : SUCCESS in 57m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4022c6a8f21a42bbbd425720423ede0f : SUCCESS in 1h 18m 57s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dce925795d554f92892565aaf237c4f6 : FAILURE in 1h 37m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/feac26e33db240678850a74f53c9e67c : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1fa9571b51b8464f881b4f2d55ea295b : SUCCESS in 6m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c32b0c52ae764c209a7bdfc0239fde6a : SUCCESS in 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f57f50a183bb8cd58624331cc6f5e5e318a47e93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 23:33:03.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/77d783ac20b94900ba1ff171579d0f05 : SUCCESS in 41m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f7a1376a989f4b94bc5eebbd8f0d66f2 : SUCCESS in 1h 16m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5ba64a41e88f41f7bd8bb535706f309a : SUCCESS in 1h 23m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3c37e60aa4164c779af74a8e0907e25a : SUCCESS in 35m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e7a7fa756a8e4f79a8a72ab8e6633f53 : SUCCESS in 59m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c706ce9b0ac84eaa9c4becc56dcb6d6c : SUCCESS in 1h 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca3adbb0d614b2194673d9aee44b1ff : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46f7d932398943eda4b2bbf944fd811c : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5710ff35f6334032a0aff811d7d102f4 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dddf054e8b584f1788f1fb734a828b4a : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/787ec38fb94a4f96842ee530799344b0 : SUCCESS in 5m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c92773c744e84fc5a1637003d5d0a2e5 : SUCCESS in 26m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/467e94cc704f4c5aa94676cf13c55c5b : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/624916216c284e54811f17d7706e5842 : SUCCESS in 2h 10m 56s","accounts_in_message":[],"_revision_number":8},{"id":"f83005d7a2c67d80d32ca11b487a4f143fb3a3cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 23:33:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"50ac3ad85b53cbd1dc2123a12540507cb20d7683","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 23:34:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35af7228c0a4cd499decd98eebd12af : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/182823c1e34147f78dc42c9ef449f6ce : SUCCESS in 57s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b6e057d694d90d50ea278411d0edf20afe427e13","revisions":{"fd94d3072376ddaff09dad25a6cfc18c70881e08":{"kind":"REWORK","_number":1,"created":"2020-04-30 10:08:24.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/30/724630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/724630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/724630/1"}}},"commit":{"parents":[{"commit":"62c8dc95bd79a1b72b8633712847d0ba44c89126","subject":"Debian: use Ussuri binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62c8dc95bd79a1b72b8633712847d0ba44c89126"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-04-30 09:58:12.000000000","tz":0},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-04-30 10:04:43.000000000","tz":0},"subject":"Refine aarch64 pxe setup","message":"Refine aarch64 pxe setup\n\n1) Only need to run pxe set up on bootstrap stage.\n2) AArch64 only supports grub pxe booting.\n3) Pxe booting no need to set grub.cfg path to EFI dir when\ngenerating grub efi.\n\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd94d3072376ddaff09dad25a6cfc18c70881e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd94d3072376ddaff09dad25a6cfc18c70881e08"}]},"branch":"refs/heads/master"},"2478c9fbf5532f03ec719223830cabb15342fc89":{"kind":"REWORK","_number":2,"created":"2020-04-30 10:56:29.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/30/724630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/724630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/724630/2"}}},"commit":{"parents":[{"commit":"62c8dc95bd79a1b72b8633712847d0ba44c89126","subject":"Debian: use Ussuri binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62c8dc95bd79a1b72b8633712847d0ba44c89126"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-04-30 09:58:12.000000000","tz":0},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-04-30 10:55:54.000000000","tz":0},"subject":"Refine aarch64 pxe setup","message":"Refine aarch64 pxe setup\n\n1) Only need to run pxe set up on bootstrap stage.\n2) AArch64 only supports grub pxe booting.\n3) Pxe booting no need to set grub.cfg path to EFI dir when\ngenerating grub efi.\n\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2478c9fbf5532f03ec719223830cabb15342fc89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2478c9fbf5532f03ec719223830cabb15342fc89"}]},"branch":"refs/heads/master"},"a95373a38e990276f809fe1c9e892d50cae498d6":{"kind":"REWORK","_number":3,"created":"2020-05-19 07:57:39.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/30/724630/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/724630/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/724630/3"}}},"commit":{"parents":[{"commit":"55eab6edc43ddb24a918298aafefdff322afb61b","subject":"Merge \"octavia: Add ovn-octavia-driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55eab6edc43ddb24a918298aafefdff322afb61b"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-19 07:33:35.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a95373a38e990276f809fe1c9e892d50cae498d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a95373a38e990276f809fe1c9e892d50cae498d6"}]},"branch":"refs/heads/master"},"ae8d6c0c73f670f102518f24e20716b4f49ee6ca":{"kind":"REWORK","_number":4,"created":"2020-05-20 07:25:06.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/30/724630/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/724630/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/724630/4"}}},"commit":{"parents":[{"commit":"629a2955f4c7838aae526f5af8c81db96f34a8b4","subject":"Merge \"Add ussuri prelude to release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/629a2955f4c7838aae526f5af8c81db96f34a8b4"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-20 07:20:16.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae8d6c0c73f670f102518f24e20716b4f49ee6ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae8d6c0c73f670f102518f24e20716b4f49ee6ca"}]},"branch":"refs/heads/master"},"b91406df21289fb96a0f420cb89da780b5beebf7":{"kind":"REWORK","_number":5,"created":"2021-12-17 03:56:10.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/30/724630/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/724630/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/724630/5"}}},"commit":{"parents":[{"commit":"5dede104bcdfa72a8da32c46fe0bd27f54221d4f","subject":"Merge \"horizon: add various dashboard plugins for ubuntu/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dede104bcdfa72a8da32c46fe0bd27f54221d4f"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2021-12-17 03:56:09.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b91406df21289fb96a0f420cb89da780b5beebf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b91406df21289fb96a0f420cb89da780b5beebf7"}]},"branch":"refs/heads/master"},"957a2219a0b20a5af734e5b4c80004b820597937":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-12-21 13:28:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/724630/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/724630/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/724630/6"}}},"commit":{"parents":[{"commit":"5171c3ed8606e8aed3389c8642beb60105861a1d","subject":"[CI] Test Ironic on Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5171c3ed8606e8aed3389c8642beb60105861a1d"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-21 13:28:57.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/957a2219a0b20a5af734e5b4c80004b820597937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/957a2219a0b20a5af734e5b4c80004b820597937"}]},"branch":"refs/heads/master","description":"Rebase"},"5bf101e932a3203f340d0cffbe078554c9594b81":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-12-21 16:55:23.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/724630/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/724630/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/724630/7"}}},"commit":{"parents":[{"commit":"02e8e76ae762663c898be4b866daea03abd19494","subject":"[CI] Test Ironic on Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02e8e76ae762663c898be4b866daea03abd19494"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-21 16:55:23.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bf101e932a3203f340d0cffbe078554c9594b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bf101e932a3203f340d0cffbe078554c9594b81"}]},"branch":"refs/heads/master","description":"Rebase"},"b6e057d694d90d50ea278411d0edf20afe427e13":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-01-26 19:48:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/724630/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/724630/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/724630/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/724630/8"}}},"commit":{"parents":[{"commit":"e8651961929a00fb8684c56983c2f9c46822553d","subject":"Merge \"Clean up fluentd image recipe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8651961929a00fb8684c56983c2f9c46822553d"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-26 19:48:34.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6e057d694d90d50ea278411d0edf20afe427e13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6e057d694d90d50ea278411d0edf20afe427e13"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
