)]}'
{"id":"openstack%2Fkolla~724900","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I4a93ba532236f89b6a963365f483b91e474515e5","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I4a93ba532236f89b6a963365f483b91e474515e5","subject":"Bump versions for Train","status":"MERGED","created":"2020-05-01 18:07:10.000000000","updated":"2020-05-02 16:38:16.000000000","submitted":"2020-05-02 16:37:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724900-1588437420581-62275d76","meta_rev_id":"f56e445dff1e6aac1ae9d05a16da136085f8eda2","_number":724900,"virtual_id_number":724900,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"date":"2020-05-02 08:59:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-02 16:37:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-02 09:28:33.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2020-05-02 15:17:35.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-05-02 13:57:50.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-05-02 16:24:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2020-05-02 15:17:35.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-02 09:28:33.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-02 13:57:50.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-05-02 15:17:35.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-02 16:24:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-05-02 16:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"426ae20b34013a6d356c49758076f4795d8167f0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-01 18:07:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12965e39386967c68e28853aab53497103135046","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-01 20:07:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/4af34fe12f9c45f8baa2a250a5df1dbf : FAILURE in 28m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7a60e2673823d5ca701d23fdae3c86f0a2a11be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 20:26:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f3c510b54542488396ff91ebba36b6ba : SUCCESS in 45m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/beb1c658c0c54d8185d550311b34b13b : SUCCESS in 1h 16m 42s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/eb5bf3277e61405a92209e6fadfd7db7 : FAILURE in 1h 20m 50s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ee4034bdcd2f4570af5022dbb9c48398 : SUCCESS in 46m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bc519b1ed55b411fa8f7d8e4e5de2c7f : TIMED_OUT in 2h 01m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7e6e7dd2e3d647359616dcd45e4901ae : SUCCESS in 1h 51m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/37b97336a59144eaa66d6b84284fb317 : SUCCESS in 1h 54m 33s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/46057452828146e289809bacfed82806 : SUCCESS in 1h 08m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cd1d26d5c3994dcda7ae655ee88ee820 : SUCCESS in 1h 39m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0ee1c5b1af1144d0821f8ce0edc4815e : SUCCESS in 55m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/de27adf631a04ef78af23fb30d29f867 : SUCCESS in 25m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b9e04d4690284438a0bf8e6808c80fa8 : SUCCESS in 27m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e7d872b694e94e1da06ab83c460e731c : SUCCESS in 1h 01m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f37594acda34e648d626acd40efa0dc : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cf481188f834bbd9d87ba55e33b67d8 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cce533a5791446628b115cc3d5afb465 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b438a745c2b4996aca1ac3059bc795f : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/443ab5e7da4d477fafd9d086d4e748ee : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a642da8147da4d039bfb1dd24945f80a : POST_FAILURE in 7m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/72287799e19843aca9bea17c319f2303 : SUCCESS in 2h 05m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f8dde163df24f2d9c143be2aa01fa15 : SUCCESS in 2h 05m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4275be5c4f9a4449b29dabf39bc53d0a : SUCCESS in 1h 01m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9006cc3dbe1a4214a478fa1001672de3 : SUCCESS in 2h 01m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be484ceac67a45feb4161b7eed4e78bd : SUCCESS in 2h 16m 08s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"5953364b76dfea9b90910d52749210520506495b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-02 08:59:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5848f233d05329fc68aa665c27a8a6c682c8db2d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-02 09:28:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/1c7f961eb0e84c44aaea944a8c5e1c09 : FAILURE in 27m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8cda2d5cf5d884b71ad026dd786926e86a4610f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 10:51:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/613714c00f7a429cab492056843405a5 : SUCCESS in 37m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a33218ac5d4441029d2b9e481d519061 : SUCCESS in 49m 24s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1a3ac01a81c043528bfda900bc8244ce : FAILURE in 21m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4cb41e64203c463a81b263a5388fd341 : SUCCESS in 1h 09m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/79025b5c4cee46dab656ab09b9c7bdd9 : SUCCESS in 1h 20m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c620809b7e584ca6b581e37c485ff673 : SUCCESS in 1h 19m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/52ffa65bea6043b69948d782b62e66ec : SUCCESS in 1h 23m 46s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3e1eb87af696464d927c0f9e000d44f5 : SUCCESS in 1h 03m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/63b77f55ee384a43ad28bfac9485f7d9 : SUCCESS in 1h 47m 11s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7ca9c55f3be1499e86be35c0ebf968f7 : SUCCESS in 1h 00m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bc2b5e9a6d634df885b76077b5a57fe5 : SUCCESS in 24m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/33d40420f08848978f73225cb216c7e3 : SUCCESS in 18m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2a7f7a96bec94c7089bb2706b9a88ae1 : SUCCESS in 56m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/494953a5b3d5425ab2a7275700f179f7 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/393e1a938b5e4c808e0b4e947c532f26 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa11e33919e44381ad3f0815f7240059 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64da49381ff4438d92802735ec3aad03 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32a0393fe2db4900af5a95b752e22271 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40523d7a870942cfbdcb4c165b5f7286 : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d3306cd90be145acbc783387a09a7f74 : SUCCESS in 21m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/646c59258c714c33aeb1011b3c61f800 : SUCCESS in 25m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c2fd9dbabfb14a3fbf3640c215f5a1f5 : SUCCESS in 59m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/294d3d9d19e24092bd05c6f4e91ee1bc : SUCCESS in 1h 06m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fdb27b350bca4c458b20d56bfaa547af : SUCCESS in 1h 10m 54s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"294cb949addc3716a22a7a23a6819dd80e5d9cb2","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-02 13:57:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"994c210babe6977af1ad81ee4c4818829dca5c83","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-02 15:17:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf4414de90adf91075fdc360abb4d20a5bb0827b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 15:17:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5efd1fbca89e6e545b06e72b5889bc5dda32910","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-02 16:24:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"815d9f3322a8d95552f5b3ea46d163126a5f3588","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 16:37:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/420b11f153fa461c83d5a468fdcd8a38 : SUCCESS in 48m 36s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/67c401e361fc49d491c242d10007295d : SUCCESS in 34m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/48a89ac5548240568db9f1bb0b99f316 : SUCCESS in 1h 11m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/41956264e90c48f48d0ef3e0f2a9c917 : SUCCESS in 1h 08m 15s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/81dec60ad7d1400ab0d982e564c97afe : SUCCESS in 1h 18m 32s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5d9c2c56221e4995abb56a9fca0ae346 : SUCCESS in 57m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ff01be9a6a046229fdf7604c28964fc : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8247773ae7c04a69b6a4f2c6644d999b : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad42e4cc9b4f4e71a0c46f62afbed715 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96612ebcf79a482ea31e6f88ad78b676 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/accd9ce757224fd0bcfe709590b04fc4 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd407ccc45354b10a86a7430ac2c620c : SUCCESS in 6m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/00767ad6b2c44ab49d16fddefc4ccef3 : SUCCESS in 32m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5f729dbccef43aea5bedca1b8802216 : SUCCESS in 42m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e73962df22774fc1a67a37b3420268e8 : SUCCESS in 1h 13m 49s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"8d1b64d670f7247fdb6dc900ff465c037fcc640d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 16:37:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f56e445dff1e6aac1ae9d05a16da136085f8eda2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 16:38:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7978b511b5074fd586b890ad01d1d5b1 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c8863b0a877a00a810dab55be818163aeb83c61","revisions":{"4c8863b0a877a00a810dab55be818163aeb83c61":{"kind":"REWORK","_number":1,"created":"2020-05-01 18:07:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/724900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/724900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/724900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/724900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/724900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/724900/1"}}},"commit":{"parents":[{"commit":"dc560d132931401edbda8e81f684512e10e14163","subject":"drop tripleo-common-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc560d132931401edbda8e81f684512e10e14163"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-01 18:06:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-01 18:06:54.000000000","tz":60},"subject":"Bump versions for Train","message":"Bump versions for Train\n\nChange-Id: I4a93ba532236f89b6a963365f483b91e474515e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8863b0a877a00a810dab55be818163aeb83c61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8863b0a877a00a810dab55be818163aeb83c61"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
