)]}'
{"id":"openstack%2Fkolla~725018","triplet_id":"openstack%2Fkolla~master~Ibad6e375c9cb88f089d7bea86d3ce059584fb2e8","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ibad6e375c9cb88f089d7bea86d3ce059584fb2e8","subject":"Remove find_disks.py since the ceph image is not exists","status":"ABANDONED","created":"2020-05-02 15:40:19.000000000","updated":"2020-06-13 06:03:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8605a92b9168ba8a630f094245f56f1ce8875351","_number":725018,"virtual_id_number":725018,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-02 16:25:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-02 16:09:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-2,"date":"2020-05-02 16:26:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-02 16:09:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-02 16:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-02 16:26:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"c8e3a2108350fb0e103444b65ecb304ed8ce9498","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-02 15:40:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09a741f77e08716dc9f3b1e1aa3864cc26e852e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-02 16:09:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/bcf08cfb1c274917a4162d85ff42d500 : SUCCESS in 27m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c158522fc8735012efeb2b92850561c6ec8458e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 16:25:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/810d7a84f5b045cd95956e523a27c5e6 : FAILURE in 23m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ce39aa6b0371466e9cc70db4a85c5c71 : FAILURE in 26m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ee2263467c2347b98f936e96c9c6fc1d : FAILURE in 26m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/66597619fc6c4b35b13cfbc8f8eab6a1 : FAILURE in 27m 13s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1c45824a070f4e048659be0d525d869a : FAILURE in 25m 26s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/32b2eb775f8946f4a1a9a32ee1b946db : SUCCESS in 43m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7b4628efe3844dfe9c8652ad09710404 : FAILURE in 22m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4f7f0c0f27054f64bb8fba7c4d219be7 : FAILURE in 20m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b785170f36364b0fb47e101f8c01c472 : FAILURE in 19m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96ca1cc51d114eae8c5b0cbaf24ab391 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac864992e4a4448815c2f6f777fc7a6 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67c41ba12cd04049a2db07917d57cd63 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31ae7fdccdf54b679b9ee86eaf8f2e17 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc8c940c92294a199662d24759034a8c : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f40054ae1454b59b8632dcf977b9ce9 : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/676db76e5ee64a67987bfe7e3ade478c : FAILURE in 26m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc870805b55f422cb257a5093ea94bcf : FAILURE in 34m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/688dffbf844c49ffb17087a8723eb4cd : FAILURE in 24m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/501577573daf47a38d1b60504e9d310e : FAILURE in 19m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d969ecc5ccb4c269b8603c6363a4eaf : FAILURE in 17m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8d33b40bc3c3ef86a22e71cb5eff48a882e59011","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-02 16:26:47.000000000","message":"Patch Set 1: Code-Review-2\n\nstill used by swift; I would accept change removing mentions of ceph if you want to pursue such direction but I do not see it much worthy","accounts_in_message":[],"_revision_number":1},{"id":"99b2d88b88f57e0fa0cce99f47e522cc6272a485","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 16:33:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/467e41ed3b914bd39855cf7addf0d0dd : FAILURE in 7m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d3b5623178947c6947aaf02616c3186 : FAILURE in 48m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c2867109accb49159bb17963e598dbf3 : FAILURE in 45m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2efb632e548a4429972f9fb37710c057 : FAILURE in 52m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8605a92b9168ba8a630f094245f56f1ce8875351","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-13 06:03:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"799854543f48be91b579ba054390aa4a82835100","revisions":{"799854543f48be91b579ba054390aa4a82835100":{"kind":"REWORK","_number":1,"created":"2020-05-02 15:40:19.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/18/725018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/725018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/725018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/725018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/725018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/725018/1"}}},"commit":{"parents":[{"commit":"8abfa0cc75be0100793d3c805e9c871d65a4d341","subject":"CI: make sure that virtualenv is available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8abfa0cc75be0100793d3c805e9c871d65a4d341"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-02 15:38:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-02 15:38:13.000000000","tz":480},"subject":"Remove find_disks.py since the ceph image is not exists","message":"Remove find_disks.py since the ceph image is not exists\n\nChange-Id: Ibad6e375c9cb88f089d7bea86d3ce059584fb2e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/799854543f48be91b579ba054390aa4a82835100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/799854543f48be91b579ba054390aa4a82835100"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
