)]}'
{"id":"openstack%2Fkolla~725221","triplet_id":"openstack%2Fkolla~master~I25cfa62dbf20490685617b4a4185bf95ab91725e","project":"openstack/kolla","branch":"master","topic":"bug/1873421","hashtags":[],"change_id":"I25cfa62dbf20490685617b4a4185bf95ab91725e","subject":"Use Python 3 for mod_wsgi in cyborg and monasca on Debian/Ubuntu","status":"MERGED","created":"2020-05-04 09:45:58.000000000","updated":"2020-05-08 19:13:50.000000000","submitted":"2020-05-08 19:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725221-1588964866284-61dee02a","meta_rev_id":"d18f3c4ee8a5979e10cdaeb53e4cb2c0cf7cc4fa","_number":725221,"virtual_id_number":725221,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 19:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-04 12:39:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2020-05-08 17:03:53.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-05-04 13:47:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2020-05-08 17:03:53.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 12:39:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-04 13:47:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-05-08 17:03:53.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-08 19:07:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d4d603f7bff2d104f9b713414a0942587832ce0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-04 09:45:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07488d468f8cd355fd33b72bdeb10c1acd03cdd1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-04 11:28:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ff6c132236ae4f8aa186abf294ba96d5 : SUCCESS in 36m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7f4d66c7dd14d067f8059508e8b14e61153d798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:29:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f77e3dae6a304c8992fa311f0e720472 : SUCCESS in 31m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a6d6009884a641f59a048701fa9ed2c6 : SUCCESS in 28m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e88a7a7d6ea0461987d8c1a9d5f302ae : SUCCESS in 1h 15m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fdb36f60a02a4b4d9d7e79b10dff3b6e : SUCCESS in 1h 09m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5b4387e83f954db4ba8d350eecaf709f : SUCCESS in 1h 16m 15s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/bc8f059b36054f13a9dd9a11ef4cd620 : SUCCESS in 1h 10m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2185a404bac2427ca6781c1074e99034 : SUCCESS in 23m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0d6d87a1a3fc4294b5d24980ec03c2f6 : SUCCESS in 45m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c472ea0e7c2c494bb145e3760a1ff787 : SUCCESS in 1h 09m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0df081dde8ac4730825987e1286da848 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47e89c7f02844649b6566c3b1a9ac42 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6787807081b849a1a65e634448e0319f : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bf387b4a1214c909f0b5c17f453af4a : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21c8856f541f45ea9f87d04a379574e5 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4189b374d3b4f2f8caf29bdbeee4b42 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/510c4161158a4d0a83de8bbf0dd29c8f : FAILURE in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f0d2bf29d2b4a9181f11c6fe91a7090 : SUCCESS in 27m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc1efbb899564e3c90186bef4784a42b : SUCCESS in 32m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b310cbc086c40b68894ec398d341057 : SUCCESS in 1h 03m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e768376fc4f438d8a7279177288fbc2 : SUCCESS in 54m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5491e4199c6148c9ac6d8f18c84bb979 : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7804df9cd412cf4b1e524315aaa6c7ee39dc21ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:32:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c259ea2a87cc47339fcfccf6c74bc803 : FAILURE in 7m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6308711ad5a43638c2bae2700119107 : SUCCESS in 50m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a87b612ef4e4e44b79c72018b2b76b5 : SUCCESS in 1h 46m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d52ffd1208740f4abb25c114f7d2724 : SUCCESS in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ee49dfdbbdf9ae44e4479d023ad8f068f5bcfda","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-04 11:46:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5c194bde6346affe26e09e158cc093837923859f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-04 12:39:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/efaa471c94694b60b54ca301215a244f : SUCCESS in 26m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c4ef3268d17fd89b8c7ea6a88dad3982e819632","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:29:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/aa02ccd625894bddbe625df38da12cad : FAILURE in 8m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6d1012c1caa4a62a3a77688d66a7d87 : SUCCESS in 51m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e9ef7fecb5e4fd5a92d265e4f12fffa : SUCCESS in 1h 41m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e10b97265b474bb3bef15b62ac354ed1 : SUCCESS in 45m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f699f355cbf2601adec03f1ab029d008edd18c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:35:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/18176529122744f59aaafefdb583e76e : SUCCESS in 25m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8a4de87512104927975e512d24918054 : SUCCESS in 29m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8a52b15c8fda4ba9a10786cd663d3a20 : SUCCESS in 1h 29m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a6b7f7af84924af6ba2cd8d6b2e02d33 : SUCCESS in 1h 09m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ab82ca68c3474a028e5ecb7139e1bcfe : SUCCESS in 1h 43m 44s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6d237f89f08448edb0d368b377e58ac0 : SUCCESS in 43m 07s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/677edb0f71cb4cc3b28740198006f224 : SUCCESS in 20m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/49953db67e2a4ee7bf35756fc5630169 : SUCCESS in 17m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c44aa50768a49818dbd340ee6fe009e : SUCCESS in 1h 03m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7089fe1bde34ceebff7d781d3f1fe33 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace1da0395404c079e54d6a1b1c75b1c : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48f1290c637c41d59af87b9f9e3eb0cb : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/378ffff414c845be8b9c7ec8265fe0e9 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d3d11fca6cc497b946293f805c90d91 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52b3e05e920644958199a22fbfe0ca0f : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/847c3781139c494b930487c2ae2ef02e : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6259aecf1a0449fe88dcc6963d4bda01 : SUCCESS in 23m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7555db2a8b1d4969b28ffdfaa60f9b8a : SUCCESS in 24m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1fd5dcd40e3646d89611eb38ded8742a : SUCCESS in 1h 03m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee13a633acc1407b98ba16132fba2c6b : SUCCESS in 58m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1faf95b1ae2644aa80cd1418c8ceeea2 : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":2},{"id":"204356538857b8a2a103799fbecce39fb56fb6ad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-04 13:47:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f2b1f7befd87d8ee46f536325f7323d7e369449a","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-08 17:03:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"08e9be7966f4f9dd78f79355d8a38e1f08d5263e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 17:04:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"188d498eeeb91da03f64160d7aeb8e24befc9970","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 19:07:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/67defef5ea7541d2866b00bdd08bee0b : SUCCESS in 1h 02m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d16d8e3750584b9697ca720a183f4736 : SUCCESS in 2h 02m 16s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d5b976aa139a4eefae9c8e556541d28a : SUCCESS in 1h 12m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adc313171a06478ca54a9c59c4f9f2ed : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6968768cdf544512a90f57a73a43a02a : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c305ebf65bb4f5398fe88a3694b48c4 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2909a41ace8b464f961ab4e12727d05b : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8281f0574149989d75a92b98d50f3a : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca81f940b7624f11b5ee59960947f446 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b81466bec363400aad614d6f6dc5a5fb : SUCCESS in 24m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5085442fc3884be7bed0bea68b4dcfc3 : SUCCESS in 1h 05m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3f387dbf91a04f13a78ed8d97ddfdf21 : SUCCESS in 1h 22m 32s","accounts_in_message":[],"_revision_number":2},{"id":"4d22dbded5d20af465ec209f47511c030fb52a86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 19:07:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1da420d3a4ef45c12fc7ef0ac867b3aa9d0526a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 19:09:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd8d546097e4ca69058269993c53c5a : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0aceafca30194c3fbdd427603794a9f6 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2},{"id":"6206798aa51571b725e08b13e751cbe39ee8e0c7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-08 19:13:50.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit f221b4f7ab7745ee20c8e97d6a1a2d3c7e74f466","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f5d31da12fa66cedfc2e270b8cd3be8e2bb9254b","revisions":{"305c19c88b43c50a8acc014489e86514f600f137":{"kind":"REWORK","_number":1,"created":"2020-05-04 09:45:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/725221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/725221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/725221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/725221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/725221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/725221/1"}}},"commit":{"parents":[{"commit":"8abfa0cc75be0100793d3c805e9c871d65a4d341","subject":"CI: make sure that virtualenv is available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8abfa0cc75be0100793d3c805e9c871d65a4d341"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-04 09:32:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-04 09:45:45.000000000","tz":60},"subject":"Use Python 3 for mod_wsgi in cyborg and monasca on Debian/Ubuntu","message":"Use Python 3 for mod_wsgi in cyborg and monasca on Debian/Ubuntu\n\nIn Train kolla switched to Python 3 in Debian and Ubuntu source images.\nFor services running under httpd with mod_wsgi, they should install\nlibapache2-mod-wsgi-py3 rather than libapache2-mod-wsgi. This was done\nfor most images, but cyborg and monasca were omitted.  This change fixes\nthat.\n\nChange-Id: I25cfa62dbf20490685617b4a4185bf95ab91725e\nCloses-Bug: #1873421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/305c19c88b43c50a8acc014489e86514f600f137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/305c19c88b43c50a8acc014489e86514f600f137"}]},"branch":"refs/heads/master"},"f5d31da12fa66cedfc2e270b8cd3be8e2bb9254b":{"kind":"REWORK","_number":2,"created":"2020-05-04 11:46:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/725221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/725221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/725221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/725221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/725221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/725221/2"}}},"commit":{"parents":[{"commit":"8abfa0cc75be0100793d3c805e9c871d65a4d341","subject":"CI: make sure that virtualenv is available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8abfa0cc75be0100793d3c805e9c871d65a4d341"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-04 09:32:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-04 11:46:36.000000000","tz":0},"subject":"Use Python 3 for mod_wsgi in cyborg and monasca on Debian/Ubuntu","message":"Use Python 3 for mod_wsgi in cyborg and monasca on Debian/Ubuntu\n\nIn Train kolla switched to Python 3 in Debian and Ubuntu source images.\nFor services running under httpd with mod_wsgi, they should install\nlibapache2-mod-wsgi-py3 rather than libapache2-mod-wsgi. This was done\nfor most images, but cyborg and monasca were omitted.  This change fixes\nthat.\n\nChange-Id: I25cfa62dbf20490685617b4a4185bf95ab91725e\nCloses-Bug: #1873421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5d31da12fa66cedfc2e270b8cd3be8e2bb9254b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5d31da12fa66cedfc2e270b8cd3be8e2bb9254b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
