)]}'
{"id":"openstack%2Fkolla~725352","triplet_id":"openstack%2Fkolla~master~I2c80080a5b2203c6b22c361f2b6126d65c0c50dc","project":"openstack/kolla","branch":"master","topic":"tripleocl","hashtags":[],"change_id":"I2c80080a5b2203c6b22c361f2b6126d65c0c50dc","subject":"Remove tripleoclient container from Kolla","status":"MERGED","created":"2020-05-04 17:35:26.000000000","updated":"2020-06-10 08:35:14.000000000","submitted":"2020-06-10 08:31:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725352-1591777908026-237d3608","meta_rev_id":"9459a69729096986a295e73a4f49546684f66e8c","_number":725352,"virtual_id_number":725352,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 08:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-05 09:24:18.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-06-08 07:24:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-06-08 09:30:47.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-05 15:54:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-06-08 18:19:02.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-10 06:54:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 17:40:34.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-05-07 18:43:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-05 09:24:18.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-08 07:24:28.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-06-08 09:30:47.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-06-08 18:19:02.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-10 06:54:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-10 08:31:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a739d34d1faf071c00cd4f4504e55c6767eb2b7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-04 17:35:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42714fccb5b15eea540b2b299917923ba87755aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-04 18:00:00.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks Emilien, could you add a release note?","accounts_in_message":[],"_revision_number":1},{"id":"1d16dc7190ff5f3cecd960bd740caeb3b1b2e266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:14:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ae365711d4dc4e86a40f3d7530b9425f : SUCCESS in 19m 30s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8a7131396e6c41ae9a3aed5d3aa05626 : SUCCESS in 38m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8ca1ea762cd54364b0aea47d53ef8bae : SUCCESS in 1h 30m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/64825f75a6414c858cc137455a237d80 : SUCCESS in 1h 13m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cc9ad963d76f4928a0b1d025ed9f3cbf : SUCCESS in 1h 25m 48s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/cbe5f304535e4a97a2db40da841e9c80 : SUCCESS in 59m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/42cb2bf1af8c45e3be262dd72b1b6242 : SUCCESS in 21m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ec30cc36e79043ea96efd76d18ae41e0 : SUCCESS in 18m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c4adf5ea3ea4d24a95da3fe650ff343 : SUCCESS in 1h 01m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beee01955fa34c9d87bf3cc864598375 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b52373de4f54bdeb54786d260027f92 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccd5025d9622469aadf030556d0731ea : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df3b66b525fa4524a7ba179e51ee1af6 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8f3c55b5d2e49a18553c785e065a88c : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2867846b8134352ab68881610e4bc76 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e3d980d7ea414223b9025da0a0dcfd30 : SUCCESS in 36m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a9e702ddac84a9abcd3c0a2e26cd741 : SUCCESS in 31m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6cd32f603204d05a09103b4037754d2 : SUCCESS in 1h 03m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b16eef1680ca4d999aac3e9469497e9b : SUCCESS in 1h 00m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1169c422938d42b9aca92d3ebe70fe6a : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b92df7965413c2d6201ecfe112293d0e48615b50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:22:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/cea23f6f20f44047abadad2603f7029c : SUCCESS in 49m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a931adfae8bb4741b07d634fb056b392 : SUCCESS in 49m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58235dac65df4aa08fa44e9ad8fc030d : SUCCESS in 1h 42m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/48765b3994374ea88fecff24b15cf46c : FAILURE in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42cc77d2bc0735aa474e8b039b077ab8f48f6a87","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-04 20:01:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/34e2a8e8d01249da9e62fc24db139058 : SUCCESS in 28m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4eccdcd40f9dec31a4ef1b380b21649db88078df","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-05 14:13:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd2e597a32985d21e3d83f52d41c8ae5ee7a00c4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-05 14:13:26.000000000","message":"Patch Set 1:\n\n\u003e Thanks Emilien, could you add a release note?\n\ndone!","accounts_in_message":[],"_revision_number":1},{"id":"ea86fbc946297a00475dd8d7afccbdc18dc06586","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-05 14:45:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"dbfe6411019d17054ef59a3bf0af1235c648abd8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-05 14:45:39.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks, Emilien. I made it sound a bit lighter. :-)","accounts_in_message":[],"_revision_number":3},{"id":"999c3ad23771ce5deb86af6e7103f0f1c91fdf1c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-05 15:46:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"591e4aa913f2c10de455f77e82408f0da97527c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 16:30:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/69a108f75ae9456c8f84ed8715954e49 : FAILURE in 52m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe550fed880a408993060423ca142073 : SUCCESS in 51m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb8e9b002c1140039fedac91b98402df : SUCCESS in 1h 44m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4652c3d59b5d4b709f0b4221658e66f5 : SUCCESS in 49m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"29584001240f2c87932a182d76c32ce029160f02","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-05 16:57:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/19c570690ea446819ab68b7f979f7c36 : FAILURE in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a5ad2a0689730321fa73fb75f9addef8223cbdd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 17:04:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c6339d9a13d548feb15910d7a3d4d148 : SUCCESS in 36m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cf188182845d4a6d84a4f0c2a7cc81dc : SUCCESS in 35m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fefdc34f8bdc47aba45fa8388019a72c : SUCCESS in 1h 32m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c95c218906044da0b2483fa8841c3696 : SUCCESS in 1h 12m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2fe7a5c821684bd39ead82fcc7ecff67 : SUCCESS in 1h 26m 22s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/aeb4bbc52a6c4702aca3ae573a28f6b3 : SUCCESS in 1h 00m 54s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/36bc9dabd5194087a2bca15dea73b146 : SUCCESS in 42m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5ad8f87dbcfe4e6ebb1bd3fd50693cd9 : SUCCESS in 1h 42m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/973a0028b1fc4261a7e203248cb31519 : TIMED_OUT in 2h 08m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4af037b11214a4282147d409eb8b4f5 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380076931f5f4fb99240f9a12ce1ec07 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90f3492225974e6f882018a1d1dd5dff : SUCCESS in 7m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f66501a98894350a68a1c1b8366949f : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c5438e48f744c3991227871607df0ba : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1b643d757147ec8801de00d218b390 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f6dc2181aec42a0b31afcfe793c132d : SUCCESS in 9m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f6e6891c31c340e89f09c987cb090c7e : SUCCESS in 41m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b70fd15e38b4f63a20038afe40c3c3e : SUCCESS in 44m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/48d7232717f24ed7b7145396b17774c0 : SUCCESS in 1h 29m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4d64f09b4b24ac7ad69f72ad6271da1 : SUCCESS in 1h 27m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6622773985df47859d4b6fdb94718c9d : SUCCESS in 1h 14m 07s","accounts_in_message":[],"_revision_number":3},{"id":"72fe3141f0dc67d2ec566e9a6502e18bb5c3e476","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-05-06 08:07:41.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"45d4564a78e595628d0917a793cde77593ed8ea9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-07 18:43:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"82412114332d89b3e87f07efcc466a80927e95c3","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-05-29 08:09:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"45965c00ad3777b7f4b6391742a943fa9e0cc497","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-29 08:23:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d4816e179b6049a990c1044be1b1185a : FAILURE in 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"33a6baa57254b63ffbb4d463fa34e4cd36c8ac84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 08:46:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7d6651a9b7847ebb6abacbaeff58569 : FAILURE in 20m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0228bd173623432cb604b35f127c7c12 : FAILURE in 22m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d181176b272c4a0298b420087e034f28 : FAILURE in 23m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ac8f445d0e34630ae26bf2057938953 : FAILURE in 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f9e53b1b34177d48b72a0b54818ad33eac39fa81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 09:34:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/015961c75ee24b12991e968cfdb56ac0 : SUCCESS in 19m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6208b6c2b3314140b0c345885aeee715 : SUCCESS in 33m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8bcd1e9c55ce4cc59840fd12aa261d31 : FAILURE in 53m 04s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/418b4429cdbe424aad52c136db5c6321 : SUCCESS in 1h 07m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/700d9e88fbd74a1c9da890b6a052aa67 : SUCCESS in 1h 19m 17s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/efbf6cbd5949469195e3d758f52fafd0 : SUCCESS in 43m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9ad26a670eb44333b9b62b72790ef4e1 : SUCCESS in 37m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5c5c79111ab84f7a94bc2090bbd186be : SUCCESS in 20m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e65c7d85c75f4257847385baf265bd72 : FAILURE in 49m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba4ff3d19dcc4aafb8b7069b8937575b : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0cde39212e4480f9206f5db2c0b298d : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd2b79db6176477d99f65f983629891c : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97c355a2e29c46ec80a6713422f5931a : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b831bf7d2b164184b0b2cf4076daa797 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15a99969111e425ca6544ab258f859ec : SUCCESS in 4m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d2ca4e27b14848b88f6172317c17c69f : SUCCESS in 37m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32262efa5df24db186be3428ee509298 : SUCCESS in 51m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/506fba6745e64721b72faf51083a48e0 : FAILURE in 55m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5f8cff1db6c24a6e8245cd40ef21a428 : SUCCESS in 1h 11m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/35cd521eb81f4a2a94776172641f82ee : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":4},{"id":"a54e7fd2534d8caedf9f2b44e5e50019ddc0d457","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-06-05 08:42:54.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"96cfdd1d83aae2a3c4e0d1877184b76a5836caea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-05 09:24:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d0958386e6f04caba5b84adf8f30647b : SUCCESS in 31m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d268be161fb4ac84870d1d3e843988db00ae7b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 10:09:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9574575b6e944390a58fc286ed806e64 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7983812fe4fe46f5b8d1426df32a8005 : SUCCESS in 31m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/49270ea83ef048148fa62ec909bd83e2 : SUCCESS in 1h 16m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bd8e9cf25e7945bb9daf156d91f41b37 : FAILURE in 49m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/11fab01eb2ab4ad694574b05d3eab2a9 : SUCCESS in 1h 21m 38s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/5d200092b1cb4679894ddae22dacdcd4 : SUCCESS in 50m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c1d1cbf8a04b44de89ee9b02969bd3ec : SUCCESS in 27m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1ab9f61d661043e79930ad79c9b9d063 : SUCCESS in 19m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/62a9038646574863881cdcc67f4dc3a5 : SUCCESS in 1h 04m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84c7afaf25f04ecdaba05db6c7062782 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e7aa5dbddb4b9592955c7ec2a1295d : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68f742ecea2f42ef950f89322b3a5b5b : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e335873609784c0c91f278e9912ec0aa : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7f5967137e411cae62ff78023331e2 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de8938a332444ec1a20474a3846dc664 : SUCCESS in 4m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ffe1e59b2d94a259bc49814a94f6a2a : SUCCESS in 27m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/71ad18d2936a43fb9911114e1439e73f : SUCCESS in 26m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d388aea77574eaf944a45d16a3b6581 : SUCCESS in 1h 06m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/28f2d850977c4cb3aad545c54ca8bb56 : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/741d3f88a754401dbc024b2fbc1dd4de : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":4},{"id":"b294e00d48743f5c12c50748c6688c6600d758e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 10:46:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/661aaaed548d46b7ab9d98b0a6e0a4bd : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef3af431ef524130873df030f8ac3d1f : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7fabc831538e43639730404919f5ff29 : SUCCESS in 1h 57m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6d5b8ed04df43f7bea039830158cb34 : SUCCESS in 1h 05m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"10f43a9dbbbd7bbbec5ad738ad6c296f5b1b6c2e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-05 15:54:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"178614364cce43287ac56ffe299bd82dff4dc5e8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-05 15:55:55.000000000","message":"Patch Set 4:\n\nneeds depends-on to merge before approving/merging","accounts_in_message":[],"_revision_number":4},{"id":"4fef98f22a86b20e04fe4235005e887b3dc73ed1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-06-08 07:24:28.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"06dd434648ad84590ca5189ea546e574f16efc12","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-06-08 09:30:47.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"164d100c35fcfad4417aafb3a84cd9f840c9cbbd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-08 18:19:02.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e717370fab2d3898209fd211bd5c6c2938db7b6e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-10 06:54:30.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e26a0b9178318b9b35017f24efeab8bf229c3461","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 06:54:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"261f0ae77ecfcd8eb688c24272573dd51c2b5115","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 08:31:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4ae03e1a2f3f4fb7960ce2b938f75751 : SUCCESS in 1h 20m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7754edcb9b764371b0e4c33234ac44fe : SUCCESS in 1h 16m 39s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3f639c385b6a437d999291e42619d7b7 : SUCCESS in 1h 22m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70b5f19a80b74f0abe97d7c17c11480a : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d97764ad8d649daa47149a0223bfe37 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/963067c592324f9584e23b37c931de80 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05aa0c7f54de48efb4adb2dcb6d3ac09 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce310e95dee45078ae3313e8fdd9ecf : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f663f793faf344f4a4254df37b09fc88 : SUCCESS in 5m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e077195da2544e6b2218143740b91f4 : SUCCESS in 26m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17e32741ee734ca58d003da098518fc7 : SUCCESS in 1h 35m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0a39d662d14743f0801277c27c309d71 : SUCCESS in 1h 35m 58s","accounts_in_message":[],"_revision_number":4},{"id":"0a7c8a5f7304b48010aec0ee9e587adb4a994384","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 08:31:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2dced29592e289b10a5b307f2637a56a2b75e836","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 08:35:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7795c8560e1e46bea4618fed6136039e : SUCCESS in 2m 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d523856d953445d9d1ab7e9d2ac09a9 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1a41c298cfcfba53956cb73208035f28c0a3d4a0","revisions":{"b897303e2dddaf7e9222e4ed09c3afc30e352f21":{"kind":"REWORK","_number":1,"created":"2020-05-04 17:35:26.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/52/725352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/725352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/725352/1"}}},"commit":{"parents":[{"commit":"fe6fd8dc5d1809f8ce91f677f6affd4e21268551","subject":"debian: binary target is x86 only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe6fd8dc5d1809f8ce91f677f6affd4e21268551"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-05-04 17:34:30.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-05-04 17:34:30.000000000","tz":-240},"subject":"Remove tripleoclient container from Kolla","message":"Remove tripleoclient container from Kolla\n\nThis container was never needed or useful so far; let\u0027s remove it from\nKolla.\n\nDepends-On: https://review.opendev.org/725349\nChange-Id: I2c80080a5b2203c6b22c361f2b6126d65c0c50dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b897303e2dddaf7e9222e4ed09c3afc30e352f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b897303e2dddaf7e9222e4ed09c3afc30e352f21"}]},"branch":"refs/heads/master"},"58bdffa13e571a832463ee0089e02cae3e9f487d":{"kind":"REWORK","_number":2,"created":"2020-05-05 14:13:22.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/52/725352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/725352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/725352/2"}}},"commit":{"parents":[{"commit":"fe6fd8dc5d1809f8ce91f677f6affd4e21268551","subject":"debian: binary target is x86 only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe6fd8dc5d1809f8ce91f677f6affd4e21268551"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-05-04 17:34:30.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-05-05 14:13:17.000000000","tz":-240},"subject":"Remove tripleoclient container from Kolla","message":"Remove tripleoclient container from Kolla\n\nThis container was never needed or useful so far; let\u0027s remove it from\nKolla.\n\nDepends-On: https://review.opendev.org/725349\nChange-Id: I2c80080a5b2203c6b22c361f2b6126d65c0c50dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58bdffa13e571a832463ee0089e02cae3e9f487d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58bdffa13e571a832463ee0089e02cae3e9f487d"}]},"branch":"refs/heads/master"},"0b4410e063095ae6050e376881a3059e7363489e":{"kind":"REWORK","_number":3,"created":"2020-05-05 14:45:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/725352/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/725352/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/725352/3"}}},"commit":{"parents":[{"commit":"fe6fd8dc5d1809f8ce91f677f6affd4e21268551","subject":"debian: binary target is x86 only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe6fd8dc5d1809f8ce91f677f6affd4e21268551"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-05-04 17:34:30.000000000","tz":-240},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-05 14:44:45.000000000","tz":0},"subject":"Remove tripleoclient container from Kolla","message":"Remove tripleoclient container from Kolla\n\nThis container was never needed or useful so far; let\u0027s remove it from\nKolla.\n\nDepends-On: https://review.opendev.org/725349\nChange-Id: I2c80080a5b2203c6b22c361f2b6126d65c0c50dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b4410e063095ae6050e376881a3059e7363489e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b4410e063095ae6050e376881a3059e7363489e"}]},"branch":"refs/heads/master"},"1a41c298cfcfba53956cb73208035f28c0a3d4a0":{"kind":"REWORK","_number":4,"created":"2020-05-29 08:09:11.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/52/725352/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/725352/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/725352/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/725352/4"}}},"commit":{"parents":[{"commit":"834ffd912a130c552cbe1ff9a416dea95599a3e4","subject":"Merge \"Make Cyborg buildable (again)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/834ffd912a130c552cbe1ff9a416dea95599a3e4"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-05-04 17:34:30.000000000","tz":-240},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-05-29 08:08:57.000000000","tz":120},"subject":"Remove tripleoclient container from Kolla","message":"Remove tripleoclient container from Kolla\n\nThis container was never needed or useful so far; let\u0027s remove it from\nKolla.\n\nDepends-On: https://review.opendev.org/725349\nChange-Id: I2c80080a5b2203c6b22c361f2b6126d65c0c50dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a41c298cfcfba53956cb73208035f28c0a3d4a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a41c298cfcfba53956cb73208035f28c0a3d4a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
