)]}'
{"id":"openstack%2Fkolla~725600","triplet_id":"openstack%2Fkolla~master~I53ff892fa7b8847bec232ad454d4a94be7b6f80a","project":"openstack/kolla","branch":"master","topic":"prometheus_upgrade","hashtags":[],"change_id":"I53ff892fa7b8847bec232ad454d4a94be7b6f80a","subject":"WIP: Prometheus upgrade","status":"ABANDONED","created":"2020-05-05 13:53:05.000000000","updated":"2020-12-24 13:52:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"31c40f0873ca951879e1c866e3052521363f8c43","_number":725600,"virtual_id_number":725600,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"date":"2020-05-07 19:18:07.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-05 15:23:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-05 15:49:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2020-05-25 10:47:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2020-05-23 08:47:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 15:23:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-05 15:49:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-23 08:47:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-05-25 10:47:03.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"691e6f11487065161f307aab6f25abf30ee93824","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-05 13:53:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75c49e05bd07b9c7dadc5ac1cc4b2ceed7066d49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 15:23:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1a32ee60ea2846e3ad6e3cd4219f80c6 : SUCCESS in 28m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f5ac07317a5748f887c26384acb3e05b : SUCCESS in 31m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0fc7dfb999ee4b668d2e203ba0ecc1e7 : SUCCESS in 1h 24m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e376b58053bc45a6b33c57433f2cfe23 : SUCCESS in 1h 15m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6c3a9a20a8a54d5c84cac57123325373 : SUCCESS in 1h 26m 08s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/0263312b43e94bcda579e172d8b26e79 : SUCCESS in 47m 52s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6b7efbf7a0664b95a11aa3c562026058 : SUCCESS in 21m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e98ba87652f04719822b89d472eaf887 : SUCCESS in 17m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eb7a7a4e95cc4f43a1a93cb9e0244ce5 : SUCCESS in 1h 10m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c106881eca0f4e8fa73a35007fbbf777 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0ba259f63924b8a84ae3f53eb8ba277 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17ed95d0c9cb4be9b80b114ae1ebdab6 : SUCCESS in 10m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/371b907bb3fe445ab11f64fd55df75dd : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed75f1a796b243fab9078bd1c9a22c98 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f1d11c95be4ca691077dfd5bba39a5 : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8339fb238aa9495c8dda978ea3e53871 : SUCCESS in 32m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b797e7018e14c26a46fbdf9acae30fa : SUCCESS in 25m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ca518e31db94985962f640e0b4f76a5 : SUCCESS in 1h 09m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/845e598e61c846e191d7c9fb0f2af1a4 : SUCCESS in 59m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9de4f95d68544dcbafebf0bad7dd8a7a : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7d73d99993943e6ebd9a5886ebd4d725a8d84d0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 15:43:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f6f6e3eaf53402e9e5fa85f360f59ce : FAILURE in 56m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ce5f6b339ef466899b490e357092c08 : SUCCESS in 51m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b044ac99a384d4ead777192de3290f9 : SUCCESS in 1h 45m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8dd0618f8a51454badf57f28d49280cf : SUCCESS in 51m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0bbca897cf6a7cd1cabf61d3c1a3c17d03442d5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-05 15:49:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/140b98c8a63940efa78568be692b49f9 : SUCCESS in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2aad85be8134f7ed9101113b71d75045b141ade5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-07 17:40:27.000000000","message":"Patch Set 1: Review-Priority-1\n\nfor the next cycle I guess?","accounts_in_message":[],"_revision_number":1},{"id":"3af5e1ec5d83ab5b1d346f907cb6a66cbc72d5ee","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-07 19:18:07.000000000","message":"Patch Set 1:\n\n\u003e for the next cycle I guess?\n\nYes it\u0027s for V.","accounts_in_message":[],"_revision_number":1},{"id":"6dd104ac794c7014f25a1802e75476385dc3f531","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-23 08:44:50.000000000","message":"Patch Set 1:\n\nussuri branched, unfreezing","accounts_in_message":[],"_revision_number":1},{"id":"6bc4c0e41fc0c661b92aa95aa083d8e53826b986","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-23 08:47:00.000000000","message":"Patch Set 1: -Review-Priority","accounts_in_message":[],"_revision_number":1},{"id":"1e4d6711d538990e4c6d2d8bd1fe608f209aeeda","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-25 10:47:03.000000000","message":"Patch Set 1: Code-Review-1\n\nMore information into commit message please. Note if some activity from admin is needed.\n\nNow I do not know should I keep 1.8 version running side by side with 2.x or do one replace other but you want to support using either 1.x or 2.x one.","accounts_in_message":[],"_revision_number":1},{"id":"31c40f0873ca951879e1c866e3052521363f8c43","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-12-24 13:52:16.000000000","message":"Abandoned\n\nNo need after https://review.opendev.org/c/openstack/kolla/+/765916","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc96a8dd627ce2c8782b4296b472a3174972e0a0","revisions":{"cc96a8dd627ce2c8782b4296b472a3174972e0a0":{"kind":"REWORK","_number":1,"created":"2020-05-05 13:53:05.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/00/725600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/725600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/725600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/725600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/725600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/725600/1"}}},"commit":{"parents":[{"commit":"fe6fd8dc5d1809f8ce91f677f6affd4e21268551","subject":"debian: binary target is x86 only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe6fd8dc5d1809f8ce91f677f6affd4e21268551"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-05-05 13:52:12.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-05-05 13:52:12.000000000","tz":180},"subject":"WIP: Prometheus upgrade","message":"WIP: Prometheus upgrade\n\nChange-Id: I53ff892fa7b8847bec232ad454d4a94be7b6f80a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc96a8dd627ce2c8782b4296b472a3174972e0a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc96a8dd627ce2c8782b4296b472a3174972e0a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
