)]}'
{"id":"openstack%2Fkolla~726460","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I25cfa62dbf20490685617b4a4185bf95ab91725e","project":"openstack/kolla","branch":"stable/train","topic":"bug/1873421-stable/train","hashtags":[],"change_id":"I25cfa62dbf20490685617b4a4185bf95ab91725e","subject":"Use Python 3 for mod_wsgi in cyborg and monasca on Debian/Ubuntu","status":"MERGED","created":"2020-05-08 19:13:50.000000000","updated":"2020-05-11 12:31:53.000000000","submitted":"2020-05-11 12:28:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726460-1589200126339-1d1e633c","meta_rev_id":"f4a872eccd95146ce4ac8a726604f75ea64d1517","_number":726460,"virtual_id_number":726460,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 12:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-11 08:11:14.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-05-11 07:39:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-05-08 19:14:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-05-11 07:40:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-05-08 19:13:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 19:13:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-11 07:40:01.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-05-11 08:11:14.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-11 12:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2cbb1a253cf70d8fb0ded0d4ce939524bce98e2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-08 19:13:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"37c63375e45af22f1d2d37038db6522cd8c41121","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-08 19:14:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d40fab8df310f649f8dcd5d73872a81b8b4741ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-08 19:41:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/cd3cb0f5b4f94b60981e94e9d699387f : FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fb7017daab9ffef26b990b4ce2aafa36d5bf6f42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 22:02:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/43c8ac9b041f43b7a7c619e9c501c960 : SUCCESS in 41m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c4940cf240ed42e89c286c7565efeafa : SUCCESS in 54m 46s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/75a88458001f410e90b6472942fe722e : FAILURE in 22m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/975dc4b8eb7c43af8ed9fc9dd35e1748 : SUCCESS in 33m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9901a1cc176541f4b6d09c077dfda879 : SUCCESS in 1h 16m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/77223f6d1bd442b38b64dde2c94aefa3 : SUCCESS in 1h 19m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/105063f5334846a48f8ed56c1448e8aa : SUCCESS in 1h 24m 03s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2c49c68b06f74fdaa46ad3e1d75df60b : SUCCESS in 1h 20m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7fc8101014134664822c5656a00f0c07 : SUCCESS in 1h 37m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/df89a5b1f82a46ebb8c967d3c637dfd8 : FAILURE in 58m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fb0d24c1ea024626ae60a767c1641b66 : SUCCESS in 25m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/179a389048134e0baebd8ba3a1a41d5d : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0b08807c27734689b6b31abcf5cb5219 : SUCCESS in 1h 22m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a42f67cf0224a3dbfaf9e2d181e1f43 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c72e73dd314c1db13f1b46d4917dfd : SUCCESS in 7m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c53af0cb14a40cc905732ad3684b840 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72d7933d2abc449582bfd8405737f11a : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a61f6db0f20e4e4facf4e5a9ee8fc374 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb65c6bd6f042a392b5918ae6bd9b5a : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90bd8fba9f9b4c309a84c44593059e36 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94e6e58fdb9248c29b17823a31b33181 : SUCCESS in 1h 21m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/951a88b4c1f949f9ae1e528ec941680a : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/58d6c4f34aad461fba2341af1e016c16 : SUCCESS in 58m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/346f13e5648e46b2b926babf05516f2c : SUCCESS in 1h 19m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1603b5f14bf143ba849253cfe132ca15 : SUCCESS in 1h 07m 10s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"791762b32c8cca14b596892873fdcf3c320d4272","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-09 08:03:36.000000000","message":"Patch Set 1:\n\nrecheck\n\ntripleo failed to build octavia","accounts_in_message":[],"_revision_number":1},{"id":"189bd26ec4811f6ae08f93ba8af0e1e60bf0b669","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-09 08:31:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/d93d8e7d5f164c289314c969aed72019 : FAILURE in 26m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f72d3ee403a8dd4cb5164df2bce087f793962e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 10:07:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/73267ac13e1b409688ae0395bd2695fc : SUCCESS in 1h 08m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0b4362ee185c42c6b1385330213be2a2 : SUCCESS in 49m 53s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2bf8c02eba5448c98349c3e8ff55d10c : FAILURE in 22m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/62dcf2585e474347a407e1519ee63bb8 : SUCCESS in 29m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/59de147215d74ed5b413c21c48bc7feb : SUCCESS in 1h 32m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/50a9c7fac1c14532acd75f8404202417 : SUCCESS in 1h 24m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3906f6030aef43fc83e43be377424b34 : TIMED_OUT in 2h 02m 11s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2ecc7b16ce074608bcefa96dfda5da5a : TIMED_OUT in 2h 02m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b6c45a845ebb462887690c6b1bd3ed03 : SUCCESS in 1h 43m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/27ee165c74574470ae68420df9f4ea7e : SUCCESS in 1h 42m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d03559d5916a4074b4321103c6929b3a : SUCCESS in 22m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9cb7f4567faa4a25993c1332260f087b : SUCCESS in 21m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bce3a8302d484d8cac14bb1724856b73 : SUCCESS in 1h 23m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b789a57ff80f4e52941f17a149a328a6 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b989cb09d3bf47f1af3d3f3d13e28511 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b29a3325979a41fb9fbf9de0ecf2e314 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18e85dc0f63140acad271d21689fa82e : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f646b786968546dfbfe719a37ac51542 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c013e452a7774b55ad418e2a1cebe26c : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3870bd85c1d44fcf8a8d52b492a4da25 : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c8935cefbf6400e9099ff2a7ffe5fe0 : SUCCESS in 20m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2e102a76a864e7381b8b97c94286519 : SUCCESS in 28m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4b826a1224c40d9bb53a6de2ea51598 : TIMED_OUT in 2h 01m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d087e373206c4971b60e07901e9d034b : SUCCESS in 55m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c9391a9367af4cfb9ba12a00fe45c5e9 : SUCCESS in 1h 36m 48s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"f5f48b44ffb27736a37b503c4ce94db6110619e3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-09 10:16:53.000000000","message":"Patch Set 1:\n\nrecheck\n\nargh, that horizon thingy is annoying","accounts_in_message":[],"_revision_number":1},{"id":"7c0bbd7b21d1a30656dd22a4f84f207a34a63f8a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-09 10:44:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/4f17fe75620d4afca572cf405e1b9dd4 : FAILURE in 26m 41s","accounts_in_message":[],"_revision_number":1},{"id":"52a33df622c5c84d0859d88042b0db1932bef6d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 12:45:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8a1cbe761dc84c548007f8f3ee7fb307 : SUCCESS in 49m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e35611c26b8141529dc02823d76bdc01 : SUCCESS in 2h 26m 51s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4534f48231494231b7fd9e9138ba79c4 : FAILURE in 56m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/21f6f9d00d2a4f1a8e2331b44623679d : SUCCESS in 1h 36m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/656054ac3ac14d859e85031d6d02336d : SUCCESS in 1h 53m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3996cd3fcda245b2a22044a32dfaa4f1 : SUCCESS in 1h 54m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8aecf9e387ed46c791a75afc72c1240c : TIMED_OUT in 2h 01m 54s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a5e62d18c0ed44ac9e987049f25018f9 : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/697c9b1234e94a65b57398fa076a7640 : SUCCESS in 1h 25m 23s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ff0d74da95594125982875946b8b38d7 : TIMED_OUT in 2h 03m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1402472e6e7441a687b63522fba66465 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5fb2f311dff1494fbba6c1fb03191043 : SUCCESS in 1h 22m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8ca76be2dc4b40bf9f7b61bb3b21c130 : TIMED_OUT in 2h 02m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6023b235aa6a48b3b79099c209a36bd9 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd8f1a78d7384e10bea0015dd7b28c42 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/698d8fd7838a4fe684e76d0bcc818b37 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/241667c8a4974a7698e00d6f1497f00d : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfc3f7e258424689b62f4398c8dedcd7 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a592c83aa4ec4c0899314b4e52ef9e78 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aa316d37f1b4da8aec12423a3715835 : SUCCESS in 5m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d35dcd5d06ca4ba39c089bdaa55e94fd : SUCCESS in 2h 15m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f77b06430e8a4c9984b298ee6661f07f : SUCCESS in 27m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee8f9391817244c0a63000598905722f : SUCCESS in 1h 16m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24950ec9e5564f52a475c263716cd4e7 : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2ba5de6a1ef5476a905f7ac1fd07efa1 : SUCCESS in 1h 32m 05s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"2f5838d19b703cbb23ceba4376fa6aee32a87592","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-09 13:33:20.000000000","message":"Patch Set 1:\n\nrecheck\n\nyou must be kidding","accounts_in_message":[],"_revision_number":1},{"id":"d89efdec6aedfdec46ee70ab9b65cb772dbfe2db","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-09 14:06:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/77b3643f5edb46468e1136630b0266b9 : FAILURE in 32m 11s","accounts_in_message":[],"_revision_number":1},{"id":"21188dddc019cbb30df18f047c99ff6dd923cf66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 16:19:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2214c6e47447488d8517a96a1c003255 : SUCCESS in 2h 39m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ab280ceab3f34f149834d3ce485fcaa0 : SUCCESS in 2h 43m 48s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e8309a3689bd459eabd6f93be94a53db : FAILURE in 16m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6cb4b6742b1f450db9f111bbe8710cfd : SUCCESS in 2h 28m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c7f5b82ac3a54388be116bd231f52382 : SUCCESS in 1h 12m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9e9a8f2252ec4585934525b04289ebbb : TIMED_OUT in 2h 02m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/82fd18912563485e9ecbdfe308fca1b1 : SUCCESS in 1h 08m 33s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0876b592af7f4415be89443f7f1dc986 : TIMED_OUT in 2h 02m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d397c5926f88452982a337fa5af2e9e8 : SUCCESS in 1h 29m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c341d71184534548a28adcf88d3c0c4a : SUCCESS in 40m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/083f9382603f49fe98b68e1e7862758f : SUCCESS in 25m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e1e0049bdd39458e86b908217ca66abc : SUCCESS in 17m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/41a44de0d5e94aaaa4eee331de651589 : SUCCESS in 56m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c18c6a27956f4881a942c1490f7c730f : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d6c33f6e2f49f48642b0c527480272 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d8c777a2b974a939f966cc527dc3cb3 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0077b29b9b4a4eb5a12f45fef1ed777c : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6f4c03523a74d928438afb5a3f460cd : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3776af2f3a234f8e8980b84933a81cbb : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92c7b9235ec344808809d4d677784977 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b0bccc8ddc214e5986853d30c7c6dfb2 : SUCCESS in 32m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c40a607ed2d84bb1ba3eb6f78959c600 : SUCCESS in 1h 06m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbc1770704b2459eb7dee25a66525163 : SUCCESS in 58m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a77ec8b769bb4f459157c2725a63adf4 : SUCCESS in 55m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c0735fc12b44e7c90ef5c1a1062ce9d : SUCCESS in 1h 15m 32s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"84282bc62a3c10a0f308c12b7c2156a5c4ea95b6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-11 07:39:52.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c3790ee06046d1ed125ab489597964a12c13fed8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-11 07:40:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d35bf89032ce03d2b49b03f6149c75c8aa3dea9a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 08:11:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/33dd8e179ef54e32992e0b5f73e75490 : FAILURE in 30m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d47582aebffe3cfd075228c89bcebcb76f454473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 10:02:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5ff80816fc5c49f187e723005f08f37a : SUCCESS in 39m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/210ae84d1fa142c9949d01ba7e7d74a9 : SUCCESS in 45m 11s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7eb6043736bc4bff9dbb7fe99213d2e7 : FAILURE in 45m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/66b73f279dd34375a8ad3cc633d75d1e : SUCCESS in 1h 37m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/49f7f66be32f4b62a76981737d3b5394 : SUCCESS in 1h 38m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8806651c18274f7c9a642bf9103cecd6 : SUCCESS in 2h 00m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3723a393af4e4404a2e1a7afea0ac95d : SUCCESS in 1h 34m 30s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/574e5529897342009bb7118cdc2c5484 : SUCCESS in 1h 06m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bd6c1f6c86c94c0c9ea4128c522487b8 : SUCCESS in 2h 18m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ac65ad8328e941d38a04bf5a2d0a0e42 : SUCCESS in 59m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1e9ecf6310764db28414a97574380948 : SUCCESS in 22m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3a2f4bf688dc4acd8c8b8f3cc9908f3c : SUCCESS in 20m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36c3aa0fd20343eebaeaed477a509dbe : SUCCESS in 1h 00m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b49d4d12728f426b89ef1b61bb9c2452 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ef97b0bce74dac86cbfbe48616f92a : SUCCESS in 6m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/441c4f54e9874cbdb9c5e0fa4b6fc84f : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39ff7688eed0463db8b6d096fc91e920 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b7b630eaf114b7fbdab442095f0b8a0 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43420b364102415f8d1b3809088de84d : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad0959ef49cd471abc2f33a21d7aced4 : SUCCESS in 4m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bebea1c9f5704581a6173e1664e20890 : SUCCESS in 33m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8792382e1dc42a580a5d0855b8a7282 : SUCCESS in 33m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7261d3b8fec940bbb3bab0436cab1178 : SUCCESS in 56m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/21550dbef2e54704a1df9d82a525d028 : SUCCESS in 57m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c713aa1c7211410c90a3bd09e736e1f4 : SUCCESS in 1h 04m 34s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"8cc5203d89a54baf796c5dec4d2220cb36dc146f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 10:02:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23559fd55a349032bd14821d37ea0dfbb81f0c09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:28:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3da14a8dc5a293f5fea81d10a17895d6655f45b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:28:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/81d6eddfd9fa42b79a1b88c6f10c3e54 : SUCCESS in 56m 06s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5be5abe31b10469bb73eb1be3987046c : SUCCESS in 47m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/64e5202ce4454aacbddbadb5a916928c : SUCCESS in 1h 30m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f862fc44b1654290abb2561b3670d896 : SUCCESS in 1h 21m 35s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/171e6e4f587041d89b8dee8974171228 : SUCCESS in 2h 20m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9a1a104b18fa4776a35601b4136cef44 : SUCCESS in 55m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7132bd3e4dae45b59d17c408365cf8b9 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a22f67828248fab04bb4a5f1b3ec72 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/320fe12a053749e884705c8a519f8a6d : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b610e348fc347238f541e7084e22f13 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a03d6a91cf984414bd127116dddd9d3e : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6786fff4592345659e35c79283ee6c53 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c766276639d42a2a7e438dee7415734 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dbb28b506bed4ba5affe64d68c32e1e5 : SUCCESS in 29m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5cf8035a89e6436fb14353abe4b6cea3 : SUCCESS in 1h 04m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/754fc728c1054826979387be6ee8e075 : SUCCESS in 1h 21m 32s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"29b7066849217ed038d34ced40b3dffe0d8630da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:31:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488408d311a44b238184ec3b95b689d7 : SUCCESS in 2m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cc7206ad7b74aeda1b8adb57ac916b3 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f221b4f7ab7745ee20c8e97d6a1a2d3c7e74f466","revisions":{"f221b4f7ab7745ee20c8e97d6a1a2d3c7e74f466":{"kind":"REWORK","_number":1,"created":"2020-05-08 19:13:50.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/60/726460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/726460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/726460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/726460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/726460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/726460/1"}}},"commit":{"parents":[{"commit":"4c8863b0a877a00a810dab55be818163aeb83c61","subject":"Bump versions for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8863b0a877a00a810dab55be818163aeb83c61"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-04 09:32:47.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-08 19:13:50.000000000","tz":0},"subject":"Use Python 3 for mod_wsgi in cyborg and monasca on Debian/Ubuntu","message":"Use Python 3 for mod_wsgi in cyborg and monasca on Debian/Ubuntu\n\nIn Train kolla switched to Python 3 in Debian and Ubuntu source images.\nFor services running under httpd with mod_wsgi, they should install\nlibapache2-mod-wsgi-py3 rather than libapache2-mod-wsgi. This was done\nfor most images, but cyborg and monasca were omitted.  This change fixes\nthat.\n\nChange-Id: I25cfa62dbf20490685617b4a4185bf95ab91725e\nCloses-Bug: #1873421\n(cherry picked from commit f5d31da12fa66cedfc2e270b8cd3be8e2bb9254b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f221b4f7ab7745ee20c8e97d6a1a2d3c7e74f466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f221b4f7ab7745ee20c8e97d6a1a2d3c7e74f466"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
