)]}'
{"id":"openstack%2Fkolla~726884","triplet_id":"openstack%2Fkolla~master~I121e2ecb30ec6e8d1b6d88d9b921894438c3e094","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I121e2ecb30ec6e8d1b6d88d9b921894438c3e094","subject":"base/centos: take care of whitespaces","status":"MERGED","created":"2020-05-11 15:31:37.000000000","updated":"2020-06-17 04:30:15.000000000","submitted":"2020-06-17 04:28:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726884-1592368130945-ff50f346","meta_rev_id":"b8ded8020da0468f9491b0568261de8a40222363","_number":726884,"virtual_id_number":726884,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-06-16 21:53:43.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 04:28:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-16 23:32:42.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-05-18 06:47:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-05-20 03:19:16.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-05-20 03:19:16.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 15:38:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-11 15:38:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-05-18 06:47:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-05-20 03:19:16.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-06-16 23:32:42.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 04:28:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"823fe2205732aaf564295526b162d9a8e6839787","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-11 15:31:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaf8ecf548c24e815e3fed8a718cea94a23981fd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-11 15:56:29.000000000","message":"Patch Set 1:\n\nthat\u0027s the templating hell I was talking about ;-)","accounts_in_message":[],"_revision_number":1},{"id":"f1e93beff688b77eefc57364bfe96f5720fe54ef","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 17:23:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/eb17103c227b405fad3de475694f00af : SUCCESS in 28m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25da9067514ef1aae2addb1608054a0fee2bd1a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:34:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5507b6b47fd74b35a16d0846d2477988 : FAILURE in 5m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/891726fa57ef4620a191466517479b52 : SUCCESS in 1h 00m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/738151d0b04a4196948f82f56f1aed41 : SUCCESS in 1h 58m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a151e188db1e470d85843c9be27f34bd : SUCCESS in 58m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72b39ce05202df641f58d848f1bbf97e0160ab9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:40:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/06201be8993948a0a7ca0a4dd5e41cd4 : FAILURE in 7m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8521c9b115fd4a38a099000e71373df3 : FAILURE in 6m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f664221d82f64cfe8c836643326fbecc : FAILURE in 10m 33s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7e8aafeb510e461699b0d6166f10ab07 : FAILURE in 9m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9aa93cf86ff64a77a6c544feba2b7fd0 : FAILURE in 12m 27s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/dbab7089eb9d468ba3b7ce9113fdef71 : SUCCESS in 48m 23s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/842fb7adc9ac4d9d94fd8461dec5c2f2 : FAILURE in 12m 19s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2f2e5f9910c74d059ec886666b5d72f2 : FAILURE in 13m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5518466af0564366b1ba3abd08028b24 : FAILURE in 11m 27s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a1e858c4be874ef68fdd0929f5c35612 : FAILURE in 16m 04s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a054c4de8190404182c18b542f2026b7 : FAILURE in 15m 34s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/0c385cec39c245cf8a94295e220c9b7e : FAILURE in 14m 05s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/652956eaf9364f8182ffae02a9e3feca : FAILURE in 13m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1d8579da8a644f3887d5a78fc6ee28b5 : SUCCESS in 22m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a0c693ec754d4acab76198b48c6e1d94 : SUCCESS in 22m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ddd334086cb4ae18c779534c0c9feca : SUCCESS in 1h 16m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d88d2a5dc28446cdbdc95bd97e76aec7 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae6440411e124bed8f1e1babcdf08bce : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81fd2d099b4446aebd7dbdc2393c1234 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f25ffa5057f443d7b26af72d3e884774 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fca2b47cbf9c4c90822072c0da4300d2 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d5919b03ac48fd92fc4cb774714fcb : SUCCESS in 7m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27f6e247e3764ae3a50b065bade4c3e3 : SUCCESS in 33m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7359ee3a385f4bba9be9bcc8f9d9711d : SUCCESS in 37m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aee97ea1c0444727acf448e3529115b7 : SUCCESS in 1h 18m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bfcdf32b33f84ce798c18842c8b0dcd6 : SUCCESS in 1h 04m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7c90fd4e592b4ed68edee1c09a2af91e : SUCCESS in 1h 08m 55s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b5e1ec3ed0244608a0ab9fe375305e7a : SUCCESS in 1h 54m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b09f52c542864d07b9c5ed7d89ea539c : TIMED_OUT in 2h 48m 48s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6fc4df46bf21472187335d7c6cddaf26 : SUCCESS in 1h 38m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0bf11bd4988d469394363a293fdf78e9 : SUCCESS in 1h 06m 35s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/f226312d214a41b58a6240b0b88e8884 : SUCCESS in 1h 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9115d0136f5c89ed09a6cca3228703b5c2ff70f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-17 11:19:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60afdb0f2c0877f2776373c8a15c878110a8d2bf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-17 11:46:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/503786bcedfc47768aeffee042886a40 : SUCCESS in 25m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bd7373913f8a3bea90def9b96973d373bd68750","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 11:48:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc0967c296594178884160c93fa97968 : FAILURE in 5m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c19ef5f10c84ecdaff495895d8fdb90 : FAILURE in 17m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/abe473f60abf40278a11eb85f0991172 : FAILURE in 24m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f09dd1b344cf4d37b6dc380f22aac6a6 : FAILURE in 18m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af66fb5141c2057b412a2eadf70ae662362d7e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 13:07:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7c35d717f0f84e0487a56e10df191aad : FAILURE in 4m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e37b0024bd0c407fa29dde19537f09cc : FAILURE in 6m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0020f643685543d1a68b757bb43400bb : FAILURE in 11m 47s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ff2667ef10f74ddb8e5e24d7d6442c4b : FAILURE in 18m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/603e0bec73304558a93d5ee731b6ebb6 : FAILURE in 10m 29s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/15d5f21d11584ced8af69483b04d9702 : SUCCESS in 39m 54s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/3b337eb0863a4ac18df51e29e8049ff3 : FAILURE in 9m 20s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f66c5d9f36354df9a42b6114576ef5a6 : FAILURE in 10m 40s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/65b606280ced4766b49008d673d78728 : FAILURE in 11m 14s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a8c2baea489749d18e6c375cd993c3da : FAILURE in 10m 15s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/e2d369e30d2b4859bd1efb3dfe0e0d33 : FAILURE in 10m 41s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/17dfe163fa704485ba1ca52934dc0962 : FAILURE in 10m 09s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/15a8269531854e21af31657934144052 : FAILURE in 10m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7b2008a75abf4a8388c69737eaa9deeb : SUCCESS in 19m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cb6b9227e4784e6ea5d35a05265798a4 : SUCCESS in 16m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2bbbf91d104943e3802717fd7d1a5d99 : SUCCESS in 1h 01m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29bd584441524b96924b5b57ee50dec2 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f547781d344669901854fb83ddf039 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30618d860ba0400ea3a2bd01dda88412 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/448499f8cba9425a8683861d84935860 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64812f79af0c4260bb0f9327120b27e1 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce9390769dd4f78ae00813562ed397b : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/90a56ce5580246ee84523939f166e08e : SUCCESS in 18m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f62788f6d414fe6b2cad34877370817 : SUCCESS in 23m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0342c14343ce4e4dba24661319025b54 : SUCCESS in 1h 00m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d48cc8406554495c9e6593641d017d5d : SUCCESS in 56m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5a350fd4efe4fd3a3a43aaa68d744dd : SUCCESS in 1h 11m 56s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/addf04a43d5e4adaa87b6e61c42df081 : SUCCESS in 1h 26m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/c013e59274784f11ac745cfbe286e837 : SUCCESS in 1h 41m 27s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/1dfff000f72a4fd19c7204438d502fbb : SUCCESS in 1h 32m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3f4e80a5b6184196b700bd99a349d85d : SUCCESS in 59m 08s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/6d7f4e515bd5441abed29311b12cf7de : SUCCESS in 1h 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bb0e2e72e7f6c10706fe2eb7f5644c1c61a978d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-17 13:21:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41e8f3419dab06ae54317c4ba4f5983dbf452173","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 13:45:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2619667d3c584d3e8af6ec81bd9c0db4 : FAILURE in 8m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f693b61551b4e54b8ac0e5ac33db6a9 : FAILURE in 18m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/26a063ffe95549cfa2e6c73607aa6c20 : FAILURE in 19m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/191200eda1614c36810982890515291f : FAILURE in 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92cc71660db3d70da01d787f5a6d2582a37a7397","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-17 13:48:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d18a3c627bad484b88c29ed7a7021553 : SUCCESS in 26m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"733cf69980882defb2741107c42a1a3ebda9d28b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-17 15:38:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd65a32304bebf12a23baa6897919c8f14ece4d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 16:02:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/63d15fb11d894e3f99b8ca58e598164b : FAILURE in 14m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ebcebcfa84446ecb91c2fea2cce845c : FAILURE in 18m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd0f57a1883a4cd194831479ba5003a8 : FAILURE in 20m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/48ac44b82e264c568178d90486b1054f : FAILURE in 20m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fea4d01937105ed743fa1518519323b0f7b0d2ce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-17 16:05:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/f1f85a0cca1940e397933d303bfa81bd : SUCCESS in 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d5159d8e53349214b64d0baefc596a011b6277c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 17:35:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9dc5cf06380448e4a4937e7c42b2abf4 : SUCCESS in 24m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/bd2c25e0efe144e5933377d0f889f4df : SUCCESS in 32m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/74e39e1636964421880997147908976d : SUCCESS in 1h 11m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cd4edbfaf7794731b7f4496703690554 : SUCCESS in 1h 16m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a6c7b501549646c38fbfcd42d8ae8ca4 : SUCCESS in 1h 19m 19s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2db00b4cac964d8abb53010c8731fdeb : SUCCESS in 39m 32s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/9c5e5b41e8df4fab933edcd63927fa8b : FAILURE in 12m 08s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/cf6ba5c35a7f4c419288bf4e2952d97c : SUCCESS in 1h 48m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/09545b78c41f4fa6a37ea263a662fba6 : FAILURE in 16m 15s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/3d195e81b8d84e0c91c325a471abbd53 : SUCCESS in 1h 47m 19s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6a31c4599221435caf85eb5a7679d446 : FAILURE in 14m 54s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/efbcac4c03534b92b8a9db011d80bfd4 : SUCCESS in 1h 11m 06s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/01373b476a5b46ba88c524aa94667beb : SUCCESS in 1h 40m 57s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/923c2e1ee3ce4b49b734a33a6a578ec4 : SUCCESS in 19m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6baca509a5d6465bbe381bec2f2e790e : SUCCESS in 19m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0d08f1be3afb4faf82a1e973b4698cc7 : SUCCESS in 56m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34bbb937b9804ea49f91babb492c7d86 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f3a8284706a456cbbf0b65eb48afccc : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4143f80398e5401cba0c9e46fe460954 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54b9bd7cbd3d46f2af68495bfad403d6 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a66c5f29135a4c029f4bcd6243b44f55 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3496a811580547468626d0c1f7312bc4 : SUCCESS in 4m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/44ac7771739d4e6fbc864a0aec81c3f5 : SUCCESS in 20m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f1481d4b7024266b2e6bf9e3a1ca5ac : SUCCESS in 24m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f3ba33ed34af4b48ae916da20edd878c : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aba3558dc29645f7965cc0b1ee597b49 : SUCCESS in 58m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aef2aa9fb3c645f38d4ef18182298f05 : SUCCESS in 1h 07m 47s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/18021a00676d48a2886d065c76d43b8f : SUCCESS in 1h 33m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/da9de5c7348446c2af690d8d791f7f6c : SUCCESS in 1h 51m 19s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/938af41cd6914d1eb6ecc8b1cd041ea6 : SUCCESS in 1h 31m 25s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/dd7ed9a6b47841c7a3f7e818d1d4e2bb : SUCCESS in 1h 00m 53s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/e9d01244fc64416ea8031aab7923ced1 : SUCCESS in 1h 24m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6a05816ad2c9b8d97bccb086d1061bdc2dc789b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-18 06:47:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e1b63608d009469e6fea002c715399accdb57002","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-20 03:19:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"295b337d6fb4707293cc110713ca8ba70b03f806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 03:19:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"50716f542a05450aeaeaaa23171e6a23999b9432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 04:45:29.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6f825b82bb9141beb7488fcb770cc367 : SUCCESS in 27m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/79c0c3d344e443b895654636279e9d3f : FAILURE in 13m 27s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/41cd9fbbc5aa41e695bcea96494f4eda : SUCCESS in 1h 21m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1494696fac01401894386778729b04d0 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802196ff829b46388e710008f97ff72d : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c2e5b711d344817ad8a5994ba4e3a69 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52673489d9494e2ab8524ac17f5d4a94 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348be00dce7d48b5b80511efbef0bc5d : SUCCESS in 6m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9c6421b0ec64bf098ae9d0990c746ee : SUCCESS in 26m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a9139fb026af401eb9b5dfdd9394c674 : SUCCESS in 1h 00m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7e3d1b76d3254802bb88b602700b537a : SUCCESS in 1h 10m 41s","accounts_in_message":[],"_revision_number":3},{"id":"77f362b9b4112ad34132f27970fffbba3e59d66a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-20 06:57:20.000000000","message":"Patch Set 3:\n\nlol, failed at:\nINFO:kolla.common.utils.rabbitmq:  - nothing provides libSDL2-2.0.so.0()(64bit) needed by wxGTK3-3.0.4-15.el8.x86_64\n\nit looks like the erlang we use for centos8 rmq includes graphical toolkit...","accounts_in_message":[],"_revision_number":3},{"id":"357100e1f05c05b830d94b4a1a017a14cf58f9a9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-20 06:57:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"866ab4d732be31c2763f2eb23c5a09d74979b46c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-20 07:29:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/eca7b5563c6f415180273cc7ba13cfd8 : SUCCESS in 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"abf9776e6a8324ad33cbce7152557624e95dbabe","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-20 08:25:36.000000000","message":"Patch Set 3:\n\n\u003e lol, failed at:\n \u003e INFO:kolla.common.utils.rabbitmq:  - nothing provides\n \u003e libSDL2-2.0.so.0()(64bit) needed by wxGTK3-3.0.4-15.el8.x86_64\n \u003e \n \u003e it looks like the erlang we use for centos8 rmq includes graphical\n \u003e toolkit...\n\nerlang depending on wxGTK is one of reasons why we wait for 8.2 with AArch64","accounts_in_message":[],"_revision_number":3},{"id":"4ebee682be5ffffe0d29e0449603a80751364f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 09:05:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/653306f28f0841cd95df536f6d9fc1d7 : SUCCESS in 26m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/598db3c7ab50480486ace84293944ff8 : SUCCESS in 31m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/eae448822caa44009df29a44ca756507 : FAILURE in 15m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/086d37113a584f5a811b4bb892ddf7e1 : FAILURE in 18m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/11fd4f0af46a4ca0b9cee6e2ea85b375 : FAILURE in 15m 19s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c547d56927634fcb9dec86b511527031 : FAILURE in 1h 45m 16s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/7245b741c33d41c1b2f083e09ebd3ab2 : SUCCESS in 29m 02s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1ac697d6461d4550827a3cc428aba87e : FAILURE in 19m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a8574a3a857e4c83b730bd5800b8d2e3 : FAILURE in 31m 46s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f684124123cf406db7b00c27d40e7207 : SUCCESS in 1h 57m 57s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/842059bd2f014403a4568b7a1158a54f : SUCCESS in 1h 47m 40s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/fcb73018fe464d6590d3298d4a988029 : FAILURE in 15m 04s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/48245df9c45a4ba0bf2408d3451f4d02 : SUCCESS in 1h 38m 59s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/81dd1849fd274f0591e35e4dd80f490d : SUCCESS in 20m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8e50d8ece0ba4efc8a8664c6c98b7e00 : SUCCESS in 32m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc747457d9c849a487816d89db200242 : SUCCESS in 1h 22m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c97843f23ea474aa827b08a1741c547 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e442973e944ec18cdfadcdf93c03d3 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/936d51ce41644d85ab7acf591c016529 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c0e028322124b8fa4ff377294e69f23 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c1d61ced51449439e6a0c4efd4abc79 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696a4ca48ed6444cbf055bebb7dfc1a0 : SUCCESS in 11m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a70fbf655c324a4aa018db823fe89a48 : SUCCESS in 40m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/694396e239274005bcfa259e6b77c1ac : SUCCESS in 1h 37m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba64b6de9e6f412099b221c0d5b8734c : SUCCESS in 1h 04m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/648ff048a2e648f0bacf576c998b15a4 : SUCCESS in 1h 01m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da082feb96db48b190e837459f960856 : SUCCESS in 1h 08m 04s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8b5f74a0b7c54b14abcf3ec7b23dd2f0 : SUCCESS in 1h 39m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/532a2b9a42314db3ab6e26ddd39b2bfe : SUCCESS in 1h 49m 08s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d6a7ca1adbb94ffcb89c698451440ba5 : SUCCESS in 1h 44m 26s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/6e4f8d6c00894001962756bb078d0e44 : SUCCESS in 1h 07m 45s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/8844cda6966145c5a4f903f51a6f2203 : SUCCESS in 1h 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45ad5a60cc9d4708f3001dd42a49fcdad1cdbc42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 09:20:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9755ef76aa0a4aab87863233c851feee : SUCCESS in 1h 24m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/deb8bf5bbba94171af4359a3f1e1efbc : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3372249de218458ca1ff5935cd25e46c : SUCCESS in 2h 18m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2854dfef013d4dc994e4356bdaaad4af : FAILURE in 1h 19m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9f81bcd13354db4449082e0ca283d9ab57ae3221","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-24 13:12:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"faa681c49b09fdb389c20c8136e40284a5794cd2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 13:38:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e3857f2b6a56462ba546f99bb69a377f : SUCCESS in 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3fbe26dc01bf4c850f6fda6d9e8519132803cca6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 14:59:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/afd97c7c4fce45e9a72faa3125c6064e : FAILURE in 6m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/01026caa79554365b5d6a8d8f9b93041 : SUCCESS in 47m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbccf36cf9004d7bb4dae3fba651717c : SUCCESS in 1h 39m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0950818f913844c6bc956a5c6cb6ad75 : FAILURE in 51m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1a85c0e20da31a0313d3f3eb37ec528ba3642503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 15:18:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5df7d2bcb7164d95817d283af7890354 : FAILURE in 37m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/283ce96d719142ba8a82df1147a4892e : SUCCESS in 27m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/076df8ab866d495d993d7033e1bb792e : SUCCESS in 1h 26m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e57c72af1ca24b66a7d6832954c7b1ef : SUCCESS in 1h 16m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6a82ff65e0af427b9e8328207f044690 : SUCCESS in 1h 26m 13s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b20ff4ec104a445ea0d42549dd81ec0d : SUCCESS in 34m 20s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/ae296529e6344fe5af7d60bfcc3e2638 : SUCCESS in 32m 28s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/30ec94f8147a49a1b91344b73254121b : FAILURE in 14m 39s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7772e6bfe31b44a1a1cf57187b330698 : FAILURE in 14m 49s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/5ca4d7bac15a466da018dd309f5fabf7 : SUCCESS in 2h 05m 00s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c18828d7b8ad45a9b73cc8234db0b11a : SUCCESS in 1h 58m 41s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/273c24c7f610454f9f58ad5f2ba92ae1 : FAILURE in 14m 21s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/cf4d4cbab85f48bf960dacc7f8b7d604 : FAILURE in 20m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b5f421788be04328aa96fc49e69e71ef : SUCCESS in 21m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4587098818f94b168cdda4dd83067c71 : SUCCESS in 20m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/94999c6d219f45f48608933e66f2ca67 : SUCCESS in 58m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e87f3b8842a3497fb1eb06229a1908ae : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc9e81c65f24a54bc8cf07e5158c9d1 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1397be6f5f54cddaf447ef473c5ad78 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e39eb360b8e45419d1f3c18822140da : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f32f847a4344918870b708a8b21b0b2 : SUCCESS in 6m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f75d8f22a784b7382911c69a24d3d68 : SUCCESS in 21m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c13eea9a4de4fb28f71f21007923e36 : SUCCESS in 26m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b04807d26b4d418fb0355166afdf55ff : SUCCESS in 59m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc2f0a9ffb5b43c9909a601306028734 : SUCCESS in 54m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d28cbd92c57f49f7bc8fd3d221a37687 : SUCCESS in 1h 13m 25s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/34795400eb0b49df89294771abfae5b9 : SUCCESS in 1h 39m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3477baa9a1f049bf91dfa70c0c9b0672 : SUCCESS in 1h 44m 33s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/05df7a6fc253480c901cfb30d9fc7fce : SUCCESS in 1h 32m 32s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/13472919d6584dc0a78d715250ef6a47 : SUCCESS in 1h 14m 15s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/379153de54324793b74b642ac7b6b615 : SUCCESS in 1h 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f483f98f3a7226b25f46283eb44ef1598a081a04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 15:18:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"87a5a311332eb1ad822481f62baab8032ed200e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 16:32:07.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/82a6dd07e9444a5cb3dc070e838f9fa2 : SUCCESS in 26m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f8d78a910891483d8cd19d0d3582e0df : FAILURE in 13m 23s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/24f11ec10584482f9916050fa68bdde3 : FAILURE in 17m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ff1537a559049f9ae2abb76887ef075 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c543376f36e3432a91b9d5104370d881 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd674aa510c8425c902ea8f85f4d5b7a : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4bc11bf56794db9864f5599573c421e : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ad841768d14be3a65ac82218de5052 : SUCCESS in 6m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b45a91fbb81487fa932eaacd9f26830 : SUCCESS in 22m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/20722f34672e4dc5ad7e768263661ad8 : SUCCESS in 1h 00m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9fa9417bfb0b4d7f9bebd99b6d8bd20d : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":3},{"id":"430c4925515a54fa2f3b860a18d280fd72f50f63","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-25 08:58:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9bba10d982554b56ff1de03594a3ea4e42dd9a92","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-25 09:36:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/2046225fe11a4c58b1bb388a88f99448 : SUCCESS in 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba85a3bc8c19a67ad2b60b779c67eedf6be0dc65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 09:55:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/81f9a5538a4944249edbfddcccb63cca : FAILURE in 21m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/99fde416f64642e5b424cf0df216a9e5 : RETRY_LIMIT in 2m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b1e33329a1340a28b54dccd438766bd : RETRY_LIMIT in 4m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e0b5bb1a54ba489f97edbdf912ef8174 : FAILURE in 56m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b04c8307ca88f695f5fa2799863b4387a72280c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 11:29:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c0228ee1f0aa4dd4b923d98fa15c24f3 : SUCCESS in 26m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/11d216378b3e43e394ce0fd328d8223b : SUCCESS in 35m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/99aca13b83ed434994200f21b2258b73 : SUCCESS in 1h 30m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a0670f4f03c44c0f92a2cae0cb98a886 : SUCCESS in 1h 17m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/31c797a5fb1e420f8a980e77cbd2e819 : SUCCESS in 1h 35m 55s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ca6d5cc3fbc5402dbbc470ee23ac7092 : SUCCESS in 43m 28s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/4d2c911ecd7d46baba9485e2aa46fabb : FAILURE in 13m 07s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e87533df1c084b46a98d32c6c8eb389d : FAILURE in 16m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/905afe79db454e0693866d4541008cdb : FAILURE in 17m 03s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/5f8dcd14c64d49eb807765aee1d554bf : FAILURE in 18m 55s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a5c2943b2b924dae8f22d003b7e19517 : SUCCESS in 1h 47m 10s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/f47143237f0c4b12922656706b9e00f0 : FAILURE in 14m 48s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/3c031d8bea5345b39f2ed99564843457 : SUCCESS in 1h 37m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/baf81a2cf7694b4ba0f260ba65575f1a : SUCCESS in 21m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7af3429f6f03419ca3286c66f8b0c8eb : SUCCESS in 18m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c9a9934cc6514f72b902a91671e800bb : SUCCESS in 1h 05m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1555e32a04de43cabd9f14d09d8cac58 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ecdc67144644d896cd40267f550adc : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acf8376870ec40bfbb9b6cbf37e2fe33 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d25246871f874b29aaf074ae36cbcbef : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78dd5ef8833045279ea6415fe2a20e20 : SUCCESS in 12m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9220087883a647b88fd664562c91ce84 : SUCCESS in 41m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fd583f12da554d7d9729ed1c29b9fb43 : SUCCESS in 54m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8cfcee6d26e94006a2c1f82e3bf942f8 : SUCCESS in 59m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b43e56819279452f9aa6b110b1310342 : SUCCESS in 1h 02m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cfbf4421ade64f8d96ddc594ddbd8f5e : SUCCESS in 1h 03m 54s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1e8703dfadb44ee38e6e1f7662bb894a : SUCCESS in 1h 38m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b4cf0365ad7a4041ad55b18d41bcf3d6 : SUCCESS in 2h 01m 30s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/692c691c8c974fa7b332abfe7199c18d : SUCCESS in 1h 44m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ffb404213ca5473a908aeb303087cf17 : SUCCESS in 1h 06m 50s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/6f17797a4d7b4194893fa2688a6292ed : SUCCESS in 1h 32m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40d17114465540b9e60515baaf35b466cea4ea2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 11:29:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2434151c56278f86e851d8967e09a1103c13d1cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 13:17:01.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/afdde5974c494c34b8ed260eb0161a8a : FAILURE in 16m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2a5c9db150f3479bafcb22ea8fc60208 : SUCCESS in 1h 11m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/11a27b42321e4c36ad1e040ef6bfce50 : SUCCESS in 1h 21m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2129fb1989e24b0cbf5063df7fac7694 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f058a915cc4d4687bb0c67ad14625d68 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4296605dfb374b35ac8dce5e677b1a25 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5552ceda41b447fa8adf71bc59c24856 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033a05a0766446adb475d173b937a37b : SUCCESS in 7m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/90e47e3638f74f92b074ab762309f120 : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e4cdcd9133f4974a5557af53df230a4 : SUCCESS in 1h 41m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/65d8e296720445ca8299455526585cb9 : SUCCESS in 1h 46m 25s","accounts_in_message":[],"_revision_number":3},{"id":"4d6999ac68f9fca7857841cfe650d856014e34cc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-25 14:05:35.000000000","message":"Patch Set 3:\n\nrecheck\n\ndear gate, please accept this patch, your faithful -","accounts_in_message":[],"_revision_number":3},{"id":"733eb2ed2643c0a7167873ebc4d60bda8a5041e7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-25 14:34:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/5de352235bea4b6fa998ddcea4800cb1 : SUCCESS in 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9ba8017a0dbea40171d99c2339f91554312118b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 15:52:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d25f3baeb2f040e2bbcf5a6ac5650340 : SUCCESS in 57m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7dff682ac7f5466ea7e0ce5d1f90a1f6 : SUCCESS in 48m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d5619141ed341158efed512bb8f97e6 : SUCCESS in 1h 42m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7436246c3b1e4e21a85213dc2dfae7b1 : FAILURE in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4f7342ffbfa164a4aaa3052ed8f7015c3083906f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 16:14:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ed689953e6b54f7bbc187bbff115db3a : SUCCESS in 24m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2e83ff8634264eef9198129f1157497f : FAILURE in 17m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8b255dc0fdd3413abb9784523dad734b : FAILURE in 13m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c12968e73599496a8a4039b2e670202a : FAILURE in 21m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/703d8cc91dc740b3bbe48f253092551a : FAILURE in 16m 17s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/aea664a0f3094d19abb8382f94eaf069 : SUCCESS in 48m 22s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/70e9e12bf1bc4d3babaf5affd67bddf5 : FAILURE in 13m 15s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2c63e42fecbe4b24950873a534a5b9a0 : FAILURE in 14m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8bf4b79c73be4986a25c3fa6e9d163b6 : FAILURE in 28m 08s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/6bd3e58c659c4b81828e80d03f3af68f : SUCCESS in 1h 47m 46s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/339640e0ba7a41829acd9da0b58b708d : FAILURE in 18m 13s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/abab56f7788a40279322eb844bbef2c1 : SUCCESS in 1h 18m 12s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/83d0d53a765443fab7dcf099fe8f251a : SUCCESS in 1h 45m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/db266220e03745fe9cbedf9ef4f23992 : SUCCESS in 19m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/25e323414fc54a56bea4d2c9f1c3a399 : SUCCESS in 17m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/068610e76a44487a9135f16ce452e211 : SUCCESS in 1h 10m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7d54d6210074d34973ed7616e3e825a : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b0f20dc7084f548d4cdd4ad8cf240a : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5145ef356db04738aefd107a8d0594fc : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95ee3697e9c54d91a978a7fed0aa3760 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/229a61ecfc3041f9be87cb2902f0f65d : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b529bac42a94651acdb5c70368a162e : SUCCESS in 53m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1468372159e2488ead98cf44d2b91988 : SUCCESS in 24m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c68e363194a41e7a46430df950a0909 : SUCCESS in 1h 04m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/001730efe7bf4d0d89b5c81a829aa9dd : SUCCESS in 1h 12m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b24263caa9554587803967c738076a09 : SUCCESS in 1h 08m 31s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/3c5d636c31c04874ac10dfc919042895 : SUCCESS in 1h 38m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b4b6c9f9e0cf4fe9a67656ae66c97d3a : SUCCESS in 1h 45m 22s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/51eca67e1ca04737a48c82c5a0558162 : SUCCESS in 2h 08m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/10c50a5c01b2477ebd60dc97bfd30c24 : SUCCESS in 1h 01m 04s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/bd26282d2ecc4ecf9146fc71470d0c45 : SUCCESS in 1h 28m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"39a9b273fc03bd17e94a92e8dae3071e77213ff8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-04 08:01:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2c367e35eb22ab5f243f9749d94653f37e3c226d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:55:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6cf68513d4a745b29da932ef9fa6a5cc : FAILURE in 21m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5c7500d1c1504b61ad1dc4b4df93a123 : SUCCESS in 28m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7750579479bc46b79b00e0a8db8e7b6e : SUCCESS in 1h 25m 13s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7fe8737764d4475887386f56139553e3 : FAILURE in 18m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6057d9b3bb654b65912b299aee0646a2 : FAILURE in 13m 07s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b87ce15460a9402b8975abfedde2d844 : SUCCESS in 36m 00s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/f54ba12698154fb2a2f96019f2724f84 : FAILURE in 13m 24s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b11153032c2f43c5b37ce4bd26a3984c : FAILURE in 16m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/4faab6e4e4fc4d08a0b10b49ce735afb : SUCCESS in 1h 30m 29s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f376e9789f7e4f588878570895d291f5 : SUCCESS in 1h 44m 19s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a6fc1edaae274a3bae9b2d0059741625 : SUCCESS in 1h 20m 05s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/c9af0ef1053349ba86a000cd07f75142 : FAILURE in 15m 53s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/3697d17a33fc47a2aa4d026c237ea2ab : SUCCESS in 1h 48m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8d40d54559ca4efd9670c43a0f63823c : SUCCESS in 24m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ed0ae390934e4e34a41c4b092fdaeadc : SUCCESS in 16m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b9e7f259c1654839943c2eda4e4d5413 : SUCCESS in 49m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/999b4d2756924503bc066620568997f7 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201e03b809fb4680b51eaff7e99f5b1b : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e3d72c9c4a47479a4d9866b8fd420c : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bce9f0c4c614001bbed617ffb4a6228 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/445c8ed006b6454d9e25bd709f3b3c94 : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24b97bb2b9334423a61c252f34da04a8 : SUCCESS in 34m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8710754a716407daa8dda39b11db394 : SUCCESS in 26m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/977ea5cdb3fe4db39ba08fb9546d6b6a : SUCCESS in 48m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e3b63cf77f8f41a58acfd7207797a9ef : SUCCESS in 56m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/135f55a012a04554aa1fdff3330a48db : SUCCESS in 1h 08m 52s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f0b771dc92b944b2bbb41603cd6d1d9a : SUCCESS in 1h 37m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3e8f3adbdc784d899a7d07e68162b412 : SUCCESS in 1h 19m 13s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/e13d38c1b1f64349a66bc83c08e3f1ef : SUCCESS in 1h 35m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/43d1df8a73b6460c92cc4f3f989581e8 : SUCCESS in 1h 08m 24s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/837d93747e18435d9f2570598b0117c6 : SUCCESS in 1h 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1a6b53ca48ff0bacfab042379a9a2877fac0d97c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-04 10:46:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/1f298bf3080142109bac3b54846e7e88 : SUCCESS in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aa3711d6f7f63ed6ba7f2e4b744d8cbb811e5ead","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:16:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b314811ceea94d73abd1c4dcb6c9b589 : FAILURE in 26m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7c568d6169144a7ac12e47d5cd0b505 : SUCCESS in 1h 42m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e60ffff7192d408eb07a3393daa7a409 : TIMED_OUT in 3h 07m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1328f24e9341421ea734c98f84d471f5 : SUCCESS in 1h 47m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"faa60d63d2ff2f6c4ae1fc3fef45bb4d817a057e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-08 10:50:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a77d971083a4de47cc4dca211af5450acc013cd0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-08 11:22:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e87aa01103294ac5a2c39d3002d493b4 : SUCCESS in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3769ffa62bf26c924e981f1b3ccbd4ae6bb0ebb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 12:47:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/db45000d1e60424bb6e2e90ead1bcb92 : SUCCESS in 1h 13m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d00f5bb6b8449b9a0e6a1ff9a9f86d5 : SUCCESS in 58m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eea55c1e84aa4b6fb91917dc8730d76b : SUCCESS in 1h 52m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f261baf6b854ef5a35262bfd1c43e74 : SUCCESS in 1h 03m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa1efe362b5886039e7abf4cf2f9598d4696ca76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 12:55:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e5fd84eab3a545fbaf29d5bb7484e7a6 : FAILURE in 22m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/88eb65b2b2f640a1842e77d42493114f : FAILURE in 19m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c7f91926856e4ef290430a1788692689 : SUCCESS in 1h 23m 10s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5567d41bd976456da88c7dda13cc17bf : SUCCESS in 1h 18m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e7a7ea0314554ee2bd9ac82114c53fed : FAILURE in 15m 53s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6f0425446610444bb6efd48bd13db44a : SUCCESS in 43m 10s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/6f634fd1ba74485c96c42d7114ba9d2a : SUCCESS in 29m 46s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f9e26eeaf5e440989d81fc69eebd7c13 : SUCCESS in 1h 56m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9f549d28a35941ae8949e84455a8f632 : FAILURE in 36m 51s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/4940c4957c514f7ebd413d4bd8a0cda7 : SUCCESS in 1h 58m 40s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/63c2272f3f7541a69b670bc72e7ea59f : SUCCESS in 1h 45m 10s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/9f3ba2a5cf024308a4dcb39fac981a5e : SUCCESS in 1h 21m 52s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/a3d1870957934a9c986c889c3e94e8d7 : SUCCESS in 1h 44m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a90741cdf38044a999ec8a51c3a48ad7 : SUCCESS in 26m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c50a1fbfa07c42daa703bfd16a786138 : SUCCESS in 20m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fd3f46ab67c640199754c23ac48a95e9 : SUCCESS in 1h 06m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe0418d2e04c409abf22707067c98621 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3dde6f73f6b413393b32e1802b01d3f : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bb7d8cb6d71478888274e79965a5db0 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0bd38f322454596a3399381f5347c3b : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/209e2dae97ea4af8a6ca7fed66beb5c4 : SUCCESS in 9m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b5739125de8c445fab84fdfab9d77bff : SUCCESS in 19m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6c9e170d61744f4a2887884b1ba194f : SUCCESS in 26m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d902b4dec8b04c9288cd277c91b1f2e9 : SUCCESS in 1h 01m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/51d395397bdd45509fad4bf726741d64 : SUCCESS in 1h 03m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/483b8fd4b4134909907e5b1b769ab15c : SUCCESS in 1h 07m 21s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/387269c1f27b49b39ad1a8f4afb680d3 : SUCCESS in 1h 40m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5b1802f0e714480da0006d13d4b24ef0 : SUCCESS in 1h 41m 06s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7f713c0e0dd346228acbfe706a263593 : SUCCESS in 1h 44m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/2e47a9bfda954e5bace8b77f32b662d4 : SUCCESS in 1h 00m 57s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/29e2ce01d18547bbbdf53146e2b9a8a9 : SUCCESS in 1h 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"313e020a1f02efb08deecbeec75c801e16eab7fd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-08 13:24:26.000000000","message":"Patch Set 3:\n\nhm. same build works locally","accounts_in_message":[],"_revision_number":3},{"id":"0eba9e1aa9a8e2c9815484dc76bd65576ff2e099","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-16 21:53:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d8e6ad48aab242431a730e733b252c2ac34243a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:18:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"16d8e97fa631a8e04b6dad433987e80fbc2e6c22","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-16 23:32:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/50114f1506d04a19b4111f15aee85cc1 : SUCCESS in 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3b675932b7efcd8ca7b9dc5735c97370f13651cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 02:58:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0c8c95dee1a740c785776fd6a6f44610 : SUCCESS in 27m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7bb6169b3a754f028c03b356f3ebccff : SUCCESS in 31m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6e934653fbd04843be86a580e34c4c0d : SUCCESS in 1h 22m 58s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/09ed15ff659b43a5a8e7bc1f5c715aae : SUCCESS in 1h 13m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/65d38000c6a14b969e8cd1ade38fdd2c : SUCCESS in 1h 26m 29s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e78aae5ecbe5491c9c4d19463e545952 : SUCCESS in 54m 30s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/dfe39dacd11749a98e30e56d92b265da : FAILURE in 16m 38s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d4629553b3a5408d8ea1457f70a31d97 : SUCCESS in 1h 58m 04s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ab8760d52d0341158494fb03d71b4c62 : SUCCESS in 2h 19m 38s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/da4aa19418c1409d89e942613a631cfd : SUCCESS in 2h 00m 56s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/9e3a56fe5f3647a1a84e8cb23bd07169 : SUCCESS in 2h 07m 58s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/ba864851ef1242dd87767cff931372fd : SUCCESS in 1h 37m 51s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/34bebb114f8a48a7a73ec37284eb207f : FAILURE in 20m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8f5cf8bd45d4433088099bb509093162 : SUCCESS in 23m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d8ffed78c7c8461da21e93af49a09d97 : SUCCESS in 27m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/58927305e73f40108abc20d5b517e67e : SUCCESS in 1h 18m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aeb8bcb805e4923993c718db7a46bc1 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c9f63d5a63f40f1a52da3d11fdeae3b : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bc2394993354dc2a9057ee9b487d989 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b28df306c9d474a81ee69e5ca36eead : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ed4b86188849789321cacc46cd5074 : SUCCESS in 10m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a53e87f9d9a46f88f48a4ba4381c3a5 : SUCCESS in 28m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d514b4090b45457589c6cb618c1e206d : SUCCESS in 30m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c33a62fdb5ff4348b3a5a17ba42c91cb : SUCCESS in 1h 06m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73185d7245ee4025a08bb727d8426e43 : SUCCESS in 1h 05m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1e8ac53c606049f3a459557da4acf7e6 : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/dce122656cc64ae1ae0633fe5d741415 : SUCCESS in 1h 34m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/dc2f283f23da4b618feb9a9adbbc3cc4 : POST_FAILURE in 1h 10m 03s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/83436b5369bc4afaa6348edc12313267 : SUCCESS in 1h 36m 50s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/23c9c8dc12074228bcb0a93a5274a43e : SUCCESS in 1h 05m 46s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/708e38780df648558ddb26f0d0f04b71 : SUCCESS in 1h 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ff4bcfdc25a3bf845b84a2a246d9d3248a24859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 02:58:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1e40152e8dbef9e913bc76064d790e56d6b7636e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 04:28:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e408d74eff534c7db20517a67bcef5d4 : SUCCESS in 22m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/07be6d546ef24b93a2cc1c74fafcd90f : SUCCESS in 1h 28m 45s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/45a6bd0a125248b3a252f7dc2ca5687a : SUCCESS in 1h 17m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85f96c0c9c5f4ff1838a1fcd6cd82d94 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f896261b7d4caf86c9ca02cd6230b4 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bdf0df77c7447309e49726410f4d560 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcb65f08f55f4e7e8e8c56dc6bfa483d : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f7ceefbf83b4e748c51b16f3b32eb9f : SUCCESS in 5m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/85a2e3f5a3624bc8be006cf51a9a4467 : SUCCESS in 19m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fe96af720ee4acaa9d01590e4a049bb : SUCCESS in 45m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b5adae99c114d3caf77910f5df734b3 : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":3},{"id":"d3a16b75ade1ce6e7036fc6edb752b1c48e8bfd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 04:28:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b8ded8020da0468f9491b0568261de8a40222363","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 04:30:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6a4075b22d34a90b4bc371afc0b03ea : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"09604a7492752c1e4afa9c0facfb4309e9862616","revisions":{"b85748f8749417cd6eabab1970bc31fda9805db7":{"kind":"REWORK","_number":1,"created":"2020-05-11 15:31:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/84/726884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/726884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/726884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/726884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/726884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/726884/1"}}},"commit":{"parents":[{"commit":"cf06869684455c3ac36567d61c38ce77a6d869e6","subject":"Merge \"Enable mod_wsgi in Nova API container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf06869684455c3ac36567d61c38ce77a6d869e6"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-11 15:29:37.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-11 15:29:37.000000000","tz":0},"subject":"base/centos: take care of whitespaces","message":"base/centos: take care of whitespaces\n\nLet\u0027s get rid of this warning:\n\nChange-Id: I121e2ecb30ec6e8d1b6d88d9b921894438c3e094\nINFO:kolla.common.utils.base:ESC[91m[WARNING]: Empty continuation line found in:\nINFO:kolla.common.utils.base:    RUN dnf -y install centos-release-opstools epel-release dnf-plugins-core     \u0026\u0026 dnf config-manager --enable PowerTools     \u0026\u0026 dnf config-manager --disable centos-ceph-nautilus     \u0026\u0026 dnf config-manager --disable centos-nfs-ganesha28     \u0026\u0026 dnf config-manager --disable influxdb \u0026\u0026 rpm --import /etc/pki/rpm-gpg/RPM-GPG-KEY-CentOS-SIG-OpsTools \u0026\u0026 dnf -y distro-sync --security --sec-severity\u003dImportant --sec-severity\u003dCritical\nINFO:kolla.common.utils.base:[WARNING]: Empty continuation lines will become errors in a future release.\nINFO:kolla.common.utils.base:ESC[0m\nINFO:kolla.common.utils.base:Step 1/39 : FROM centos:8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b85748f8749417cd6eabab1970bc31fda9805db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b85748f8749417cd6eabab1970bc31fda9805db7"}]},"branch":"refs/heads/master"},"842b081eeff3a45d4338a17202dc276b6a589a9c":{"kind":"REWORK","_number":2,"created":"2020-05-17 13:21:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/84/726884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/726884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/726884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/726884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/726884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/726884/2"}}},"commit":{"parents":[{"commit":"aabd532da9fe3cac64626d481c4b3113b5925e69","subject":"Remove Monasca Log API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aabd532da9fe3cac64626d481c4b3113b5925e69"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-11 15:29:37.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-17 13:21:11.000000000","tz":120},"subject":"base/centos: take care of whitespaces","message":"base/centos: take care of whitespaces\n\nLet\u0027s get rid of this warning:\n\nChange-Id: I121e2ecb30ec6e8d1b6d88d9b921894438c3e094\nINFO:kolla.common.utils.base:ESC[91m[WARNING]: Empty continuation line found in:\nINFO:kolla.common.utils.base:    RUN dnf -y install centos-release-opstools epel-release dnf-plugins-core     \u0026\u0026 dnf config-manager --enable PowerTools     \u0026\u0026 dnf config-manager --disable centos-ceph-nautilus     \u0026\u0026 dnf config-manager --disable centos-nfs-ganesha28     \u0026\u0026 dnf config-manager --disable influxdb \u0026\u0026 rpm --import /etc/pki/rpm-gpg/RPM-GPG-KEY-CentOS-SIG-OpsTools \u0026\u0026 dnf -y distro-sync --security --sec-severity\u003dImportant --sec-severity\u003dCritical\nINFO:kolla.common.utils.base:[WARNING]: Empty continuation lines will become errors in a future release.\nINFO:kolla.common.utils.base:ESC[0m\nINFO:kolla.common.utils.base:Step 1/39 : FROM centos:8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/842b081eeff3a45d4338a17202dc276b6a589a9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/842b081eeff3a45d4338a17202dc276b6a589a9c"}]},"branch":"refs/heads/master"},"09604a7492752c1e4afa9c0facfb4309e9862616":{"kind":"REWORK","_number":3,"created":"2020-05-17 15:38:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/84/726884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/726884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/726884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/726884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/726884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/726884/3"}}},"commit":{"parents":[{"commit":"aabd532da9fe3cac64626d481c4b3113b5925e69","subject":"Remove Monasca Log API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aabd532da9fe3cac64626d481c4b3113b5925e69"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-11 15:29:37.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-17 15:38:10.000000000","tz":120},"subject":"base/centos: take care of whitespaces","message":"base/centos: take care of whitespaces\n\nLet\u0027s get rid of this warning:\n\nChange-Id: I121e2ecb30ec6e8d1b6d88d9b921894438c3e094\nINFO:kolla.common.utils.base:ESC[91m[WARNING]: Empty continuation line found in:\nINFO:kolla.common.utils.base:    RUN dnf -y install centos-release-opstools epel-release dnf-plugins-core     \u0026\u0026 dnf config-manager --enable PowerTools     \u0026\u0026 dnf config-manager --disable centos-ceph-nautilus     \u0026\u0026 dnf config-manager --disable centos-nfs-ganesha28     \u0026\u0026 dnf config-manager --disable influxdb \u0026\u0026 rpm --import /etc/pki/rpm-gpg/RPM-GPG-KEY-CentOS-SIG-OpsTools \u0026\u0026 dnf -y distro-sync --security --sec-severity\u003dImportant --sec-severity\u003dCritical\nINFO:kolla.common.utils.base:[WARNING]: Empty continuation lines will become errors in a future release.\nINFO:kolla.common.utils.base:ESC[0m\nINFO:kolla.common.utils.base:Step 1/39 : FROM centos:8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09604a7492752c1e4afa9c0facfb4309e9862616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09604a7492752c1e4afa9c0facfb4309e9862616"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
