)]}'
{"id":"openstack%2Fkolla~729298","triplet_id":"openstack%2Fkolla~master~If80dcfe4a99893ac82797c505a78ceeecebcb0b9","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"If80dcfe4a99893ac82797c505a78ceeecebcb0b9","subject":"Switch to tarballs.opendev.org for source images","status":"MERGED","created":"2020-05-19 15:30:25.000000000","updated":"2020-05-24 17:52:24.000000000","submitted":"2020-05-20 01:21:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729298-1589937676011-d1623aff","meta_rev_id":"f133851b0b9c83fe04443c11b5f890e3187e730c","_number":729298,"virtual_id_number":729298,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 01:21:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-19 16:58:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2020-05-19 23:50:09.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-05-19 19:20:23.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2020-05-19 23:50:09.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2020-05-19 18:20:03.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 16:58:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-19 19:20:23.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-05-19 23:50:09.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-20 01:21:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0739ff9e3700ff580fd3db7c51417eaaf9d01a2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 15:30:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5f79c94e971fb03f79c0a956fbe1073a02e0b33","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-19 16:58:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b6eb0fcaa07847dfbbdd2c7101694951 : SUCCESS in 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a7d9358737f75e3e2e1edb8841a1f027a35dfb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 16:59:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a4c3fa65d2834aa9ae90509a804da099 : SUCCESS in 44m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6b3730696f2b47e788731724b85d8545 : SUCCESS in 34m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1cc55f44d8fa4f73b8ed1d0f94a3dcf1 : SUCCESS in 1h 23m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e3b76ea7a9934e24a4c5121305d5ff0d : SUCCESS in 1h 19m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2e064f778ab944d39b9eb8d41dcd861b : SUCCESS in 1h 24m 19s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/66575420903b4a218ca137209f93c77c : SUCCESS in 1h 03m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cf9ab3ac798a4b1c92b8aaa8e4867907 : SUCCESS in 28m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d25f0327a3af45108443b119830fe793 : SUCCESS in 20m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/092bd08872a446a1ac8fe15a5e5e298c : SUCCESS in 1h 07m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9f638d576c049cc9785823b705b7e16 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f4cfbc91764b63ac5f472210e22fe0 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/956ce46934654198a4c2d82f38b29e3e : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79cfb8e5fd5947eeb3334d02f825b457 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/552320465fba463a8174d8d3766bb52d : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cf83fe27dab4d43a6881036f6354c87 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2bd5017249941f880aab5fd7039a590 : SUCCESS in 4m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/694afd6470954c0185a663167f23adf4 : SUCCESS in 26m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff9dfbeb630b4475b683497aa3085f83 : SUCCESS in 28m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec18560749dc488abbc87a83658ccb69 : SUCCESS in 1h 12m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f0aa3a28d47a41709fd644dc46647af6 : SUCCESS in 1h 00m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ff509680bfc94b55849c93c59da206e3 : SUCCESS in 1h 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"52bc939171797ad8770f87076dd993648f773ae1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 18:05:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4689585cd7947fa9674c3eb94333ff8 : FAILURE in 1h 20m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/99b70e7cb1284b0fb8c12079d52c98e4 : SUCCESS in 1h 19m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/068ef07e5b6d4f7a93a92ad54316eeac : SUCCESS in 2h 27m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b545ea68cb849eaa3843fb2d65507ee : FAILURE in 1h 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d653e974a9e7528956f1acd44f3461faf518d463","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 18:20:03.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"5255614daad0fbe86ded679c775caeb76c6f9fc5","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-05-19 19:20:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"34baab953826f75a400b15217ca338f76a52fd77","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-19 23:50:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78601794134a192a0b4bc9f4fdb801e59133dba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 23:50:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d21fe6ccf1a65a87e94c70128e1bfed6ca413c16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 01:21:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d7614bdfcf614b859ed7d1025b2dd86c : SUCCESS in 49m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/29ff70e2783e4ff197d96caec53dc64e : SUCCESS in 1h 29m 28s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7e60bf5b7cb748669964a76e822ba350 : SUCCESS in 1h 15m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84963bc8bd0f461e97d57cc682dc0f33 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177308258e8e4b22b55cfadaf3221cfe : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3686e43b148b42adafe0112414daaa9d : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52e50372d35d4aa1ba813b4633a7bfaa : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2959e2a7b056471fa200312f06b5f440 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96aa56eebf574067a93e470b8dc9a040 : SUCCESS in 4m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e3bd2a70ac4411dbfbd47d3dec7d649 : SUCCESS in 36m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b828d96209f45e9a69ca27f9893d939 : SUCCESS in 1h 08m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2548d19636094435bd9cf8323d9af236 : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b87e6e061eb06b60d821b412fbb860255c9090f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 01:21:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e13705adc4bffdab0e7dcdc67c251bb90efa9fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 01:23:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/613254debbd344859fe3aaf66f5b3c77 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f5700eee1f44a3087b44a45bdd03d65 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":1},{"id":"883f7c37c8aa5d113c749ce6603034db737786e2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-20 06:59:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f133851b0b9c83fe04443c11b5f890e3187e730c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:52:24.000000000","message":"Created a revert of this change as I0700fbc452f395bd21757ec57155308441ce9753","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ededc45cf8a7fe9398032d305218d31660bcce79","revisions":{"ededc45cf8a7fe9398032d305218d31660bcce79":{"kind":"REWORK","_number":1,"created":"2020-05-19 15:30:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/729298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/729298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/729298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/729298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/729298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/729298/1"}}},"commit":{"parents":[{"commit":"55eab6edc43ddb24a918298aafefdff322afb61b","subject":"Merge \"octavia: Add ovn-octavia-driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55eab6edc43ddb24a918298aafefdff322afb61b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-19 15:16:34.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-19 15:28:24.000000000","tz":60},"subject":"Switch to tarballs.opendev.org for source images","message":"Switch to tarballs.opendev.org for source images\n\nOpenStack tarballs are hosted on tarballs.opendev.org.\ntarballs.openstack.org is just a redirect.\n\nThis change switches the tarballs-base configuration option to\nhttps://tarballs.opendev.org. Since the new site is namespaced, we need\nto add openstack/ to the URL path, or in some cases x/.\n\nThe following projects now publish under the x/ namespace:\n\n* networking-ansible\n* networking-mlnx\n* vmware-nsx\n* vmware-nsxlib\n\nChange-Id: If80dcfe4a99893ac82797c505a78ceeecebcb0b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ededc45cf8a7fe9398032d305218d31660bcce79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ededc45cf8a7fe9398032d305218d31660bcce79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
