)]}'
{"id":"openstack%2Fkolla~729299","triplet_id":"openstack%2Fkolla~master~I08230d5b9eedbbbde9616a0dfba5ba6f6ef491d8","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I08230d5b9eedbbbde9616a0dfba5ba6f6ef491d8","subject":"Switch to Ussuri release tarballs for vmware-nsx*","status":"MERGED","created":"2020-05-19 15:30:25.000000000","updated":"2020-05-24 17:42:27.000000000","submitted":"2020-05-20 13:11:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729299-1589980306750-e2f5466d","meta_rev_id":"c986ac2316a4a8a88699593216c92319a41fa8f6","_number":729299,"virtual_id_number":729299,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 13:11:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-20 11:34:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-05-20 03:24:12.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-05-20 06:59:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2020-05-20 06:59:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-05-19 18:20:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 03:24:12.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-05-20 06:59:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-05-20 11:34:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-20 13:11:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12330ff76b6f0c507f2e5b3d511971fcd998caff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 15:30:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d4667a0bf7a758c24c3a777dac409c74671f770","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-19 17:03:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/36f2c4b2912b412ba3e74e34e2549ecb : SUCCESS in 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acd3133b4621a5a0bceb6c818185ed01e39ce201","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 18:08:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c9a67453e8d4b6a8b4184dfd5191e30 : FAILURE in 1h 15m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ed29f548b0a40efa1860a658f70012b : SUCCESS in 1h 19m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b810b6a661d4e639138b5cfe62a5082 : SUCCESS in 2h 28m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c67638f8097d4e168dce4aa58c9ff7b2 : FAILURE in 1h 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c4b786aab6e7d9f9599646171509decd1a63f70","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 18:20:14.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"a484e7e4b78fe2519230b950556a1c40ef0c2ea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 19:05:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/213726e079fc429e93802417386099fe : SUCCESS in 25m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/39c62df44605423dafb8fedaeb0fe089 : SUCCESS in 30m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3ebb0cb03d974834b3b33daa548658d8 : SUCCESS in 1h 19m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0643670df80f464f909f2cd68ad90fdc : SUCCESS in 1h 10m 32s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3519200c5142475781205dcd8b7dd67d : SUCCESS in 1h 27m 09s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ad1be4acab0247b38ecca756da3e10b0 : SUCCESS in 56m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3efe9192078549d5a0596abe2e2ab7ef : SUCCESS in 29m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/44c883e13a2045ec88d792b818e2d213 : SUCCESS in 26m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7628006c730c4a28a653400700745d4f : SUCCESS in 1h 01m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58ec1daedfdd42a3bafe67374d14a6bb : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc19f6c4d93d4c87858d20b85c739d18 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfad30ebb5af47f48d2165f7dab16e8d : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9d02fd63387474f837209d47c45f0bf : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edd6b17c55a243d5996b031986e32eed : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be05d65e8b545c8b0f111ff31d7eca8 : SUCCESS in 7m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b6b617a1d7db4527aa374c46914a7df9 : SUCCESS in 33m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1985fcbd7a04fe8a9ceb3ef357b11ed : SUCCESS in 41m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/086f4225a054485eb638640644b67258 : SUCCESS in 1h 07m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/520f35722019496db434002460e368ae : SUCCESS in 1h 10m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/214cf9705ba44e0796dee7dcc1ddde4a : TIMED_OUT in 3h 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5207ee9d849a3ce97d01eddc85cfcdecc833d4d9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 19:07:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c3ad0524a16a5f9459b999f710e25317d5075ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 20:39:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cbe0e652bf3f49c89e8446a790a66108 : SUCCESS in 25m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4e3357ed631f4e139305706f007261ab : SUCCESS in 31m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/04ca8ce4fa4c4aad8d53fe78efc40622 : SUCCESS in 1h 27m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/96cd166199dc44adb345e2109783fa65 : SUCCESS in 1h 22m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cc7d6d12699e4f44bf451b354a0b8645 : SUCCESS in 1h 20m 03s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/fc54cd54ada14ecba285dedaa37a70ca : SUCCESS in 41m 18s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b25f4a44e18f4f3b8b1b9e20735b0249 : SUCCESS in 19m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ecb1f201ca7b4db2b4002b2558ea915f : SUCCESS in 21m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/613101d63acc4abeae4986b49e134aea : SUCCESS in 1h 08m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17a69d62d4094e45931b72e2ebb73fd6 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0150c99131074b91bde97781b5239f73 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72abbcdfdcf44061816f6c5852ccd4c6 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01352a716efd4d1abbcd6fff6f3d54b8 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea4022c8cbf74213b6438c90af530e67 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d922fab6e834e79845162e707683d5e : SUCCESS in 8m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fe928afd228f4768823a2d9d4b099bf7 : SUCCESS in 24m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d76eddfdab94de987899ff82830176d : SUCCESS in 35m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7992d84d56340f38244e7b6c0f1a91d : SUCCESS in 1h 16m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b0a7c1a19a52426ea78ab4a3a389bd28 : SUCCESS in 1h 00m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4e21b8fbea9a4fb7bb04b2da29f2874e : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"42888b837f093fcedce4af69eae19590fd3a8732","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-19 20:40:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b903bb71e2f64cefb689e69d114becf5 : SUCCESS in 29m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddab4934c8340c6edbbc51a6a8d69224756b9fda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 21:06:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc5bc4b5028a4b078d928bea497e292f : FAILURE in 10m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/921fc9f3bd9845d4a2a8f3257a69fc74 : SUCCESS in 58m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b40857a272af4bd7b7a3e11d1ea4c225 : SUCCESS in 1h 52m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d986c70bed9a4e0091f4160fa8b28c1d : FAILURE in 56m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39c0d5cee5084d2b76904801291fb3f0a08d65b1","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-20 03:24:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e51af60de20e39b30e2fffd437f7c59e17487ae2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-20 06:59:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c6e876538c5db8f1230bc301a2e427207d3e3ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 06:59:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"491c4f072b9dedd9823971e736a027e549f4ac7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 09:10:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7f7362842cd341f4ae7c1748891858a9 : SUCCESS in 32m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/29f6ec5c07634a7daada1899b9c36a0a : TIMED_OUT in 2h 08m 40s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5a2e4bfb0a284a798e17384d764e861b : SUCCESS in 1h 29m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce1d01a4dc5345369f9e2afa11ae1419 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f81793f2444228affdc479713c2751 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cf5be58a78f4f44864a327b878d944e : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2fe863369ec04b5eae2edfb926431e11 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aba9bfc69f7478c8dd42127f9e4b536 : SUCCESS in 12m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/73c2c82e1aff4528ab9c106285efd738 : SUCCESS in 33m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16316ed133c64e8c9437df4d9991e37d : SUCCESS in 1h 15m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0c7fc806d43a40639e6bb10fe65a6a93 : SUCCESS in 1h 24m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b65192e692c064cae00e0272b459ef14e89d4a75","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-20 09:51:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e6e3b84f77e74b879df40f59d828db2674e304b7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-20 11:34:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/653773b540084b6c922199b66f4e2021 : SUCCESS in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f32b17249d57ff146215521c8adf431e2e0a8de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 11:40:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2aa388410cd949eba19b688a5d2884b9 : FAILURE in 58m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/25de3218e6f64db3a49a2f19ec313d1f : SUCCESS in 53m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/634db9d3c94f420bb215d15ad8265a42 : SUCCESS in 1h 45m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cfbe5ae219604ddab2417ed3364c2386 : FAILURE in 56m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb7b78a66b0aff9a985d22341361a9bea6a43576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 11:41:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b1045ed7d4d649c186f2fc6716998c2a : SUCCESS in 35m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/59ad758703a14e39beb93341d0a23f9f : SUCCESS in 34m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d15da33fa76641109cc9339d1683c8bd : SUCCESS in 1h 14m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/71606628a8054fdeac621ef86c8e074b : SUCCESS in 1h 08m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/613ffd2b66d34601af5e261391c88ebe : SUCCESS in 1h 16m 04s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/1d68b4cc99c749eb8d3200bd40256ca2 : SUCCESS in 50m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ea46af69a0da4c9bb9b3a0fb69742d65 : SUCCESS in 25m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2e808c28d14f4fc0a7fa5402603d9a43 : SUCCESS in 40m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1a8fadb0742c4714874cbb65a3909e5d : SUCCESS in 1h 02m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7ad7f38bb504f4592fc5a9ce2313e19 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003650ed4357449c854fb49df17e9d9d : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93dd8e55c956463fbbd2a3e932a8f3cd : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b5c4a13a59e4f77981683bd7016fcd8 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d83120818e3b427cb626a54c56dd883d : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d7137b846e45b095a12795fb285118 : SUCCESS in 6m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f5f248188df4e33b7d0f84f96ecccba : SUCCESS in 24m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e22c964758e04cc99ff76ee5b2621584 : SUCCESS in 26m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/52bfe9dc605644d0a6df469ca4571b40 : SUCCESS in 1h 20m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6956a33a458f4bdc9dcab2628938d88f : SUCCESS in 1h 16m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1df6aae35c944ab78a6c127ce13a8a6e : SUCCESS in 1h 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"47d582ff041e100b11ac67af49d0cc055b61b9a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 11:42:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f8163dc7897a63877ff0055a43c2c9886dabb47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 13:11:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d12cdc6286c2bc358ae9f2619f2c218090678e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 13:11:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/aa75d5dbca524cd680534b0d5a77feec : SUCCESS in 52m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/806cabf5cb934eda9cf704f01c0fec33 : SUCCESS in 1h 25m 15s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c5d2d5ec3cd3462398bbf6fbc48c3ab6 : SUCCESS in 1h 25m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab9625c2acd241d58dd04c8e27d055ec : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e7f9ae105845e0b2b59af6da1f89f4 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1573c5cde68b479ba2c103cbcffa09a0 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff9020c9c97c474dbb03e6adb995d5c4 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b97d9746024d97987fe72f5df638db : SUCCESS in 8m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4592189945ae4c3ab10c5d26c95c1357 : SUCCESS in 50m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64310d0cc7134ff9902255ed5cd5e9b8 : SUCCESS in 1h 12m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e045d53a3b04159a6bb15bc684302a6 : SUCCESS in 1h 17m 01s","accounts_in_message":[],"_revision_number":1},{"id":"dd1650c69025ac2021140432983cc691110c2f19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 13:14:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da71bd97c384a988c9d00d98fcf5da2 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c986ac2316a4a8a88699593216c92319a41fa8f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:42:27.000000000","message":"Created a revert of this change as Ia6d4eb684f0ff1893d65b3694939f4caeb9bb471","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a6904de217b2f35b70b7740c4198fdfe7b25d07","revisions":{"4a6904de217b2f35b70b7740c4198fdfe7b25d07":{"kind":"REWORK","_number":1,"created":"2020-05-19 15:30:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/729299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/729299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/729299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/729299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/729299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/729299/1"}}},"commit":{"parents":[{"commit":"ededc45cf8a7fe9398032d305218d31660bcce79","subject":"Switch to tarballs.opendev.org for source images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ededc45cf8a7fe9398032d305218d31660bcce79"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-19 15:25:23.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-19 15:28:24.000000000","tz":60},"subject":"Switch to Ussuri release tarballs for vmware-nsx*","message":"Switch to Ussuri release tarballs for vmware-nsx*\n\nChange-Id: I08230d5b9eedbbbde9616a0dfba5ba6f6ef491d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a6904de217b2f35b70b7740c4198fdfe7b25d07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a6904de217b2f35b70b7740c4198fdfe7b25d07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
