)]}'
{"id":"openstack%2Fkolla~729460","triplet_id":"openstack%2Fkolla~master~I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3","project":"openstack/kolla","branch":"master","topic":"retire-congress","hashtags":[],"change_id":"I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3","subject":"Remove the congress images since it has been retired","status":"MERGED","created":"2020-05-20 01:04:41.000000000","updated":"2020-06-20 00:58:52.000000000","submitted":"2020-06-20 00:57:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729460-1592614636379-0cb2b021","meta_rev_id":"79a3f1e5f317bc7ee0bbbaca3fdf64ec23bde1d5","_number":729460,"virtual_id_number":729460,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-20 00:57:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-19 19:20:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"recommended":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"all":[{"value":1,"date":"2020-06-19 18:13:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":2,"date":"2020-06-19 18:13:53.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-06-19 18:13:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2020-06-19 18:13:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 18:13:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":1,"date":"2020-06-19 18:14:00.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 09:50:07.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-06-19 05:36:59.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2020-06-19 08:04:10.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-19 10:01:37.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-06-19 11:01:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-19 18:14:00.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-06-19 19:20:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-20 00:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe5ada81064c06655f382826537df06828f4d3ac","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-20 01:04:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c12d3f548ee8ce8a6dea2e5d3fbad3692d05e67d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-20 01:32:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/c58d0c60eb4a49169d376002cfb33b4b : SUCCESS in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31a98ea7dd86dd248c02392d8a6dd2cd1b94bb8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 02:38:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4f329709ab6d42ad82ec964e24aefd29 : SUCCESS in 26m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9608122413ad4b20b0c8e2e4eee2aa69 : SUCCESS in 31m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/47fc51203b954030971aa5a5f412936d : SUCCESS in 1h 28m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5b4b1acd1b8f4e1fbb47ebb7940d68cc : SUCCESS in 1h 06m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2da40e0c62a94411b5fd86ed0ac8171d : SUCCESS in 1h 17m 58s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e79481d3b28042acb2a9ca1d2a9e75ea : SUCCESS in 49m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/debe0ce8ced04a5d95dad23ec992c650 : SUCCESS in 23m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/015437f5d5e44cccab08f5143b0bce8f : SUCCESS in 19m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/077d06750acd4f1bbf67e5cd55921b5c : SUCCESS in 1h 04m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5557fbc44a34016a70cdb8a9606f2de : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ab6edfe232a4dbf8989e80d29548955 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3309a7212264437596ae391cb0931999 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0af8940b6f3745edb8daf20374ed6f6f : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54a1f7b18f3346369d30709bca1c0af0 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4081deb0f9d4d3f99714079c5cf9789 : SUCCESS in 6m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f4632477ebe42dd86b9ecf3297dc068 : SUCCESS in 26m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f461a4c19d9847f78356ac6121807ab0 : SUCCESS in 1h 13m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/80a864889e4f43afb85cbc43e5de67be : SUCCESS in 1h 16m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef435813a2844593a29c62d9c13da0a0 : SUCCESS in 1h 12m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fb843ff9094d40fe84a5ae9864d73451 : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"afa7a0a35cd1886007eee8e30d174ddb42569a93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 03:00:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9df2e9825b0411f8c6666357fa9a2c6 : SUCCESS in 57m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d5cda7f8e3a484db37d973ef4aa16a7 : SUCCESS in 52m 29s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9fbc589e1bec4d62bf886128a362cd99 : SUCCESS in 1h 49m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/08092f75320a40549818a458ee99947e : FAILURE in 54m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f221178fc4b3e88a4c9b866e6cf93cb761bde1ad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-20 06:36:11.000000000","message":"Patch Set 1: Code-Review-1 Review-Priority-1\n\n1) it is going to be retired, not retired yet\n2) we need normal deprecation cycle anyway\n3) kolla-ansible must deprecate\u0026remove first","accounts_in_message":[],"_revision_number":1},{"id":"44086c7b05c589fa412b1a279d3bb8911b9d4a64","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-23 08:44:45.000000000","message":"Patch Set 1: -Review-Priority\n\nussuri branched, unfreezing","accounts_in_message":[],"_revision_number":1},{"id":"e945435c528f9bf7cc0e5f02fd0d006fe7463a56","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-29 05:41:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a9c96f75ca8e9711479b556efc5b86b441592aa4","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-29 05:41:55.000000000","message":"Patch Set 2:\n\ncongress is retired now, we may need to review this ps","accounts_in_message":[],"_revision_number":2},{"id":"16cd7ed2772698a15c2ead9f4f370de884b0c41f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 06:08:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f0eb3b33e86444a9ae9af1d7b3e1281 : FAILURE in 7m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c33c86dd6da3417ab4428c8cfcc94f2b : FAILURE in 18m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/95bffeecb94e435fa2ad657e56af5a91 : FAILURE in 23m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/64297b5f46b14d2681520a609424279e : FAILURE in 20m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"509f62e6fb80d1ea3f2f5e45f74514826a03af81","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-29 06:11:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d55142d7fe264bf4a8da457ad8b2b57e : SUCCESS in 28m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5943160ac6b7a0c9472651feebb6dbe59cb02ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 07:07:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/48331c251a3f40a6a7a9186431419e24 : SUCCESS in 21m 32s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ff5414ac9d8546c4b631a07d82c24c3a : SUCCESS in 28m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/225fcd5d1a8f425a86f596a1865944af : FAILURE in 55m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3fe1a93ee20a49ed9ce5ae92b4cdc6ec : SUCCESS in 1h 07m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/22ab78dc3a6a46108f44e5af3cd3f5a4 : SUCCESS in 1h 18m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/467e4588090345d8a5d837a30a267c3c : SUCCESS in 48m 19s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4a784bd872fd4b63822e8ce50d27e4db : SUCCESS in 19m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/480a8f7f867943ba9804c55c96ae5edf : SUCCESS in 17m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/43c7bb77b06a470584668be8462074ad : FAILURE in 45m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/155e994b49b7495385f339585d2713e6 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dab9efc2aa4f41ac8ec126f3f06e5d8d : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/367a73f04e094243b0a96d7151018c85 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96ec415d066e47cda2e02302f669060b : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba0b50047db949aaa520810ac3e9cc14 : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5915401b3aae4bfb9e7b3ae481f6aeb3 : SUCCESS in 22m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9dca1824e18e4f54b5cbd726b0745234 : SUCCESS in 28m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2be5362167449c2a33009a1e229aff3 : FAILURE in 47m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8438009361be4090b3da832455d7a5b6 : SUCCESS in 58m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c92c2dc08f6148aa90f48b0fd9efa4c0 : SUCCESS in 1h 18m 53s","accounts_in_message":[],"_revision_number":2},{"id":"666ca814bca054e38aa4c6306701cb8a631183b1","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-29 13:46:13.000000000","message":"Topic set to retire-congress","accounts_in_message":[],"_revision_number":2},{"id":"50f2f033d56b90b2f5dc8e9e206cc33c38fe2e82","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-29 15:02:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c634c2e68c28b4eda3ed6449b2acda712e1d7c95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 15:34:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/07e96625df824e43bb3e492d83cca020 : FAILURE in 10m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/76183ce6338845508ac066e2b533ad81 : FAILURE in 23m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/445b900a2e02426590c4c7b483dd7712 : FAILURE in 27m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/eeff0b94d5e048559cd8443ba56bd8e9 : FAILURE in 21m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8fb9294a38567d86676cd5969bc6d714b0fe298","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-29 16:16:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/41f7aaaf65404c6f9e3930c5f6c151f4 : FAILURE in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a630bd9a0f6bee037f9843d900e69f3678e2db8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 18:23:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e9834ddc8633448886247ccc235a6dd7 : SUCCESS in 3m 19s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/eb4eac4394e843bf857043c6487e18db : SUCCESS in 22m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/74b5cea8140f4b0fa678ff49005492fb : SUCCESS in 29m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ddc23e381e8746d4b907c56d9945f4c0 : FAILURE in 58m 22s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2716d7b2ca174855b08995decf985014 : FAILURE in 51m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4d82ba79182946d99e96ee0a532664e5 : SUCCESS in 1h 16m 11s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/676fde2d44a848a28331ee04fce88e9d : SUCCESS in 53m 11s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/caf9f1253a7e4f19a62f6d8d4fa1d78c : SUCCESS in 32m 47s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8f2d66db64d54f778f52115e84ad7fcf : FAILURE in 1h 24m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/be31e0da2ed04347b6b10eb9bd69d76e : SUCCESS in 2h 04m 35s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/aa3cfabae7ce44368052bb4f914ee818 : FAILURE in 1h 49m 29s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f776ee1f54a544a08cc59239ef33f454 : FAILURE in 1h 24m 41s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/64c1dee7b84b4f4a833490d1d6997411 : FAILURE in 53m 23s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/e57ba460e7a94c628a5e339caf1a1938 : FAILURE in 1h 14m 15s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7b8ceb96ab8c4f1f8cc3bb8f05fce5ea : SUCCESS in 32m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/295e1804ae984254a88a043c5b8a1ff8 : SUCCESS in 19m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/134f99c5ef9b4600ba486297c678f6a1 : FAILURE in 37m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdad4268cb814a1b8180e68ac39d1618 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd0998974f9640b0926165e0e7ab344b : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4262f77976034f368b9d5b2572a9cf12 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3638b7f7ee364e34aa65935f5bff4103 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6418283d170454dbd7d4b3c74cfc356 : SUCCESS in 6m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/777547246c0649e69df811a708eb6f37 : SUCCESS in 23m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4898a057df741b98d5fb4f14dd2a344 : SUCCESS in 23m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/686707a879ea4381b898edac693cc56b : FAILURE in 45m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ff952cbf3fae4f228ba8832081b8f269 : SUCCESS in 58m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7dd940da53074d4b8a500fc690efed26 : SUCCESS in 1h 10m 46s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/726f72cd811b4905b24e6c586b2e5a6d : FAILURE in 1h 11m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9a15f4e3b1af4425b2be91a1fc1726d0 : SUCCESS in 1h 25m 12s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/910876982dc843de844c1c5461730b09 : FAILURE in 1h 05m 23s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a484530d0c1648a1ae9276025a924573 : FAILURE in 46m 17s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/3010fb398a864116b9b6222e7a07346e : FAILURE in 59m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c93df18e533bb700ecb16d6d9d9ab794e08c792e","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-31 14:27:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e6e0ca162ad6ef990f551f84707c9d157238df9f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-31 14:45:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e6c96ac6999a4c07b029b77cbddbae5f : FAILURE in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9a274080298c189e9a59e9cf14a920c749eeb06e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 15:00:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f855c0ac02f40f9b1cb8cf78ac5081c : FAILURE in 10m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/16c7070d455d4714ab7dfac839d567e2 : FAILURE in 24m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f09ce851ab441f7bbb70b88e9029a98 : FAILURE in 27m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c0d3856f8fe48758763bc8992afc823 : FAILURE in 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c5b9713bf4a770c3d558cf922140dbea5478dd05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 17:45:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9ce6f6d517b545f1bfb740638a792361 : SUCCESS in 2m 02s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/24f1239320ed4f44a1c4296902e96af9 : SUCCESS in 23m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fd4485a48be442a484675fe0457c9323 : SUCCESS in 30m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c146009dfd6d44229ac901ace8f8e27e : FAILURE in 1h 29m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1d96def822c5464a8c032eb464cfbfb4 : FAILURE in 1h 27m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/de49a876b4ed416f8777b9f4bab87999 : SUCCESS in 1h 22m 03s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/98af0ee0624d4e94ae83eac9ee107f87 : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/e16521add2dd4a2fb7c9092e3f0eea93 : SUCCESS in 49m 18s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f5ddb8b3b2e541da9d290b29117c2a70 : FAILURE in 1h 11m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b403418648de4572b268f77fb646b1b6 : SUCCESS in 2h 11m 22s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/670dad3525774417acc0bf125dfa7f58 : TIMED_OUT in 2h 05m 06s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/605a647707f9473ca5a43b56c824d4f2 : TIMED_OUT in 2h 06m 51s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/c5823b348cf841808f9909fe5ab865ed : TIMED_OUT in 2h 04m 10s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/8a7ddffd40c24c588c57cf184fcf2d74 : TIMED_OUT in 2h 07m 22s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ed07e71dddf44a2e80895c3c6367fd05 : SUCCESS in 20m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6f21d8d833c047fdb480e33dd6c99f82 : SUCCESS in 20m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc16f1dd5fe945ada1b74ff723768bc3 : FAILURE in 46m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be16db6ac033470d83e6b197a29ee767 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ef4ef820704084b664bb0ca00210f2 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d74b3c2542be4ab2ad7ac21252e35940 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/097320da86714a9d8351895b854a6ffe : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5b6a5e0a224521a17392dd49da4423 : SUCCESS in 4m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3fcb66411d34d13953c5273dad69972 : SUCCESS in 25m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ea63518d26c4e5fa8c77605c456f7e9 : SUCCESS in 28m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/593f5d3e6b134d34b434e0f962058782 : FAILURE in 40m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c6326ba7df54e29abd115559d230c0a : SUCCESS in 58m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e83032083d7d4ae79ecce51f421a06a8 : SUCCESS in 2h 13m 36s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b76b9df96587423c9bb9020a812618b2 : FAILURE in 1h 03m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/0e9568da1072491f83bf3ba625979f75 : SUCCESS in 1h 49m 15s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ab9981c9c30d444aabfb4cfa19229dab : FAILURE in 1h 10m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c44ee694eafb4442a045dec9d5700ee0 : FAILURE in 1h 45m 03s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/c361d4f61907420d9acf1f24cbd4d297 : FAILURE in 57m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"14cf2cb38b666bdc5e0ff06ee45acfe11feef7ed","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-08 15:39:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4c9274915c45ce93368c0443a8b7160625e04703","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 17:35:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/60bdcaa303ad4ca4badda4b4fdbfba9f : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c6f0e2b17806451c983b9ee2665651b2 : SUCCESS in 1h 01m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b714e50b3cb04035888606631931d303 : SUCCESS in 1h 51m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/15ecca0079c148b5beb63029d0180097 : SUCCESS in 1h 08m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"622a183e88922e50087e3d882521b15342121df1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-08 18:06:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/8a9992c569984230b746a9a17eec6aba : SUCCESS in 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"963104dcbca25212f2f92658725956d9b645229b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 19:21:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/477bceede7624b1894bc1a8c6e7fef26 : SUCCESS in 2m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/77dcdbc9c21448d3af4e01f3471442da : SUCCESS in 46m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b574e30ca494470ea0456f5f3b641d0e : SUCCESS in 27m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/eab41e1ed77b42eab09d3fe3e045341b : SUCCESS in 1h 15m 27s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/47a96c8eeecf4c248221878d0c28b691 : SUCCESS in 1h 13m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ceda4db912a5464b885059c37b9d9614 : SUCCESS in 1h 16m 36s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3e75c29ea01d45428bc4d17dfa828d66 : SUCCESS in 54m 14s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/5475c603ea5048e89ccc5406ebe81b31 : SUCCESS in 36m 23s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/7d31ad1f85db403995e8d922f10b9e08 : SUCCESS in 1h 50m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d51c451419c5449dbf29d5fa14de10a9 : SUCCESS in 2h 23m 41s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/b35b47751e214026a3e9cea5acb24879 : SUCCESS in 2h 07m 39s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1f904c2e9020491394b07e9148254a70 : TIMED_OUT in 2h 13m 54s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/3b658185e6ed429699c363190a08d6ce : SUCCESS in 1h 20m 53s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/77085af08560423a9736348b6f0a34d9 : SUCCESS in 1h 52m 13s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5b0798d20a984bf8a923ffe31633bb9b : SUCCESS in 39m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c2cca117765848a094e711d10842d4b6 : SUCCESS in 36m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d3db893818b146ad85e9c8e9bd2eb9ae : SUCCESS in 1h 03m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f76e46e1c2454a8eabbcd0cd6e30b423 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538a030efafa4fd5a005cc2e64f3ffd4 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2207bae6637e4579bb118c4660cdae12 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da386e0d54084936a713e142a3b10716 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d2ca1b62944f5dbd14cde10c91b5fc : SUCCESS in 7m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4200e2187e741fbb8b0eab9fdbe21a0 : SUCCESS in 24m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/07b0d3994042477b8a20e31fae520b3c : SUCCESS in 26m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/80ca8a166fa84fdbb2b9c0bfd0f8f8e6 : SUCCESS in 1h 05m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c4b320e0bf8b401ea0b17d4f4866fb82 : SUCCESS in 59m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e3829631e6924558975879338bfb6c4c : SUCCESS in 1h 14m 12s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5428fcb45d4a4b4bb9ebd1fe633aa818 : SUCCESS in 1h 34m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7f929ccc39674fdd8d5b37cdcc6f143c : SUCCESS in 2h 24m 05s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0db0b29b50b249e18b65028e45cffc97 : TIMED_OUT in 2h 10m 32s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/fb3edfa6caa64428a7943a2d79df90f1 : SUCCESS in 1h 11m 10s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/926d5320c8544abc866ebd672eac8c78 : SUCCESS in 1h 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5dfd5ca42e72e8fa5a3854d4207ec0d5cbaab0b2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-08 19:28:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f0ed300db927e32b3936b3e607e54d3d09539ccc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-09 08:39:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nplease also amend the currently existing reno according to comments left on that previous change (be explicit about removal and mark as \u0027upgrade\u0027, not deprecation)","accounts_in_message":[],"_revision_number":4},{"id":"d86455f7f8299b098ff1763e7c067dffb2dd2a7e","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-06-16 14:29:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"59581ec019bdd7099d6a2920260278f360a7b66d","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-06-16 14:44:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bc5b187ea71f819bcfa4a45659d4efd5cf157a32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 16:04:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/83c45694880e4342bc52a20c1ade63fa : SUCCESS in 57m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"577d1997fc86bfc4339709a12784cb9e66aad900","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-16 19:01:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/43f8231fed714cad9a977852283014be : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f10763d5d81a83fae1018b0409e084823f677fa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 20:50:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9fea46f5425847af8feda2838c846af6 : FAILURE in 4m 19s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/aabcb3bd15c04a728f97a263139d4e16 : SUCCESS in 26m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/51431f329af24b68897f2cd958ba2d7a : SUCCESS in 31m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fcadb85c21794a7f932a42e8f281a758 : TIMED_OUT in 2h 07m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/21e76fb3a57349af8ef7966391cc3abe : SUCCESS in 2h 02m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b1c5f548332c46b69050cb41d6e62c3a : SUCCESS in 1h 20m 23s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/92d28165223b4704be048b2d75d57655 : SUCCESS in 49m 15s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/82559f85b68b4738880b8d1daeedaac5 : SUCCESS in 49m 34s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5a3eb2e6cbcd4456a51e5242ba2e0bae : SUCCESS in 2h 03m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/95f30d3b34fd479a91c694359adf802f : TIMED_OUT in 2h 44m 07s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/49e565ca4b76420488b5f9a1b34e4f2f : TIMED_OUT in 2h 15m 21s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/751e08ad8925400399e0332c4038db67 : SUCCESS in 2h 04m 03s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/436bf4bad45748e3b41f7f7621d0201a : SUCCESS in 1h 27m 07s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/2539488bbfc148919ca67979f289da27 : SUCCESS in 1h 21m 04s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bb0847e930c648ee9f77366e881dee34 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bce1e20cce95449eb9880ac8abc2b09f : SUCCESS in 1h 08m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f2cdc894f78a4138a844ba32f6f2937d : SUCCESS in 1h 57m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b815b0ebe74b4b5681b17b40f49c6f5c : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0ee5307f2145569f58b3cbb17da3e4 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08412f492da749f7b8e75e5a4773518c : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9137337a6cec4a4984fb9a6bb3485549 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e35b71ff79c487680c31e7189174314 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97bf451217a34518958a026925aee73a : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/14f5071688124643a9a7af28424e6a8a : SUCCESS in 1h 05m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b60bf2238c9a48af9da0a3f8bd2b07f4 : SUCCESS in 1h 24m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f692434e5834d9eaf314360afded281 : SUCCESS in 1h 15m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/caa0876481da4b90ac75792d3a1ebfdf : SUCCESS in 1h 35m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ed849cc08d754e9389e223e08ecb68c2 : SUCCESS in 1h 21m 53s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d2cdf94a2e4a42c0a5b409e4b49d9eb6 : SUCCESS in 1h 47m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/df5e769d22a94315a475cad50ae59759 : SUCCESS in 2h 30m 13s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0f41d318620d49f3b38aa2a118d8db51 : TIMED_OUT in 2h 10m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7f6b5807e2f94ad18c663532d808fa3e : SUCCESS in 1h 47m 15s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/12ee4cc9e9dc4d4cac717b40dca06a6a : SUCCESS in 1h 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"de198319da0c033ba9717e8b4490606efaa18925","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-17 05:05:13.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"b342f78a26f02470bf6deb495d7dd584bdbd3080","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 05:46:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b7457168a765d6eb3e5c4c6e56ef4a6ac0ee2c58","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-17 06:44:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b8b87f3a43d74e6884120ed8942622e6 : SUCCESS in 30m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"366fd751b7c544f28b624d82a4d67698587eb855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 07:53:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4fae9220d0b14baeacb694d93d7bfbee : FAILURE in 2m 10s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8038e0c5f3594b3cb79b95a114db344c : SUCCESS in 19m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/83397f7c7385402f983620c31dbf6665 : SUCCESS in 31m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a383ad85279a4f27b9e416edf2de4400 : SUCCESS in 1h 21m 20s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1053effb89d34a48b0fe6131b6b01d65 : FAILURE in 1h 29m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/361db781a67344a1a0d18552e3271a5f : SUCCESS in 1h 15m 34s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7d424c4b99214d638191a0ac3ddf177a : SUCCESS in 48m 34s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/94908adfa7294021a237c65638ffaa44 : SUCCESS in 1h 51m 33s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/336f8829c173465f9b399624fe857771 : SUCCESS in 1h 56m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/276cb8c63b774845ab56cd0d42e06402 : TIMED_OUT in 2h 46m 01s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/b70b12d9212d4de19f0e5ba847c27707 : TIMED_OUT in 2h 06m 16s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5c16875f14124c019f0c3083ef754085 : FAILURE in 32m 01s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/86ae20a63e6a4fec9cf390b312583889 : TIMED_OUT in 2h 04m 33s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/2aa072024bad409ca71a1f111c2158e5 : FAILURE in 1h 07m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9d27ab4ac1064538b313a5185ac164f9 : SUCCESS in 1h 48m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/983f344a504b4a47b6d27e10816e28b0 : SUCCESS in 1h 44m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6eaf36e445084cd89e44d51d9b34b802 : TIMED_OUT in 2h 05m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65a380d64c894ce79ca0f0263c1302f3 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63473b615ca0444d928785739f3152ab : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00d7642b129447d08cf841b9f221b58a : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96d3ef13b83842dca609d96ee9eaab5b : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfe48c968d041b8bd4134f79773477e : SUCCESS in 24m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf0dbfa9f3ba4e4386659a6197d01871 : SUCCESS in 6m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d6b8102057b34e8aa6f960c02f28e5b2 : SUCCESS in 1h 51m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0334471bd02a4f6489b84e17bf95d970 : SUCCESS in 1h 50m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c630c7ce9bba4101a41654498f3d6b7a : TIMED_OUT in 2h 05m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ffae40f5879641c2b730fbd074d61d2a : TIMED_OUT in 2h 04m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/79d6d2bd69624c89ba16e0cd5918d596 : SUCCESS in 1h 29m 25s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/72252d7919154da191f97c42a2331caf : TIMED_OUT in 2h 40m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2d549163f3f248aa8dddbabfa8b0b39a : TIMED_OUT in 2h 40m 34s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/64d1eee8d326490e8c5f79a5bdf359ef : SUCCESS in 1h 51m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f994e8eca7884afd93f05275b40a5a02 : FAILURE in 28m 29s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/2033cd0a325048b6aee9c31c774edb17 : SUCCESS in 1h 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d874d62e5e3c17fb4f4fbfcfe495db334ffe481c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-17 19:20:30.000000000","message":"Patch Set 7: Code-Review+2\n\nit has some conflict but looks fine now, please fix the conflict and gtg otherwise","accounts_in_message":[],"_revision_number":7},{"id":"cb9e88875562601513aac55829e900b918d9350f","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-06-19 00:48:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0d72501d18a363eb7958d18df4d9a9b5c04d5fae","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-06-19 00:49:47.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5f18d5f19417ccb5d6a4c9cec1c9dffc8f6e635f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-19 01:18:09.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/369bdda1902e4aad89713214b739e96c : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3a256cfa4513c7ee37d3bbcb7707a544e1cabe7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 01:41:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"eb28250d83ac10bcf7fca774e812542fb6c81e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:01:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/478ebc9990d4457d8a78bd8366e9a6fa : SUCCESS in 3m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a7c5a5d94be048e281b4c6fb5e3c916f : SUCCESS in 23m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6a7f365aed1c44c5911cec2727d6b663 : SUCCESS in 32m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/04a49f6e48954422b6ece70c86c6d87a : SUCCESS in 1h 22m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7ab431f1be3d40e9b5ac0a1a0078e046 : SUCCESS in 1h 15m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b24f1cf2ae2248a6a9a5a865891e5a9b : SUCCESS in 1h 30m 07s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/76054c1baa7b400ab8a1da03eca09d58 : SUCCESS in 38m 22s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/570719c9d12245d3bb02531c063555ee : SUCCESS in 32m 59s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e7b99dd758664d22b57314d9c35f2613 : FAILURE in 1h 23m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/217faa5866a14583aa4bd39fc4e6607c : SUCCESS in 1h 56m 59s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/30a451651ba84a29961161c653b4f40c : SUCCESS in 1h 49m 57s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8c9b4b02a4c740a1956987cad735780a : SUCCESS in 1h 51m 13s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/ed0271dde2414f8fa55fdc0e154f281d : SUCCESS in 1h 41m 39s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/c68a3f019da241ba808a40efe3169022 : SUCCESS in 1h 09m 53s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6a7350f470474a8eab57a0bea7e9dde4 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e173a0160f194ecf9d25396fc610f5f0 : SUCCESS in 16m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e80c161ed4d7454cae8bec14570f005b : SUCCESS in 1h 03m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e24f72f398164da3a0b494fe3a699851 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fedcb2fd69814376b95885316e2548a8 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9127523de7b84a7dbdfcca56e3d9f5de : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1495aafd97d04ffcb9d2c201be59d85d : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04100fda4414733bd1afd0a2401124b : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ccdc0dbfd564a9ea15e5f2d4f6ace08 : SUCCESS in 4m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/019a0067f6c8456bb1d72d420f09d1c3 : SUCCESS in 20m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0fd178a2e3924c13a592f4d1c368d493 : SUCCESS in 59m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1aec0996f7ae4b1fb5bf9c839a0bba19 : SUCCESS in 1h 01m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/faf2c53bf4c949f48a58294ca384ea74 : SUCCESS in 1h 24m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/41c7e5566d7d4f9e82ecbe9581f34345 : SUCCESS in 1h 12m 34s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/734b79cc8de8454dab63938d6a684a3d : SUCCESS in 2h 29m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/491c1cdf9283460c860af6296452c694 : SUCCESS in 2h 04m 29s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/64643927618c4eff860dbb917515536d : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/87c20f39d43a4bb48840b1297dacb8c0 : SUCCESS in 1h 37m 51s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/89d6c3f993a14453a044a1307993a712 : SUCCESS in 1h 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8e218253d18e1af79ab69185fa423d125114e353","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2020-06-19 05:36:59.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"6b6eee39c20fc433c4694dc054520540bd36dad6","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-19 06:45:02.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"c2e054db4e64fe932803aa224a853ae627289419","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-19 08:04:10.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"2eaf0bbe19c7b2b231559b9b575c738e7b488503","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 08:04:24.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"539d42e4d18a8f2a1987ec5209d12a65d5269bb2","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-06-19 10:01:37.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"09a6c12de7370456bfbc0a3b99bed68429c80a13","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-19 11:01:31.000000000","message":"Patch Set 9: Code-Review+2\n\nCI is read, please hold","accounts_in_message":[],"_revision_number":9},{"id":"e19d831b30e337bf8e1c26ffccb152d675c00d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 11:22:28.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2eacbec2cd9242318d5e2e078bb36aa0 : SUCCESS in 3m 12s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e085289af1e94968af8e13080d5c731c : SUCCESS in 29m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/04aeca89f4f0461197636faf4fdbbfb9 : SUCCESS in 53m 29s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1cb6966f3d8c4f50bcaacfe0bbdb49f5 : SUCCESS in 1h 04m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/341aa81fa80349969440fb3f0fa7e7b9 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d23025e82e4c1c88b4563f9bc44d19 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95d0497380f9465693ba517ccd2f0202 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d52284ac2014a44b80f2024fc6ac413 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d05f25fc0f47c88388050c63c057d9 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cff3f2382ec489f8615b2efd6c57510 : SUCCESS in 5m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb9e330b24904c4393d371ab195748e4 : SUCCESS in 20m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1077f0286d7b46e0a66161da84179463 : FAILURE in 37m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/af9f4eb265f749daa70245ff137a76ac : FAILURE in 40m 43s","accounts_in_message":[],"_revision_number":9},{"id":"253ca0c78334084bd0ff78e9c4eae98ccd8e1227","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-19 18:13:53.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"24906e6e232e3f6c634d8472f27f50aa295a672a","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-19 18:14:00.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"7b6f538b903b341ed1d735296e0eda6c1b64fe10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 18:56:14.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a5fee3cb5153c8c76710bb159545cd227cb2c08b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-19 19:20:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/61a3c51d4b0f466fb66c3b31776f7232 : SUCCESS in 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e7a02eff2289854abc34237ffa1bc5eb98dba488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 23:23:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/41087a18c14b439e988275c02cb83bd5 : SUCCESS in 3m 39s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8eb63de91e864a4399038c6580b84a0e : SUCCESS in 23m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c53d358c9a7c4e55845fd7a3c0195e18 : SUCCESS in 36m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f3f3bb5ee06143678ded35061506dec3 : SUCCESS in 1h 19m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a7db8e97cd4142e7b4f332b336896d58 : SUCCESS in 1h 08m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c2cb40a8c45742589bfd05d8bb19c685 : SUCCESS in 1h 33m 15s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/14906f2d601d4fb7835b1402664cd61a : SUCCESS in 43m 37s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/cb24ae0c511443dda1e5c89336d0b3f5 : SUCCESS in 28m 48s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e2afa754053b47ada5a84bde355d3dae : SUCCESS in 1h 56m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1e22bdf5c75d4497880af5c630b47002 : SUCCESS in 2h 36m 34s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f0bad47d023141e8bc1c4fa78ca89a7e : SUCCESS in 1h 55m 36s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/def615cd0a7b45acadec158010976021 : SUCCESS in 2h 04m 49s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/ded9e37f08844c49baeb4b5c4470c974 : SUCCESS in 1h 12m 37s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/a5ae61eff25b4350b6a83ec6ce8638c7 : SUCCESS in 1h 53m 25s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d0787e6039444f3499c4e1956ff06323 : SUCCESS in 18m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a7f12cfd94c449ee8adda639715270ea : SUCCESS in 17m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7b294c0389a9445bb62247c1b9a048f0 : SUCCESS in 1h 06m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7e177a48af34a11b505c4acd756ff29 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e0e51849ff4e58b27944b85003ff89 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/144cda8bd5184dc4acc9b98ac0798a3f : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b46af80e30f1440db2126359ae00d24e : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a98c29b3c343cc9fddba4f0904702b : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e041536caef4ce5b346db73842a5021 : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c87ce8029aa541bf894e6ff782c31638 : SUCCESS in 22m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b99b1d79d277424a9195aefc335507b2 : SUCCESS in 29m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc79e37227234dabbebf684e10df7c63 : SUCCESS in 1h 02m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e7416bffde847f5814b6e6d0f7f9eb7 : SUCCESS in 1h 09m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e4d88d9b5861429e9880cabd234bda80 : SUCCESS in 1h 14m 19s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/015bcaff59364534aceca398135438a6 : SUCCESS in 1h 11m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/18a0970e6eb34c718086dba5cbdb1a19 : SUCCESS in 1h 42m 58s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d6bfab24f74d4893bf265a25dd9d915b : SUCCESS in 1h 36m 43s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/63d6b659ec1142d6ba2825f3b5cca8ed : SUCCESS in 1h 04m 35s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/0cd17b03cda541358af4c142bd1e90ab : SUCCESS in 1h 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bcf1bbdc9ce6bcbfac60d0a3a0df9fbb09fed6dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 23:23:27.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"c6d6bff5a6af497323950474bc12c8584bb772cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 00:57:15.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ae750719300b4b8fbb7f3e97b8986cd4 : SUCCESS in 2m 37s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/70b8142505bd40be8e1c938a5f1163a6 : SUCCESS in 28m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1534191258d64c13927a2766e27f630c : SUCCESS in 1h 18m 40s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/888e3dd1ecca4619a447c54d4b788f89 : SUCCESS in 1h 25m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1caf83748b2046158549efb6f223d7b2 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6dc0762528044bf98d6942add17fd09 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a09cb3c2f4194ffaaea7d5a9bb974edb : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ffa700c44fd4379967617733e7f03da : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a190bee41935493f9ce414858d999cc2 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c85ebf1b87247b59f2dd999d2ff91f4 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/56924d9cc89b439ba82c7891850ab259 : SUCCESS in 33m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2096aa7e1881477ab03815d27c2c71c7 : SUCCESS in 1h 26m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e866c62a02d348df9d7411bc29c03600 : SUCCESS in 1h 32m 36s","accounts_in_message":[],"_revision_number":10},{"id":"56f4badfba878cb6ff516aae00015750491b658f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 00:57:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"79a3f1e5f317bc7ee0bbbaca3fdf64ec23bde1d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 00:58:52.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7134d21c28477b90cfed2922668425 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abc65084f644406991c70f993ac2317a : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e332891a8af55cecfcdb1b96932029690b63037f","revisions":{"491d1103a9e9cf963677c2c3c9f60dd1f9b864c6":{"kind":"REWORK","_number":1,"created":"2020-05-20 01:04:41.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/60/729460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/1"}}},"commit":{"parents":[{"commit":"105053f2aba98af40bc7a508fd86ee18ec3a8edd","subject":"Merge \"Tidy up release notes for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/105053f2aba98af40bc7a508fd86ee18ec3a8edd"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"subject":"[WIP]Remove the congress images since it has been retired","message":"[WIP]Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/491d1103a9e9cf963677c2c3c9f60dd1f9b864c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/491d1103a9e9cf963677c2c3c9f60dd1f9b864c6"}]},"branch":"refs/heads/master"},"727c102913d3c21258518725774d8f115daabf4a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-29 05:41:13.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/60/729460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/2"}}},"commit":{"parents":[{"commit":"834ffd912a130c552cbe1ff9a416dea95599a3e4","subject":"Merge \"Make Cyborg buildable (again)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/834ffd912a130c552cbe1ff9a416dea95599a3e4"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yingjunz@gmail.com","date":"2020-05-29 05:41:13.000000000","tz":0},"subject":"[WIP]Remove the congress images since it has been retired","message":"[WIP]Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/727c102913d3c21258518725774d8f115daabf4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/727c102913d3c21258518725774d8f115daabf4a"}]},"branch":"refs/heads/master"},"bb6e00bfc609d564e6e07f130b3a2334ce09106e":{"kind":"REWORK","_number":3,"created":"2020-05-29 15:02:58.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/60/729460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/3"}}},"commit":{"parents":[{"commit":"834ffd912a130c552cbe1ff9a416dea95599a3e4","subject":"Merge \"Make Cyborg buildable (again)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/834ffd912a130c552cbe1ff9a416dea95599a3e4"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-29 14:58:49.000000000","tz":480},"subject":"Remove the congress images since it has been retired","message":"Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I12d74b53b64268d0358220af35f7fd719d82b397\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb6e00bfc609d564e6e07f130b3a2334ce09106e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb6e00bfc609d564e6e07f130b3a2334ce09106e"}]},"branch":"refs/heads/master"},"0ea1860dcdf68bcffe48b3c847f9299077d657d2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-31 14:27:55.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/60/729460/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/4"}}},"commit":{"parents":[{"commit":"6815f917077a3494be04420e3906dde1d5fe68fd","subject":"Victoria: Add releasenote for remove congress images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6815f917077a3494be04420e3906dde1d5fe68fd"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-31 14:27:55.000000000","tz":0},"subject":"Remove the congress images since it has been retired","message":"Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I12d74b53b64268d0358220af35f7fd719d82b397\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ea1860dcdf68bcffe48b3c847f9299077d657d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ea1860dcdf68bcffe48b3c847f9299077d657d2"}]},"branch":"refs/heads/master"},"c56485e9fb6bb07ab41d98762dcabf8e9dceba89":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-06-16 14:29:23.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/60/729460/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/5"}}},"commit":{"parents":[{"commit":"14ef775c6ea849964d9afa03575a415687d9b51c","subject":"cyborg-api: install proper WSGI for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14ef775c6ea849964d9afa03575a415687d9b51c"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-06-16 14:29:23.000000000","tz":0},"subject":"Remove the congress images since it has been retired","message":"Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I12d74b53b64268d0358220af35f7fd719d82b397\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c56485e9fb6bb07ab41d98762dcabf8e9dceba89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c56485e9fb6bb07ab41d98762dcabf8e9dceba89"}]},"branch":"refs/heads/master"},"1739a3491407faa554e9c1f2c419d515f3f44503":{"kind":"REWORK","_number":6,"created":"2020-06-16 14:44:15.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/60/729460/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/6"}}},"commit":{"parents":[{"commit":"14ef775c6ea849964d9afa03575a415687d9b51c","subject":"cyborg-api: install proper WSGI for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14ef775c6ea849964d9afa03575a415687d9b51c"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-06-16 14:39:51.000000000","tz":480},"subject":"Remove the congress images since it has been retired","message":"Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I12d74b53b64268d0358220af35f7fd719d82b397\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\nCo-Authored-By: jacky06 \u003czhang.min@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1739a3491407faa554e9c1f2c419d515f3f44503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1739a3491407faa554e9c1f2c419d515f3f44503"}]},"branch":"refs/heads/master"},"1fd9a00205026443c2bfae80d36c025d59967345":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-06-17 05:05:13.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/60/729460/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/7"}}},"commit":{"parents":[{"commit":"2b140296fc7511c85740a7480019af5e4abe8332","subject":"Merge \"base/centos: take care of whitespaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b140296fc7511c85740a7480019af5e4abe8332"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yingjunz@gmail.com","date":"2020-06-17 05:05:13.000000000","tz":0},"subject":"Remove the congress images since it has been retired","message":"Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I12d74b53b64268d0358220af35f7fd719d82b397\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\nCo-Authored-By: jacky06 \u003czhang.min@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fd9a00205026443c2bfae80d36c025d59967345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fd9a00205026443c2bfae80d36c025d59967345"}]},"branch":"refs/heads/master"},"9e190ebbc50f7e1aadba78559b95878835a0160e":{"kind":"REWORK","_number":8,"created":"2020-06-19 00:48:25.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/60/729460/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/8"}}},"commit":{"parents":[{"commit":"039a6d431ae1b2ddfb7b593c1c2d10bb58556b9e","subject":"CI: fix line merge in CentOS builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/039a6d431ae1b2ddfb7b593c1c2d10bb58556b9e"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-06-19 00:47:36.000000000","tz":480},"subject":"Remove the congress images since it has been retired","message":"Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I12d74b53b64268d0358220af35f7fd719d82b397\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\nCo-Authored-By: jacky06 \u003czhang.min@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e190ebbc50f7e1aadba78559b95878835a0160e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e190ebbc50f7e1aadba78559b95878835a0160e"}]},"branch":"refs/heads/master"},"b214681462cb82f6d48d422e916afef909856fe7":{"kind":"REWORK","_number":9,"created":"2020-06-19 00:49:47.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/60/729460/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/9"}}},"commit":{"parents":[{"commit":"039a6d431ae1b2ddfb7b593c1c2d10bb58556b9e","subject":"CI: fix line merge in CentOS builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/039a6d431ae1b2ddfb7b593c1c2d10bb58556b9e"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-06-19 00:49:02.000000000","tz":480},"subject":"Remove the congress images since it has been retired","message":"Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I12d74b53b64268d0358220af35f7fd719d82b397\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\nCo-Authored-By: jacky06 \u003czhang.min@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b214681462cb82f6d48d422e916afef909856fe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b214681462cb82f6d48d422e916afef909856fe7"}]},"branch":"refs/heads/master"},"e332891a8af55cecfcdb1b96932029690b63037f":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-06-19 18:13:53.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/60/729460/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/729460/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/729460/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/729460/10"}}},"commit":{"parents":[{"commit":"c42b882aff3d8c911157b52f705dc755294f23f6","subject":"CI: use wheels from infra mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c42b882aff3d8c911157b52f705dc755294f23f6"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-20 01:01:30.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yingjunz@gmail.com","date":"2020-06-19 18:13:53.000000000","tz":0},"subject":"Remove the congress images since it has been retired","message":"Remove the congress images since it has been retired\n\nmore info: https://review.opendev.org/#/c/721733/\n\nDepends-On: I12d74b53b64268d0358220af35f7fd719d82b397\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I7bd03b4f70ddfb15c96ed68e659cb6dccb0c36a3\nCo-Authored-By: jacky06 \u003czhang.min@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e332891a8af55cecfcdb1b96932029690b63037f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e332891a8af55cecfcdb1b96932029690b63037f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
