)]}'
{"id":"openstack%2Fkolla~729542","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I8a802617952f80cfb9f906398054ef6d17559589","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I8a802617952f80cfb9f906398054ef6d17559589","subject":"Ensure the confs exists for cyborg and masakari","status":"MERGED","created":"2020-05-20 08:54:47.000000000","updated":"2020-05-23 08:06:56.000000000","submitted":"2020-05-23 08:05:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729542-1590221133268-f4975ef0","meta_rev_id":"72204c50085a5410f839baae57fb3340b850b20b","_number":729542,"virtual_id_number":729542,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-23 08:05:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"recommended":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"all":[{"value":1,"date":"2020-05-20 15:19:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2020-05-23 06:01:53.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-05-20 08:54:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2020-05-23 06:01:53.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 15:19:45.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2020-05-23 06:01:53.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-23 08:05:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5af67a0b9260edfc2af1052d677eb118ab59734","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-20 08:54:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7e92dd1f2cb2a9ffc851a4d3eec2babef393ed31","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-20 08:54:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a3eb12f15d6eb377062241b13c3355b680a63a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:58:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/11827a2a06f44f40986a16322897746f : SUCCESS in 41m 00s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e461510c722049efbc8deeb3672381f0 : SUCCESS in 52m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/054fba2b75324f2d880fd2a355df5018 : SUCCESS in 29m 59s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/65dd28247f1749bf8797c6fd6c28aad0 : SUCCESS in 31m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/36ababdb5d3342beae00f37abb073f28 : SUCCESS in 1h 32m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c6f42b93b55a4684bd20d1442093b7c4 : SUCCESS in 1h 09m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/98db330ef6474455bebdc8f20302fab4 : SUCCESS in 1h 23m 33s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/79badbbc332e4037be2f225e095592b2 : SUCCESS in 1h 09m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4a1662455f274f73bb642e5ea5c1224c : SUCCESS in 2h 00m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2b17625edefe4717beb0987af9b373b1 : SUCCESS in 1h 38m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ae57d5e6feb34c889bfd459d32376b40 : SUCCESS in 29m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9373f3784f7e4b119e08c63483161407 : SUCCESS in 21m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f54f9de0e1a48438b11011fe5b09b60 : SUCCESS in 1h 04m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68c3d60499c14bcea8931d81c3204e5c : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455beb6797f5445788f81e3e3c599e8a : SUCCESS in 7m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1cc07114f1a84a5a82e852184d2637ae : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c8c5257ee594692b577027372c39452 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d070c2ac54041a5abd2412e345bae17 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98fd0d71896450b943750da7cd6074e : SUCCESS in 10m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef71ab7ff4774729a8cbbe25df10c336 : SUCCESS in 28m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d6b1e07b80f4571a1dbaf3991ac7592 : SUCCESS in 26m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92941666961e4b939b1e4d6fbedd8591 : SUCCESS in 59m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d4df96f7cb74b46b48318bb89b90e8f : SUCCESS in 56m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/71ba29dc7cbb4f8db5a0d5816a9e74b2 : SUCCESS in 1h 11m 11s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"7596f9623ed0f6036eecf1ee5f86b1debfafffb4","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-20 15:19:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ef2cbfb29d1eb0f74dd61ed5c74ee6bab15af7cb","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-23 06:01:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eef1c041eac3cd973e43149649a43edd57fa3ad6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 06:02:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1264829216ae457c041d334b61501329a0d2acfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 08:05:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e5bec6c1c420c03cf4662ba4527b9bbc5ca8d056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 08:05:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/574b1546a87147bcb0c8fbde525571f6 : SUCCESS in 1h 03m 27s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4d780c064b614e8d920486ba21ebf42b : SUCCESS in 31m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3351d2b440bc4df8ae80aacd42fc731e : SUCCESS in 1h 30m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8e4084376fb24b4388881fe8e6f9ac8f : SUCCESS in 1h 10m 17s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1b70d116507947a0b469e6c3e65700b1 : SUCCESS in 1h 42m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/028f1dc15a314979ad1787dd0311ea5a : SUCCESS in 55m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d9c1e91cb384397a78e1687471f9274 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de736d956a8b4afaa3bd8a7214bbd38a : SUCCESS in 7m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91725895866043c8b9ab6b4844e6b697 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2c8c720421b48a29b4504bba21cc48d : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4be179edcaab4d96aa0615bae5f7d294 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8a1ccf8d96b421d8e02ccb6aed73263 : SUCCESS in 10m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/94b65d74d6904707aea04d173e6f7ff4 : SUCCESS in 32m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e47ab375f4db40be8b22a8df737e426d : SUCCESS in 59m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/db2e92fcf9974ab9a9c55dccae9bde09 : SUCCESS in 1h 13m 34s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"72204c50085a5410f839baae57fb3340b850b20b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 08:06:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d909fadc4ef343c4ac91abfd4a9af847 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2a4822fb0a1d8238ad206d523a56e8965fef826","revisions":{"d2a4822fb0a1d8238ad206d523a56e8965fef826":{"kind":"REWORK","_number":1,"created":"2020-05-20 08:54:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/42/729542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/729542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/729542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/729542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/729542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/729542/1"}}},"commit":{"parents":[{"commit":"e6f30e87020383f27c17560d0bf544c87520af84","subject":"Merge \"switch to official CentOS 8 repositories\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6f30e87020383f27c17560d0bf544c87520af84"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-17 14:19:02.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-20 08:54:47.000000000","tz":0},"subject":"Ensure the confs exists for cyborg and masakari","message":"Ensure the confs exists for cyborg and masakari\n\n1. When cyborg images build completed, the /etc/cyborg folder is not\nexits, and there is not confs, especially the api-paste.ini\n2. When masakari images build completed, it\u0027s configs is missing,\nthis ps add them together\n\ncyborg should keep consistence with the other images[1], this PS\nto add it.\n\n[1]: https://github.com/openstack/kolla/blob/master/docker/nova/nova-base/Dockerfile.j2#L137\n\nChange-Id: I8a802617952f80cfb9f906398054ef6d17559589\n(cherry picked from commit 2e6d2f144cd2de18193d9d87d36c95dfb75fb1b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2a4822fb0a1d8238ad206d523a56e8965fef826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2a4822fb0a1d8238ad206d523a56e8965fef826"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
