)]}'
{"id":"openstack%2Fkolla~729864","triplet_id":"openstack%2Fkolla~master~Ib6a6f38efd9860851e558f13ee3bdedd8bb57903","project":"openstack/kolla","branch":"master","topic":"bug/1873744","hashtags":[],"change_id":"Ib6a6f38efd9860851e558f13ee3bdedd8bb57903","subject":"Make Cyborg buildable (again)","status":"MERGED","created":"2020-05-21 08:11:36.000000000","updated":"2020-05-27 17:32:52.000000000","submitted":"2020-05-27 16:36:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729864-1590597409465-06556da8","meta_rev_id":"e7c76a4f75936272b29f0ad07051fd8fd99439ea","_number":729864,"virtual_id_number":729864,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-05-21 08:19:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 16:36:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-21 08:46:56.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"all":[{"value":2,"date":"2020-05-27 10:52:28.000000000","_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-05-27 15:01:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-05-27 15:01:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-05-21 08:17:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-05-27 17:32:52.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 08:11:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-05-21 08:13:04.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2020-05-21 08:13:07.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2020-05-21 08:13:17.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-21 08:46:56.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-27 10:52:28.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"},{"updated":"2020-05-27 15:01:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-27 16:36:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd02e5696f1e02d6412dacbaeed28e3ca39ff44d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-21 08:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a12100f77aeaca88db0351c8e6960b200c21a85d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-21 08:11:50.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"8612099ecac851573ab30499e62aa347b2f10841","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 08:16:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a2bb8f06dab045a484a9f3c5239aecc6896c0651","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 08:17:21.000000000","message":"Patch Set 2: Code-Review+1 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"3079bbbf9669f8c33f8f31643f837b3304283b6e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-21 08:18:38.000000000","message":"Patch Set 2:\n\nthanks, Mark, good catch","accounts_in_message":[],"_revision_number":2},{"id":"ccef7c37ddc5557ab326f5d06e93af358df857bd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-21 08:19:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ab5e11bfe4516b10f05c8bffcd9341be0d014cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-21 08:46:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/c023da48ab104814844d3ed53db36f73 : SUCCESS in 28m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"82e4ae815894aefce358646f61175167a92cf369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 10:13:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/96567bab411a433e98f788743a092840 : SUCCESS in 46m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7c00c34bb17543a18169b078538b782d : SUCCESS in 25m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8517b1a646aa4b67ad814bf7e63384ec : SUCCESS in 1h 17m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/23a1825c7c234e358ee05447104cf5b3 : SUCCESS in 1h 20m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/16102a2eb25741a18b1c7214b1a991f7 : SUCCESS in 1h 55m 36s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/71627454527742938b3a5435a07ce1d2 : SUCCESS in 48m 04s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ef419fbf87074a9f871b4a4f6b7d6155 : SUCCESS in 56m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a5900e59c35e40569edf59cf2f420f61 : SUCCESS in 19m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/449c7af5006f42b183e13c190ef3a8a9 : SUCCESS in 1h 10m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/881ded8dc1e543e4b9c88174fda496d7 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7da62f2d346481c9a0cf3a8e6c115b7 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/711ec856eca34510a4b90ae89b92c361 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38afd8c16b244e99b2a8bfd9435f2b9e : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3804169271647e78670c061a76b90d3 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf5ccc2a46341e9b11b6c394cd20a99 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b49f3a3e90af4e4bb09f777afe2c8359 : SUCCESS in 6m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be509ec6bad445e09abfb4065b18d533 : SUCCESS in 29m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f69979e3e8d54ce18c4ca76d35a10956 : SUCCESS in 25m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aca4a95628d34b13a957cfaf2f0e65fd : SUCCESS in 1h 02m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/864a582f35b045b8b20a21e928ca90f1 : SUCCESS in 1h 10m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1d7840b74bb340a7a7d5fa3c7ecc61bc : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d33aef67a9c7fcc4bf4f8264f277e963c4a0865d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 10:14:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b15ab2c257b048469784b3305d5005e9 : FAILURE in 10m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9e5b49c0fcb4e65ab737b098de88ea9 : SUCCESS in 1h 04m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4484b8ec5fb44f77bc3e194b15f07273 : SUCCESS in 1h 52m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/716de380096340a091fecabb76d2303a : FAILURE in 55m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1c930e7be5d737c004957c5d7c2ed6fa4056fa9","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2020-05-27 10:52:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cfa9a1eec5baf1d6cc7c52a80df7f9f3ec89e3d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-27 15:01:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9513a7dceb24ea40e7bd2983b1ec288dd278d96d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:02:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ab6bdf699afdd4054cb15b20381600ba5702f118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 16:36:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ca1f5d5759b246bca1a96eef67d08be5 : SUCCESS in 34m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/75583291f55a42b3a0e9da889bbb5f1d : SUCCESS in 1h 28m 38s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0e018631663045b18a2489d9753be710 : SUCCESS in 1h 30m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf281d7c216f4692b859d18229c14993 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/842e7235b9d6458e80c72f1d6d9e8ec6 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00243d45f1994528801bb5e5ced890a0 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c080960dfcd9411cbff2eb1208ae7fce : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949acac721e94d7896f8559ffe0519da : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3781220be8ca4b818209a6f86ecdb1c5 : SUCCESS in 4m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/318440117a144fa496f7698f9a5f52e7 : SUCCESS in 24m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/638b429a528e41e6bb98924d63f314c7 : SUCCESS in 1h 00m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d56952a8f95426b818632c796d166fd : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":2},{"id":"3d7f49f5ba78f011cf789e8220f788c46e9392e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 16:36:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7a0bb28cd44d3e5dda4b685d41546dbb9b8f4171","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 16:39:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44197df9d00c4ddbb230512d9f21cca9 : SUCCESS in 1m 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fe2b327242a403288ce3d6ddf4bafc0 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ebf85911c7959bd4ebd01fe15b8bee2dccc239e4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-27 17:32:52.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit a08ed2adbf079f11e81df6c41ab866cb5ee40266","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a170353b47d359e39f29cd951f73507b07fefb04","revisions":{"1fd4802f61904b465421c3e6d374bb47e7481873":{"kind":"REWORK","_number":1,"created":"2020-05-21 08:11:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/64/729864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/729864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/729864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/729864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/729864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/729864/1"}}},"commit":{"parents":[{"commit":"1ebb97c9f266e9be222d53d423031c75b2eb3ea4","subject":"Merge \"Switch to Ussuri release of ovn-octavia-provider source tarball\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ebb97c9f266e9be222d53d423031c75b2eb3ea4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-21 08:09:10.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-21 08:09:10.000000000","tz":120},"subject":"Make Cyborg buildable (again)","message":"Make Cyborg buildable (again)\n\nThis is done by removing the dependency on OPAE SDK.\n\nAdditionally, cursive is no longer tried to be installed again\nin the agent image as it is already installed in base via regular\nrequirements.\n\nDetails in reno and the referenced bug report.\n\nChange-Id: Ib6a6f38efd9860851e558f13ee3bdedd8bb57903\nCloses-bug: #1873744\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fd4802f61904b465421c3e6d374bb47e7481873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fd4802f61904b465421c3e6d374bb47e7481873"}]},"branch":"refs/heads/master"},"a170353b47d359e39f29cd951f73507b07fefb04":{"kind":"REWORK","_number":2,"created":"2020-05-21 08:16:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/729864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/729864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/729864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/729864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/729864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/729864/2"}}},"commit":{"parents":[{"commit":"1ebb97c9f266e9be222d53d423031c75b2eb3ea4","subject":"Merge \"Switch to Ussuri release of ovn-octavia-provider source tarball\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ebb97c9f266e9be222d53d423031c75b2eb3ea4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-21 08:09:10.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 08:16:55.000000000","tz":0},"subject":"Make Cyborg buildable (again)","message":"Make Cyborg buildable (again)\n\nThis is done by removing the dependency on OPAE SDK.\n\nAdditionally, cursive is no longer tried to be installed again\nin the agent image as it is already installed in base via regular\nrequirements.\n\nDetails in reno and the referenced bug report.\n\nChange-Id: Ib6a6f38efd9860851e558f13ee3bdedd8bb57903\nCloses-bug: #1873744\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a170353b47d359e39f29cd951f73507b07fefb04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a170353b47d359e39f29cd951f73507b07fefb04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
