)]}'
{"id":"openstack%2Fkolla~729893","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ic1b74ccc7c8e6ee2e4c025e8b29150d1a1d96f2f","project":"openstack/kolla","branch":"stable/train","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"Ic1b74ccc7c8e6ee2e4c025e8b29150d1a1d96f2f","subject":"qdrouterd: ignore failure of disabling epel repos","status":"MERGED","created":"2020-05-21 10:24:05.000000000","updated":"2020-05-21 15:04:54.000000000","submitted":"2020-05-21 15:01:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729893-1590073302213-f6fe6744","meta_rev_id":"e5b882d96d807cd646b56691783205996d50ddf9","_number":729893,"virtual_id_number":729893,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 15:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-05-21 11:39:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-05-21 13:05:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-05-21 13:00:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-05-21 13:05:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-05-21 10:28:01.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2020-05-21 13:05:43.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 10:28:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-05-21 11:39:20.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-05-21 13:00:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-05-21 13:05:43.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-05-21 13:05:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-21 15:01:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ba32f1ea2f24bcbdccea1dfec31744a180325de","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2020-05-21 10:24:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec89acac49c2313996aa04d009668792039f20a7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-21 11:39:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8fb631913098100ef4350cad48e71a7bb30da787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 11:54:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/19d12bb179164bfd8cec785a6c910b47 : SUCCESS in 32m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9aa4f6edc124453ba2a75bc2ed2928db : SUCCESS in 45m 10s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/48178cb002064c45ba389dcb2041160d : SUCCESS in 25m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/dad6ffe517044e9db6a8e90c883beb17 : SUCCESS in 29m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c580421dfbe44164a87a6f55876e455d : SUCCESS in 1h 18m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/52bf1ad819f74342931b68fa4d0d1d0d : SUCCESS in 1h 23m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ead6b352c5dc43d285f23d4dec04053e : SUCCESS in 1h 29m 06s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/306af61d61f748dda4322792d878f4e9 : SUCCESS in 1h 17m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2c367c07d72b46f7963a58d5ac3a55ea : SUCCESS in 1h 23m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/312cfff71a094d06bab87d00be870907 : SUCCESS in 42m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/59543efb799c46928e8356c9ec73ffdc : SUCCESS in 23m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2e3b48c963e5455996bd75065108175c : SUCCESS in 18m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/96705e92b789475eb8555b3d3b48c6c2 : SUCCESS in 49m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afb9e55348c94e41b36d3841d9bcc39d : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a714058a674df193ac588c3afc0c06 : SUCCESS in 7m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/abda3d814f064a7aac59a156ed6af482 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8d6c1e00a8d4ec3aebb929551784f0a : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58fccb912c2f4cac8371bd97e5eb495f : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f35b4b7c3d44e2599ea2b61b6280703 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d45f4c84f9db41ec9dfc6a631bab0f62 : SUCCESS in 22m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1599a318c1ef46e8b70eb88ee3f544d5 : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de4d1adc6437429fb56f4b136275f2d6 : SUCCESS in 52m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/463979dd46ef45e6ad25c17dba34ddcf : SUCCESS in 54m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4d05d159d502486da9efc0e687b01a58 : SUCCESS in 1h 08m 53s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"8be78a6b033e3e0d42ad7cc90e6f356c34baf5fa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-21 13:00:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f4125e122b877fbdf706dc5319556bc10efc8354","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 13:05:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3b55fd4d6b7695ef84dcf4fd72e49b3423616d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 13:05:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d5e76b25b57ffdb61eeccf001974f128329e9e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 15:01:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3fa8e85a59904140ad2dc42cf783047d : SUCCESS in 1h 12m 09s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/33a9248de19c4af6be2a171de4f5d729 : SUCCESS in 55m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0e93169943d840e4a6759c6db741eebc : SUCCESS in 1h 44m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/796e9413abf84ad0b6da3cd889da3d00 : SUCCESS in 1h 33m 31s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/22cb9429eb504b77addb7265a9bf0d46 : SUCCESS in 1h 54m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/94fc82a56a344b4dbc7f8a33000dd824 : SUCCESS in 1h 09m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a710c8eafbcd4316909d81da2ff25862 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a03f1332b9a34b99ab19f40508e08f86 : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/288e0410f7e04f9bbe708477bce44d8e : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eedec0e7130c4719b6993b76771976a9 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66181423c05a48cd8c6a66630f9a24bf : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3171e73d77f4c1fb749d1209feb98fe : SUCCESS in 10m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/247d58522dae4a5b8211ac74addfefae : SUCCESS in 37m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/57d16e81d83b4fffbc70080ca01187d2 : SUCCESS in 58m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/80403f6b9c464d83bb7836d8074dfab5 : SUCCESS in 1h 04m 17s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"d0517745c0b2ee260a6145134aa96d541fc57b3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 15:01:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"12864f9adc7a6abfbe0fe77100e3b3723c9da906","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 15:04:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00eeb88c452f4843a9e7ff0e46ff93df : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"359f61b05a4eb5614a703e021bedcbaa10e3b9b7","revisions":{"359f61b05a4eb5614a703e021bedcbaa10e3b9b7":{"kind":"REWORK","_number":1,"created":"2020-05-21 10:24:05.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/93/729893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/729893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/729893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/729893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/729893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/729893/1"}}},"commit":{"parents":[{"commit":"e6f30e87020383f27c17560d0bf544c87520af84","subject":"Merge \"switch to official CentOS 8 repositories\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6f30e87020383f27c17560d0bf544c87520af84"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2020-05-21 10:19:40.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2020-05-21 10:24:01.000000000","tz":330},"subject":"qdrouterd: ignore failure of disabling epel repos","message":"qdrouterd: ignore failure of disabling epel repos\n\nThere are cases when epel repos are not setup in\nthat case dnf config-manager will fail.\n\nLet\u0027s ignore the failure in such cases.\n\nThis change is needed only for Train branch.\n\nChange-Id: Ic1b74ccc7c8e6ee2e4c025e8b29150d1a1d96f2f\nPartially-Implements: blueprint centos-rhel-8\nSigned-off-by: Bhagyashri Shewale \u003cbshewale@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/359f61b05a4eb5614a703e021bedcbaa10e3b9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/359f61b05a4eb5614a703e021bedcbaa10e3b9b7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
