)]}'
{"id":"openstack%2Fkolla~730520","triplet_id":"openstack%2Fkolla~master~I266cc411248804ea4e6693ff768d546e2b56fb37","project":"openstack/kolla","branch":"master","topic":"kolla-set-master-to-master-victoria","hashtags":[],"change_id":"I266cc411248804ea4e6693ff768d546e2b56fb37","subject":"Revert \"Switch to Ussuri release of ovn-octavia-provider source tarball\"","status":"MERGED","created":"2020-05-24 17:42:16.000000000","updated":"2020-05-26 09:16:46.000000000","submitted":"2020-05-26 09:16:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730520-1590484606072-48d602ed","meta_rev_id":"c80031fc7f89d84be92430a33ddb1e797f79d264","_number":730520,"virtual_id_number":730520,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 09:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-24 18:34:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-05-25 07:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-05-25 06:42:53.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-05-25 07:02:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-05-24 17:57:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-24 17:42:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-24 18:34:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-25 06:42:53.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-05-25 07:02:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-05-26 09:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12a53c7febec4c089a8f9f29f0ef39bf4cf17a46","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:42:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8daf3f2bf948697c6b7f4ee869696be098bdfafd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:57:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8ae5d306f37bd0f887d3fb6c62afc10ef1315c81","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 18:00:05.000000000","message":"Topic set to kolla-set-master-to-master-victoria","accounts_in_message":[],"_revision_number":2},{"id":"b348f495d07bf3e4370c1eae9841bbb39a3a75c6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 18:34:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/2ff91c0d21a647a782d3eb12180bc717 : SUCCESS in 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ee189325b2230b0941072a95b22ed1f58c72201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 19:20:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/16c18c17c8e346dd8bc9acbc3ee610f8 : SUCCESS in 27m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4e5ea4be0fdd4790890877b17ee05c50 : SUCCESS in 26m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ecccb85dc2414a09a039215d3d120bc0 : SUCCESS in 1h 10m 22s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4b52756ea5404f49a9b88037bde98c1a : SUCCESS in 1h 15m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ca970c3011e34d248297feca86ab5156 : SUCCESS in 1h 18m 31s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b6eba761112c441bbfcfbe1b66fb5bf4 : SUCCESS in 42m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/191b1f5bc3614011a198a6acd35439da : SUCCESS in 21m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f5dfb1e221184565897e9606b37f7ade : SUCCESS in 20m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2e65e384a3ea4cd099f35901a0b100d0 : SUCCESS in 1h 08m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99ef84797d824ad08395635724be5177 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a7f206c97a4978851b3bbb531f66e5 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83fc51bd115b42e5a041d5d17f069249 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcc95ac749d640c8aad2650e6a3fad07 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b19b2c9920e4a1e8bcf49ebb45b3415 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c973908e690f4429a51fd26252893eae : SUCCESS in 24m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/83f9ef9af6f34ac3a7c1f2a339904d00 : SUCCESS in 24m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/546b336fe3c94876b5358e139a11b7ca : SUCCESS in 1h 06m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e2452c6a1c5465cbbd4225e8d81ba73 : SUCCESS in 1h 03m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/43ddc031e5a2404195132f9b9f6f2c27 : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f72e26c7e86b990ae059bc755a923b92560e5cef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 20:40:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b0686bdc20f4171b4bc70f73a1371c0 : FAILURE in 1h 16m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86fef027285641cfa51162237b281503 : SUCCESS in 1h 26m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4dbe844b0efd4152843a39c6015f0015 : FAILURE in 1h 20m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/49a6c86a32514f108c691499364b1470 : FAILURE in 1h 13m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eb0d667d3b190c946fda7c7a1e46ad1e1f3647c6","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-25 06:42:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"529c1bc6e68d2d86ce3d84b942328bce2dae013d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-25 07:02:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"221250bc29fc51e4ec99ca66f8558b03d18348ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 07:24:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d3e3c11de92ebfc681d5333e5e5bdf9769b1d36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 09:16:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b193105e31694519acd263c8d1c5b99d : SUCCESS in 30m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/64a7dcae11f7426fb52d1ddf3caed414 : SUCCESS in 1h 50m 36s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f31fcd935a8c41b3bacfd2dd171475c7 : SUCCESS in 1h 19m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e189d66795e434a86ad41cbd692901c : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d31ff9aa94245f39d3b186e16becce6 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ff5380f4c1d446fbd8d63eb1ad5f676 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b286e1ae6774503b4fb6e4c324363b0 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85cafc78a7ee4b8eb0e168f093ae46d2 : SUCCESS in 6m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d0f8c03fbfe4fe8a4bca67854dbf0bc : SUCCESS in 25m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/60b498fbb54546b4962060bfbd91f895 : SUCCESS in 1h 01m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9ad6e6d37d0349b99114c048825f4c5c : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c80031fc7f89d84be92430a33ddb1e797f79d264","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 09:16:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1697bf198b14b8e77bb3f347acf200327143d1a9","revisions":{"ceb81a14e92b3e92b284713038ebfc5634940421":{"kind":"REWORK","_number":1,"created":"2020-05-24 17:42:16.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/20/730520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/730520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/730520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/730520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/730520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/730520/1"}}},"commit":{"parents":[{"commit":"d9baa171bd1e3ba492f0870715f780a03c0699ef","subject":"Switch to Ussuri release of ovn-octavia-provider source tarball","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9baa171bd1e3ba492f0870715f780a03c0699ef"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:42:16.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:42:16.000000000","tz":0},"subject":"Revert \"Switch to Ussuri release of ovn-octavia-provider source tarball\"","message":"Revert \"Switch to Ussuri release of ovn-octavia-provider source tarball\"\n\nUssuri release. Standard process per [1].\n\n[1] https://docs.openstack.org/kolla/latest/contributor/release-management.html\n\nThis reverts commit d9baa171bd1e3ba492f0870715f780a03c0699ef.\n\nChange-Id: I266cc411248804ea4e6693ff768d546e2b56fb37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ceb81a14e92b3e92b284713038ebfc5634940421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ceb81a14e92b3e92b284713038ebfc5634940421"}]},"branch":"refs/heads/master"},"1697bf198b14b8e77bb3f347acf200327143d1a9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-24 17:57:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/20/730520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/730520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/730520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/730520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/730520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/730520/2"}}},"commit":{"parents":[{"commit":"fdfbbf77147cd8969350d57904922a9eef67fb3f","subject":"Revert \"Switch to Ussuri release tarballs for vmware-nsx*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdfbbf77147cd8969350d57904922a9eef67fb3f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:42:16.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:57:01.000000000","tz":120},"subject":"Revert \"Switch to Ussuri release of ovn-octavia-provider source tarball\"","message":"Revert \"Switch to Ussuri release of ovn-octavia-provider source tarball\"\n\nUssuri release. Standard process per [1].\n\n[1] https://docs.openstack.org/kolla/latest/contributor/release-management.html\n\nThis reverts commit d9baa171bd1e3ba492f0870715f780a03c0699ef.\n\nChange-Id: I266cc411248804ea4e6693ff768d546e2b56fb37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1697bf198b14b8e77bb3f347acf200327143d1a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1697bf198b14b8e77bb3f347acf200327143d1a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
