)]}'
{"id":"openstack%2Fkolla~730522","triplet_id":"openstack%2Fkolla~master~I45251e80aadaab3f2a540c17e8e60b086f1adfa5","project":"openstack/kolla","branch":"master","topic":"kolla-set-master-to-master-victoria","hashtags":[],"change_id":"I45251e80aadaab3f2a540c17e8e60b086f1adfa5","subject":"Revert \"Update OPENSTACK_RELEASE to ussuri\"","status":"MERGED","created":"2020-05-24 17:42:36.000000000","updated":"2020-05-26 09:06:32.000000000","submitted":"2020-05-26 09:04:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730522-1590483892254-48d602ed","meta_rev_id":"1d8725b14be19575fdce2f5e503ff0cadce8309f","_number":730522,"virtual_id_number":730522,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 09:04:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-24 18:32:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":2,"date":"2020-05-25 05:50:55.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-05-26 07:24:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2020-05-26 07:24:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"date":"2020-05-24 17:57:26.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"date":"2020-05-24 17:57:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-24 17:42:36.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-05-24 17:42:36.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2020-05-24 17:42:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-24 17:42:36.000000000","updated_by":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"reviewer":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"state":"REVIEWER"},{"updated":"2020-05-24 18:32:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-25 05:50:55.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-05-26 09:04:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"341c591f114d52c5c6578e946a0b61018b883d29","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"638552df1d277d007023d376484bbe5a5fa33cee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 17:42:48.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"e899d15ee102b2763f5c0b1f4b6b5ca42cc315f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 17:42:49.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"8dc42d425365d4ca733388244657cfd4264b45e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 17:42:51.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"a747f6620b78d1f3da291587e6631f3f352bb5a7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:45:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2fa9e3d8c50f0c5d26b6c91675970d1876479bc2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:57:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5c6c0610bc6a9dfae2334ca9e3f0ff4ad29b0773","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:59:51.000000000","message":"Topic set to kolla-set-master-to-master-victoria","accounts_in_message":[],"_revision_number":3},{"id":"d2fb1fa0bf11f1772846ccc58b66b8f6ca72f1e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 18:32:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/8f454de8df994eabb417200bf905bfd2 : SUCCESS in 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e40fed180b35a5bd648c227a967be8a70e8bee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 19:26:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a85cb9cdb626488783cc50eedbc8e11a : SUCCESS in 32m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1cbf484f659447779b59de2a15cd0add : SUCCESS in 28m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1b62ca6d956a479ca1f95a4d0f25f0b3 : SUCCESS in 1h 26m 35s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d8efce640bd1494ebbf55f6e2ee4ca49 : SUCCESS in 1h 17m 32s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/00bef5e52a504f5bb712fc7db0abaafe : SUCCESS in 1h 18m 17s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e1f34c3f724143e5a657dfad54bdd977 : SUCCESS in 44m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f087612aadc0412382d89bfa4438958e : SUCCESS in 22m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e6f21d7929654ef38ae66f3e81a0fdbd : SUCCESS in 32m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/da75aeb204a8467d93e207ef8ae54a3b : SUCCESS in 1h 02m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b6cc026c79e455b942967208f63b253 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e3d0866ee7413fbb6fa8c803b9b0b9 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa7407f8abf340cda8711c86aac0592a : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a29a005cdabc47569dd80051a76d2744 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d33b6eaac6746f4882a9ac1bf5b53fa : SUCCESS in 6m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d3045d877644f7f8e561029ee246a3a : SUCCESS in 40m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4af8b64eafb0481b8527a268d91f0925 : SUCCESS in 26m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16dae256e96b4352a9e3a370048c0d64 : SUCCESS in 1h 13m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eb3424e95ada41acb2344aca43401ba4 : SUCCESS in 1h 00m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1e72efbaa3814dbfafeae61afceb3544 : SUCCESS in 1h 14m 27s","accounts_in_message":[],"_revision_number":3},{"id":"5393cf33ac4dded0779f605ed28b22ebdd6a631c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 19:57:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f46d90bbe3dc43ef9f0b011391e9747e : SUCCESS in 1h 37m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ebb7ecba06347608f37fcbdb490abb2 : FAILURE in 1h 30m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/07abe14bded74877bd7ba6acf7d59c13 : FAILURE in 36m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3af477a5c70541b19419dc5ce63c6578 : FAILURE in 1h 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6b860edece9e0e6a3c6223793dd9c2a4ef06894","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-25 05:50:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"87ee00fb5c7c6fbefe6eb528e71b52eefa5b5c93","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-26 07:24:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"39198916b978eda45e9b3f8167d16c73f6f6ab00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 07:24:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"93fceb78e0f2f7aff858814440a752bd81132c88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 09:04:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5f3d2443fccc648d6732fba7774525ae6643e459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 09:04:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1c2f618094d44a38941f3de448ddc6fb : SUCCESS in 1h 05m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/553420dc5d2043d1ad447fc95466f009 : SUCCESS in 1h 39m 32s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/599b5660397d4df1a1c76a5531d34a3a : SUCCESS in 1h 17m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45d3accabd8f412cb94804a498e1ef6f : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/819b2decd3304538bbf49ab9ebc61a3d : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f747ca95805640a6aa6173a9cfde28d0 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93da87e6c58347f28d0ac7e0abda272f : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9016bf42ff74fc7a11225617e1af3af : SUCCESS in 7m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4040f7e1223401c9600febe9af9757c : SUCCESS in 1h 02m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cbc66d6566b408ba8338b55a30aace5 : SUCCESS in 1h 14m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ca04253539fc4d6daacf7440ca2f2f80 : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":3},{"id":"1d8725b14be19575fdce2f5e503ff0cadce8309f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 09:06:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912e372dcc03434c95fa616d25951769 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a3ef253dfb8da927bd996e64e19c9cd5ba1aefdc","revisions":{"4fcdc4cc19b60025b21eab11c5e87879b8ed98c0":{"kind":"REWORK","_number":1,"created":"2020-05-24 17:42:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/730522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/730522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/730522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/730522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/730522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/730522/1"}}},"commit":{"parents":[{"commit":"4e41d1416e6a4d4296b4b27f447e58973b34a9c4","subject":"Update OPENSTACK_RELEASE to ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e41d1416e6a4d4296b4b27f447e58973b34a9c4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:42:36.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:42:36.000000000","tz":0},"subject":"Revert \"Update OPENSTACK_RELEASE to ussuri\"","message":"Revert \"Update OPENSTACK_RELEASE to ussuri\"\n\nUssuri release. Standard process per [1].\n\n[1] https://docs.openstack.org/kolla/latest/contributor/release-management.html\n\nThis reverts commit 4e41d1416e6a4d4296b4b27f447e58973b34a9c4.\n\nChange-Id: I45251e80aadaab3f2a540c17e8e60b086f1adfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fcdc4cc19b60025b21eab11c5e87879b8ed98c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fcdc4cc19b60025b21eab11c5e87879b8ed98c0"}]},"branch":"refs/heads/master"},"f9d38bb7e2a8b5d8e9f04c681293dd1a13e2d641":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-05-24 17:45:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/730522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/730522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/730522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/730522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/730522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/730522/2"}}},"commit":{"parents":[{"commit":"be8fff1aee49a4be74192be40f39f7668422b9e6","subject":"Revert \"Switch to Ussuri release of networking-l2gw tarball\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be8fff1aee49a4be74192be40f39f7668422b9e6"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:42:36.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:44:25.000000000","tz":120},"subject":"Revert \"Update OPENSTACK_RELEASE to ussuri\"","message":"Revert \"Update OPENSTACK_RELEASE to ussuri\"\n\nUssuri release. Standard process per [1].\n\n[1] https://docs.openstack.org/kolla/latest/contributor/release-management.html\n\nThis reverts commit 4e41d1416e6a4d4296b4b27f447e58973b34a9c4.\n\nChange-Id: I45251e80aadaab3f2a540c17e8e60b086f1adfa5\n(cherry picked from commit 4fcdc4cc19b60025b21eab11c5e87879b8ed98c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9d38bb7e2a8b5d8e9f04c681293dd1a13e2d641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9d38bb7e2a8b5d8e9f04c681293dd1a13e2d641"}]},"branch":"refs/heads/master"},"a3ef253dfb8da927bd996e64e19c9cd5ba1aefdc":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-05-24 17:57:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/730522/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/730522/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/730522/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/730522/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/730522/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/730522/3"}}},"commit":{"parents":[{"commit":"9836fb9267afac298141e4a3f5a748c457029af7","subject":"Revert \"Switch to Ussuri release of networking-l2gw tarball\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9836fb9267afac298141e4a3f5a748c457029af7"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:42:36.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:56:53.000000000","tz":120},"subject":"Revert \"Update OPENSTACK_RELEASE to ussuri\"","message":"Revert \"Update OPENSTACK_RELEASE to ussuri\"\n\nUssuri release. Standard process per [1].\n\n[1] https://docs.openstack.org/kolla/latest/contributor/release-management.html\n\nThis reverts commit 4e41d1416e6a4d4296b4b27f447e58973b34a9c4.\n\nChange-Id: I45251e80aadaab3f2a540c17e8e60b086f1adfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3ef253dfb8da927bd996e64e19c9cd5ba1aefdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3ef253dfb8da927bd996e64e19c9cd5ba1aefdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
