)]}'
{"id":"openstack%2Fkolla~730525","triplet_id":"openstack%2Fkolla~master~I0700fbc452f395bd21757ec57155308441ce9753","project":"openstack/kolla","branch":"master","topic":"kolla-set-master-to-master-victoria","hashtags":[],"change_id":"I0700fbc452f395bd21757ec57155308441ce9753","subject":"Revert \"Switch to tarballs.opendev.org for source images\"","status":"MERGED","created":"2020-05-24 17:52:24.000000000","updated":"2020-06-16 09:21:09.000000000","submitted":"2020-05-26 09:16:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730525-1590484608254-d3b1a624","meta_rev_id":"5af535a06cea881ddc176eb9df0f766a47f5864f","_number":730525,"virtual_id_number":730525,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 09:16:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-24 20:19:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2020-05-26 05:41:12.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-05-25 06:42:15.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2020-05-26 07:23:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"date":"2020-05-24 17:57:26.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"date":"2020-05-24 17:57:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-24 17:52:24.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-05-24 17:52:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-24 20:19:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-25 06:42:15.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-05-26 05:41:12.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-26 09:16:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b8f7db7e94c2ec63a7276469622a9672e507807","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:52:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05fb8a14e1cb6f72e9f99a0c532a4fb80fa32f70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 17:52:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7fd244ebbc447a31a5dbfd93ffad17dcb4f9cc22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 17:52:36.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"63de499336d00f0b5aadf4a19932ffd7a65e48ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 17:52:40.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"daf4088eb9c647dcf3c4afb23b7e0ea65e0a0bcd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:57:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c23c9796cfae6000085d3b66ffb5a8cf5f513476","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:58:52.000000000","message":"Patch Set 2:\n\nI know this is orthogonal but the order was funky and skipping this revert makes conflicts all over the file. Better clean dropped and redo.","accounts_in_message":[],"_revision_number":2},{"id":"973e1a41e211d3ff259052dfa4c2723b090f01a8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:59:21.000000000","message":"Topic set to kolla-set-master-to-master-victoria","accounts_in_message":[],"_revision_number":2},{"id":"5da39bcb8490391385165c1e4d7b59d5e828bca6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 18:33:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/6c7275efaed74183970c95ebdab1f972 : SUCCESS in 34m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50f16114005b0f298db63b7658fe69937804efe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 19:24:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fe0198ee3f294569ba661500902be065 : SUCCESS in 32m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ab6f64f7322a4fe6ae32e6e25e195cf2 : SUCCESS in 31m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8ed4e5a3db3344d79c67509c01d46267 : SUCCESS in 1h 21m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1cfbc2d2c98f4fd78c4b70b45cd160b8 : SUCCESS in 1h 14m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a3a129ee370949b5b07e061d2adcd36c : FAILURE in 1h 09m 41s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/79053b576d904b70806269a383b0e255 : SUCCESS in 45m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/670a9dc9271a4197a4200e34bf58dc92 : SUCCESS in 21m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d8a545d4ac9b4fdcaa1eae3a023b31c9 : SUCCESS in 18m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4191ca15f4394254af510c0708ad0de1 : SUCCESS in 1h 01m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a00d016f920431e802215043a4239ad : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ca10e488054b2a817c0fed188d9772 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99e33af0d59c4afc949efcc0c199a99e : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e67425792d95474d946de29b2bd86808 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/947301bd0020486aa7f6711d77172889 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f13f460530040a5819ee84b0a844485 : SUCCESS in 3m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c07f282de5c44a77a908a5f28e5fdaf3 : SUCCESS in 20m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/297ad98cf1a342438b16f493d6d34804 : SUCCESS in 22m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c23021d2d6ca4a2d8949a9b46480f806 : SUCCESS in 1h 05m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b0a641defac849dc946efb764a063edb : SUCCESS in 56m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a8c7dbc41ff4440a0836fea5b1c9068 : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":2},{"id":"ef406fc01dec5cdb4cb9420ec00b0ff24e2aa886","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 19:51:57.000000000","message":"Patch Set 2:\n\nrecheck\n\nrandom issue\n\nwe might want to squash all reverts to avoid a cascade of problems","accounts_in_message":[],"_revision_number":2},{"id":"ed56ed2c9604a86ec21e27b075a52ee0e0f2e97f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 20:19:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/fdea3be9716343669996f1c823f9eda2 : SUCCESS in 26m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"534029fcfe3ed8b51a5e32f134c7f666440a3894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 21:19:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/933802cb045d4640ba3b6884626ed7f6 : SUCCESS in 26m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/206a1884a089449e8afa97ed87035a56 : SUCCESS in 37m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0a8be311d3e942cbbb1acc5a30ed5cde : SUCCESS in 1h 21m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ab63054bcb4d4368bc6116d6ce088faa : SUCCESS in 1h 18m 40s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4f1da589a5b840389c69778f5b1026fb : SUCCESS in 1h 26m 09s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/a76905109c13407ba51aa97b7f97ddc2 : SUCCESS in 45m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/77b89a23dbc94fbf9cd390b768848c91 : SUCCESS in 29m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2f0093d539a14cf193a644c7fab91d24 : SUCCESS in 27m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0806390cd4ce439abdb6bc7fdaffe0e8 : SUCCESS in 1h 03m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/297e49ccbd5748f6867891c2d0b93e26 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2d9b0aff3354f34ac358ede6587673e : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e23d36a78a14be08a2fe4736caca229 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d033840c0a7405f88d09363eaf28abe : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e87ac4bf75ca429a88a283a544821617 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ba94a1aadec4b3abd380a65eef60a4a : SUCCESS in 3m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ff4997fbb1d49aaba3baee089e3af2d : SUCCESS in 36m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4006559bffb54ab0ba14138238e0dcbd : SUCCESS in 28m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dcbc962dc9014cdbb7a4465076cfd3fc : SUCCESS in 59m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f300f8099564915a8b15f689fcfc2e6 : SUCCESS in 49m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a8f6b9fe34fe4e3e82d9c26899ff0806 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f0b99edd370fbaaad91021694c9bb40ed6fc06a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 21:36:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/964b10eb7dc84963bb48279fd82e1ac4 : SUCCESS in 1h 34m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fadb455c9bad473697a3fcd69c7d77ce : SUCCESS in 1h 23m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/23f2f64c8d884d2c9c786b944d2908d7 : SUCCESS in 2h 05m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9bff7cc187dc4ebbbee6e32a4fd8c92b : FAILURE in 1h 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3514ef3a2ec7c6344ac3f6de0d2278adaeae02d0","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-25 06:42:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9a96925163ba81d11573d1a849b693fddb65582c","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-26 05:41:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"15f2f95351f7c3136b0f76212c348fe5bd8d4bb6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-26 07:23:46.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4607fa3c2c67bae681387af94b4fe9303deaa166","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 07:24:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d7846c5024b454abf09a014cb3a9dcf4954cfb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 09:16:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/46e30e03c8694ad4bb8f2ba18e9c5e72 : SUCCESS in 48m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/dd9905917e444b268f33f9fc22f9dfca : SUCCESS in 1h 15m 06s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b54e1977f9884650ae72b95f53b79691 : SUCCESS in 1h 14m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3c17d83d9ac4bf2ab96e187c5572c5b : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01e0530b04c4933bf676ac870509755 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60d94c71600a4886af1fdb53d9bb6434 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c47a1d27a164aac92435942a56ae9ef : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ebe10e35ca4fd3b015927c40f1f81d : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ef1215aa19740b39479ed4dbbf2af3a : SUCCESS in 4m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c5adf6ac8d84cbeb415e40173209939 : SUCCESS in 26m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e90d31266d50493296a8ebcd80c15b34 : SUCCESS in 1h 29m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/791054700dde48bfbb2d9f86efc65fb0 : SUCCESS in 1h 14m 32s","accounts_in_message":[],"_revision_number":2},{"id":"1f62d246b3985a759ef229a882054f88b3b1b5fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 09:16:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"519d510fdb7d27761cef1b6c06539daef272285d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-27 11:24:33.000000000","message":"Patch Set 2:\n\nHas it been redone?","accounts_in_message":[],"_revision_number":2},{"id":"d4781787a469b511af91a79a19a289e55845d4e1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-27 12:54:51.000000000","message":"Patch Set 2:\n\n\u003e Has it been redone?\n\nNope, pending. But I did not start it.","accounts_in_message":[],"_revision_number":2},{"id":"5af535a06cea881ddc176eb9df0f766a47f5864f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 09:21:09.000000000","message":"Created a revert of this change as If9bd73a2e758c9da87f9a9a60fe075d807e3ca00","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c2ee4993d824ed4dc5331706a86491a459c3c4e","revisions":{"397350752df0c7af268c19971161de879dac891e":{"kind":"REWORK","_number":1,"created":"2020-05-24 17:52:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/730525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/730525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/730525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/730525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/730525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/730525/1"}}},"commit":{"parents":[{"commit":"ededc45cf8a7fe9398032d305218d31660bcce79","subject":"Switch to tarballs.opendev.org for source images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ededc45cf8a7fe9398032d305218d31660bcce79"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:52:24.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:52:24.000000000","tz":0},"subject":"Revert \"Switch to tarballs.opendev.org for source images\"","message":"Revert \"Switch to tarballs.opendev.org for source images\"\n\nUssuri release. Standard process per [1].\n\n[1] https://docs.openstack.org/kolla/latest/contributor/release-management.html\n\nThis reverts commit ededc45cf8a7fe9398032d305218d31660bcce79.\n\nChange-Id: I0700fbc452f395bd21757ec57155308441ce9753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/397350752df0c7af268c19971161de879dac891e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/397350752df0c7af268c19971161de879dac891e"}]},"branch":"refs/heads/master"},"1c2ee4993d824ed4dc5331706a86491a459c3c4e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-24 17:57:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/730525/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/730525/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/730525/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/730525/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/730525/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/730525/2"}}},"commit":{"parents":[{"commit":"1697bf198b14b8e77bb3f347acf200327143d1a9","subject":"Revert \"Switch to Ussuri release of ovn-octavia-provider source tarball\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1697bf198b14b8e77bb3f347acf200327143d1a9"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:52:24.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:57:04.000000000","tz":120},"subject":"Revert \"Switch to tarballs.opendev.org for source images\"","message":"Revert \"Switch to tarballs.opendev.org for source images\"\n\nUssuri release. Standard process per [1].\n\n[1] https://docs.openstack.org/kolla/latest/contributor/release-management.html\n\nThis reverts commit ededc45cf8a7fe9398032d305218d31660bcce79.\n\nChange-Id: I0700fbc452f395bd21757ec57155308441ce9753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c2ee4993d824ed4dc5331706a86491a459c3c4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c2ee4993d824ed4dc5331706a86491a459c3c4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
