)]}'
{"id":"openstack%2Fkolla~730526","triplet_id":"openstack%2Fkolla~master~Ia2f036922406cec40e94ed8093ae07239bba4b2e","project":"openstack/kolla","branch":"master","topic":"kolla-set-master-to-master-victoria","hashtags":[],"change_id":"Ia2f036922406cec40e94ed8093ae07239bba4b2e","subject":"Revert \"Switch to versioned tarballs part 2\"","status":"MERGED","created":"2020-05-24 17:52:48.000000000","updated":"2020-05-26 09:16:50.000000000","submitted":"2020-05-26 09:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730526-1590484610300-fdce670a","meta_rev_id":"0287403b6f96a2549485a5002c541f2b74248bcf","_number":730526,"virtual_id_number":730526,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 09:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-24 18:32:57.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-05-24 20:54:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-05-25 05:55:39.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-05-25 05:55:39.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-05-24 17:57:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-24 17:52:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-24 18:32:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-24 20:54:55.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-05-25 05:55:39.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-05-26 09:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"453fd3cfc691d23e3ea5b42081929fdc00fb8869","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:52:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ee2fdd0b75795b4a46b15355d3703c791259c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 17:53:00.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"4c55aedcee249da01f809b33b978d8826997d740","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 17:53:01.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"3ace39c5d97051d4742b8fb2970cace2477e5bd2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 17:53:05.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"a976cf2a1fdd9765d4864923c2831222d8248484","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:57:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"563ed490cf2590c42bcaf530f5c332fc64f4e092","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 17:59:28.000000000","message":"Topic set to kolla-set-master-to-master-victoria","accounts_in_message":[],"_revision_number":2},{"id":"5c865756d5f82cf4c0eb4a49ed12e7a7a51afb04","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 18:32:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/529782997aea4ee0b3b792edd4c32d94 : SUCCESS in 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74b8426d4dcce4e0eae1736fab79818980e5a32f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 19:20:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/768be23d5a594fc6942977ed3c855b04 : SUCCESS in 32m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/10d61074d2e9490c8bccd47918ec4f3b : SUCCESS in 34m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/43b284fcc7c549d7a009f96325489cbd : SUCCESS in 1h 19m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ed0512e816864c4a969bee3ebe4fc8fc : SUCCESS in 1h 21m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d431460195e749a084e3d7aed0df2bf7 : SUCCESS in 1h 15m 27s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f2359793a29f4fdb8115497cac64ff2d : SUCCESS in 53m 05s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d935a73d5e354c8d88b27b9c2b091315 : SUCCESS in 26m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8ee1570aa85c44fa9c2fae0a95e685d2 : SUCCESS in 16m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/81ecb1e01d7e40f5a2a8d5825051e510 : SUCCESS in 1h 03m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbe1246ada324d1fb59e6e6cd4291499 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130142600bc947dfb8f502b002a4d14f : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/534983c4e27a4c7086885606659ad99a : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f47eba1a3e74e9891ff5e88da32c7c6 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2530758c1964ce2a7f73728e079ca6e : SUCCESS in 6m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a973837659b64957b85b603ae2b0313f : SUCCESS in 22m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be1453b78e1a4609b687a718741c8697 : SUCCESS in 50m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f666d3caaf1b4ecd93a1b8b781cb2a59 : SUCCESS in 51m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b208e07de2ba47d78526f7326fe97c38 : SUCCESS in 1h 02m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0fc3f29f63bc4ef296a9564ae8333bf4 : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":2},{"id":"c5a3968aea1b7b76c8b61baf8288ddc2980a8627","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-24 20:54:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3c50e651c36ca525184f2909aa35b446e7050cc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 21:47:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/758d7cfd1e514543b35d456e29dec024 : SUCCESS in 59m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f33990147dc464cb3da7175730e78a7 : FAILURE in 5m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c14cd4668bdc42ccbd6da15ddd65a7c6 : SUCCESS in 1h 52m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4036b18e7c3c4afa844208ee39fbbc05 : FAILURE in 1h 01m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22206fc28fa6d0e01544988c39dc51cb03909339","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-25 05:55:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f26d33436bfa95c6b4b8eff9078a0d14ac56446","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 07:24:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"98ede1f84c6b3bcd126921f8fee94eac23936e78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 09:16:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d7582aca0c414c598d53e6308ac1cbef : SUCCESS in 25m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ad03e5cf53c24ea6b5b0cabbad881f20 : SUCCESS in 1h 38m 54s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b35a744e274f44beafb919aa05273387 : SUCCESS in 1h 22m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6178462a5af4bde9758259768cf8dd0 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d68f935d6d4689b005d0831c72c460 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2fb0cb6ccf547cd984fa44391ad5b19 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f830abd49ae48d7afdf9f937550b783 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc88d54ccdf47a789b73cd7fb4fed78 : SUCCESS in 7m 24s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a12d7ea716904662a3a15aa951e2a1c3 : SUCCESS in 29m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23abf5f1f9d34426a25d42b3fcec52ed : SUCCESS in 1h 13m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5e3a69b72484c82a25bedcf0f9780b1 : SUCCESS in 1h 16m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0287403b6f96a2549485a5002c541f2b74248bcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 09:16:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a50bddee03cedad0ade823b2365b5e56f5610bc3","revisions":{"3f96ebf0988a65ec1fbef55014090bcaa79c1eeb":{"kind":"REWORK","_number":1,"created":"2020-05-24 17:52:48.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/26/730526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/730526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/730526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/730526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/730526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/730526/1"}}},"commit":{"parents":[{"commit":"d99e859c85be0654bceea594f6838b5d79d46024","subject":"Switch to versioned tarballs part 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d99e859c85be0654bceea594f6838b5d79d46024"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:52:48.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:52:48.000000000","tz":0},"subject":"Revert \"Switch to versioned tarballs part 2\"","message":"Revert \"Switch to versioned tarballs part 2\"\n\nUssuri release. Standard process per [1].\n\n[1] https://docs.openstack.org/kolla/latest/contributor/release-management.html\n\nThis reverts commit d99e859c85be0654bceea594f6838b5d79d46024.\n\nChange-Id: Ia2f036922406cec40e94ed8093ae07239bba4b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f96ebf0988a65ec1fbef55014090bcaa79c1eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f96ebf0988a65ec1fbef55014090bcaa79c1eeb"}]},"branch":"refs/heads/master"},"a50bddee03cedad0ade823b2365b5e56f5610bc3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-24 17:57:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/26/730526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/730526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/730526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/730526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/730526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/730526/2"}}},"commit":{"parents":[{"commit":"1c2ee4993d824ed4dc5331706a86491a459c3c4e","subject":"Revert \"Switch to tarballs.opendev.org for source images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c2ee4993d824ed4dc5331706a86491a459c3c4e"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:52:48.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-24 17:57:08.000000000","tz":120},"subject":"Revert \"Switch to versioned tarballs part 2\"","message":"Revert \"Switch to versioned tarballs part 2\"\n\nUssuri release. Standard process per [1].\n\n[1] https://docs.openstack.org/kolla/latest/contributor/release-management.html\n\nThis reverts commit d99e859c85be0654bceea594f6838b5d79d46024.\n\nChange-Id: Ia2f036922406cec40e94ed8093ae07239bba4b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a50bddee03cedad0ade823b2365b5e56f5610bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a50bddee03cedad0ade823b2365b5e56f5610bc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
