)]}'
{"id":"openstack%2Fkolla~731798","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ifa0614512a314a114f06ac2a39f646fddf839e80","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"Ifa0614512a314a114f06ac2a39f646fddf839e80","subject":"Switch to RDO Ussuri release on CentOS","status":"MERGED","created":"2020-05-29 16:18:11.000000000","updated":"2020-06-05 09:28:52.000000000","submitted":"2020-06-04 19:57:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731798-1591300640407-5027401d","meta_rev_id":"34ee0a937fbaaa8e8438b971d5a59d0ec3b0a20e","_number":731798,"virtual_id_number":731798,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 19:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-06-04 15:35:20.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-04 14:14:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-06-04 15:35:20.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-04 14:14:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 14:14:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-04 15:35:20.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-04 19:57:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7aeed4673c6581b4d6ace8eebe2c79c1e8577cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-29 16:18:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a0364255e3902de8813456a823c9aedb76896f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 18:40:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c21d714098f4aed9bca3e202c6a9e5b : FAILURE in 7m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/89fb41fcff0843f599de6ad5ff67068e : SUCCESS in 1h 15m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b766e53367bf4c62b80236f92fa8c8b9 : SUCCESS in 2h 16m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4f92f3a3da745d4b6171683a0a2bd8a : FAILURE in 1h 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f072387540a3e27ac5e1e5ed329e52b8b766688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 20:45:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3864a06308564b368cfd6c476a80319c : SUCCESS in 24m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0f981b57ad6942dba5157f8c4828e60d : SUCCESS in 30m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a32bbc8603bf4d1abe7134afd2c1203d : SUCCESS in 1h 25m 19s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2d8a6cb7374640a080d08e624a91b8f8 : SUCCESS in 1h 15m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/de5b62ab3b504a95b5a7663cd1662b85 : SUCCESS in 1h 27m 52s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d36429d253e54dcda2ac1c3c13d3e026 : SUCCESS in 46m 11s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/f4f5ddce4e794264997491ede90d44ae : SUCCESS in 29m 25s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/acac8e18914c469faf887be54d7ce300 : SUCCESS in 1h 50m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d3b9bcf9089e4234ae8c70bb073693d7 : SUCCESS in 2h 12m 21s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/71cc8529478b4259bdd677ed0a3f04c7 : SUCCESS in 1h 53m 56s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/02e11851091b41839cb39118881d55d2 : SUCCESS in 1h 50m 20s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/4ca31b4b68c942c4a2d26fc896245726 : SUCCESS in 1h 15m 18s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/b3a0c5fa5420464f819e3c11d9653fd7 : SUCCESS in 1h 49m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/71e9104485114310b9de76fa53657eb1 : SUCCESS in 23m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8ed16acf43024bf8a6bc92ce7dbe7b5d : SUCCESS in 57m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f744ed1e997446e4aaf38f5b6a738a6c : SUCCESS in 1h 38m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/125e58005fca47a5b9ca21d6ee21ee94 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0915ce40074c42a0c4f1b3499e6024 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1e7766ea1e146fd92c892ee4a7a1b6d : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69d5a39853b44242bb4b9c18c0a1dab6 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271f8fd122f64908be814992230b2264 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/402f43df7b8e48728810878b2ab056f3 : SUCCESS in 29m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc6ef33b7bb44d87bfea3a1c40180c41 : SUCCESS in 39m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0025b1ad3a044314911c855cf0ecc30e : SUCCESS in 1h 04m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/58150b6637fd41049e15be344b293224 : SUCCESS in 1h 00m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9e46ae5a3d0e4dbf82815e5b926f0bf3 : SUCCESS in 1h 07m 35s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/20445684f89b4a9aa7c8fce896403dde : SUCCESS in 1h 40m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bb14ee58846649f682da0ccdcabb4034 : SUCCESS in 1h 54m 43s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8cd5c4a30f474da6a4fcc49a68efdfeb : SUCCESS in 1h 34m 28s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/dff5556baa034ad28451dfdb3c09795a : SUCCESS in 55m 40s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/04b38ccdbfa84c808d102580e0fb5789 : SUCCESS in 1h 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"893980ec63527661d6340ffe297ac541d89c1257","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-30 14:40:58.000000000","message":"Patch Set 1:\n\nhmm, the patch is OK in general - but we seem to be getting ceph nautilus as extra in here - relevant comments are invalidated as well as the need for ceph.repo","accounts_in_message":[],"_revision_number":1},{"id":"e9af80dc7ee30c0bda3239b01d252db92bf1a0f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-01 09:38:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b35fae6efddd6b0593d2ece1056ae5cb49a32b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 11:28:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/dea515de227f41aca3c98c6ef395fac2 : FAILURE in 7m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d336fcda087448c58ce2ed7c17d747de : SUCCESS in 52m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/24e623b0f27c40e4bba4218f4830181d : SUCCESS in 1h 49m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/63d3b21ecfec43f0b7e7abe47992292c : FAILURE in 58m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c8b388ca56657ec827e592a4344257bb8aaaf0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 11:56:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5f21a88847144013aa05212425233e61 : FAILURE in 6m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f7a42c7273964513b7991660f48dde68 : FAILURE in 6m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9398602e679142e2b2368237f84afd24 : FAILURE in 12m 44s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/81ffff9e0b35468392468d6fb987dd04 : FAILURE in 12m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/433c089b2b574713915450a9d4728bee : FAILURE in 13m 14s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c52ed0adfe084382986b23cf98bc4106 : FAILURE in 20m 19s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/ec908eff4f144b019cc384b3ab66d5ca : FAILURE in 11m 31s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1e5e5031d6bf43a488de90ca5b6354ea : FAILURE in 13m 32s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f0385bb6e845482591728d4a81747417 : FAILURE in 15m 54s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/e310593527444f1c9ede3754debdd61e : FAILURE in 14m 19s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/34fc783e34584895876297b9c96b5bd7 : FAILURE in 13m 26s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/ff5a98c3d7fa4b158768ec1b137b5cc4 : FAILURE in 12m 05s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/f8cde36fe4cd4ec492a7f550e2b415c8 : FAILURE in 18m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/04121dc16f55440790b22c5a7e5a1355 : SUCCESS in 28m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4694c06ff9cd4f6aae9de7be39a4ba22 : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7ea8b345758c4270a25578f7d8869266 : SUCCESS in 1h 56m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5998b269491b4c04a62ea647ad25a059 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0aac5323944666ba51b9c18f573142 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecc68cd01f33414c9cc142f0eb3db417 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9877a3af13c94064b2d35255a161030a : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/706baefb4c8542c38ed3d402a22e5c65 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c97e2fc433d34548835fb442f7a77f3b : SUCCESS in 1h 52m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed7589560ce34bcfad1b1a0b029b776d : SUCCESS in 1h 50m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc05699f15454fbdbc97913699a1e63b : SUCCESS in 1h 57m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b6177b8c7d39441188079d9f3c4961ae : SUCCESS in 1h 00m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9eedf06bf22e4987bac1a2d059052d03 : SUCCESS in 1h 07m 29s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/520dfdfd21f04eb99be8ffc723e56557 : SUCCESS in 1h 11m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f205ece120f8491384ef3900637f09cc : SUCCESS in 2h 16m 19s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/700d2a8653814fa9b39679b24428c553 : SUCCESS in 1h 11m 49s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/15535b765dc14f369664ed8bf9b6b417 : SUCCESS in 56m 02s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/128364cc8f35449db1f0953c4f0e931c : SUCCESS in 1h 27m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1f5e9f768ff9b30c288f6317d551750c0761537","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-01 12:51:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f19d7f5689a365f92508b3da50f173af5d40cb73","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-01 12:53:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"df15cf531949f186e1e1e58de567a22b17a64d99","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-01 12:55:21.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2c8acc30bb49daf905aaeb4dea2cafe0c2a4b3ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-01 12:56:01.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"02c22be5cde5e8468c10141dcbffe16f2ec8c64a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-01 12:56:43.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"453c91444989057513e4044350b137c9748625bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:56:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/66e40a27f75343f1b59b1028a83b3777 : FAILURE in 9m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f071dc1a416745358a76cb0eefee7286 : SUCCESS in 1h 01m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cf50593f3804c0186406e8dd54a745f : SUCCESS in 1h 57m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4710bdfe8504759a424dcc052be9319 : FAILURE in 1h 03m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f8fd7f6d8546a5c4dc398818aa5d7fd7d3e824dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:57:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bc3d55ad9fa24b3398a90317a4e43e12 : FAILURE in 9m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9c8f70e68e654623bcc133e7a4d61721 : FAILURE in 8m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/adfcdc134ed2454982c0c4a7851fb8be : FAILURE in 14m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/45e2a362d72d47d6bf59d704fcd7cff8 : FAILURE in 15m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f9746aeacb674dcab4cf1c348256b6d5 : FAILURE in 15m 40s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/97d754cbf8444fc4ace86ef685a322ce : FAILURE in 17m 45s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/f4fdd53f431a41c0a463b21822adee08 : FAILURE in 12m 13s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f9f462060a2848e1bf543507d9e62451 : FAILURE in 17m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bcf77ca4b2ef4e58a63256ae9b35f715 : FAILURE in 14m 05s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a7698e5d940443b0b219246413fc1eb7 : FAILURE in 12m 30s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/b31af5f6cd2144b58faebd89cb60651f : FAILURE in 12m 22s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/02304b8aeb87445ca714f0683071176d : FAILURE in 11m 50s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/9cf1b221b7c3404e8c907f5995cfdd52 : FAILURE in 13m 37s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e93b5d3a5d6f4e049622fc7760624591 : SUCCESS in 24m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ac2998940e13432b82643ed5c618fa33 : SUCCESS in 27m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0a50bae1e1ca4604bc71294d7b63aeec : SUCCESS in 1h 05m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe3b1ee7254340b487c2f0f956486f71 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1322fbefdcc142d9868d70e41cfd2688 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15d7ac7317a3472189c23ff23c85bf9a : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60cb11154b4540e1b886d9d069f71250 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7881f488a3df40b0815ae40a1c09e55c : SUCCESS in 5m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c8e8277605742f9a2eb483cb1230bbe : SUCCESS in 22m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a98a54ca7eb4c819bcfa71e53447ae3 : SUCCESS in 25m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e59fe291badc4605a52fe3877e6e1613 : SUCCESS in 1h 05m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/843d423651624d87afd94a3f92573566 : SUCCESS in 1h 01m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/580e46691fdc46aebf6c771df9965e29 : SUCCESS in 1h 12m 15s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/a1baba04eb0a42639211dffeb1e12a16 : SUCCESS in 1h 37m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7880f93a840244ce8105a22f24d2b6ec : SUCCESS in 1h 56m 03s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/5a7494903b574c4e9296524d355428a9 : SUCCESS in 1h 53m 14s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c3208f9ffc8044969a335bcc19461320 : SUCCESS in 1h 05m 46s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/ff9817f3a0984a13895d670d9d845376 : SUCCESS in 1h 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"273ed12b912ad0df99bd4f05b08be911b671a346","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 08:45:39.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"76fbd101bd1c6855ed94d4c155cad4c99c2e72cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:53:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/099cf5e6878943668064fe815fbffa7b : FAILURE in 6m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4370f29fe13c489cae9bbbd8d3ea7e97 : FAILURE in 6m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/db3d58cc912f4781a33091f1887d59e9 : FAILURE in 11m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6e126ab50cc64dec9d88682dd10e8551 : FAILURE in 14m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/940ff050de214cc9a2c5fbc39fd077d5 : FAILURE in 13m 09s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/594178ab58c64cff86c35b7e7d881fc2 : FAILURE in 18m 47s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/60dff851353b4f1a81ec99893c790ce4 : FAILURE in 15m 27s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e7ccca9d415d40558f1282a67c8e8680 : FAILURE in 14m 20s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/213075a7009c4d40b1623040ad3173a6 : FAILURE in 15m 19s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/461a484223e249da8d12a1edc663a29e : FAILURE in 18m 50s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8cd5ffc6163e41f0a67f480c3a013e23 : FAILURE in 20m 52s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/aa6730e747f24987b1c61f9072f61126 : FAILURE in 16m 12s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/1f645da813f0467db15bf838dad44fd6 : FAILURE in 13m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bc8aa32376fd484fb730198d7c4e6aea : SUCCESS in 31m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b99a1d2a15f64b0b856f8a3e8f530f8b : SUCCESS in 30m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e0bbcc9161d64064a930d205b10e9ea2 : SUCCESS in 1h 10m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b31b2638a6d74c61bca07bbb06607dfb : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c17a06c8047b4e9481eca2538c4aeb52 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a30cfa2d34ff453db9cd90dc9592fdf0 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd12544727d5497c87d1f380951f72da : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b8ad8daada4370b4408af476e9f35f : SUCCESS in 9m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0982610b16204b319ff7cbca29e8a507 : SUCCESS in 25m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f97134358589492e93d254f451ef0cce : SUCCESS in 24m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2860bfdd607b4261bd22e87e347d0753 : SUCCESS in 1h 01m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5539ff31cc1b413aa030dd74d42f0f40 : SUCCESS in 58m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/584523b369954928b92c4e8cd8dcaff1 : SUCCESS in 1h 20m 31s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/03fcfe6dffba4ef0aac740f75d8ed86d : SUCCESS in 1h 44m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d6cc224550dd4142aba9a4f77875ccd0 : SUCCESS in 2h 07m 26s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a2df179c8a2b4a7992c51566bd6a78bb : FAILURE in 1h 02m 56s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a5598d133cc44bd480297f9df25ab394 : SUCCESS in 1h 05m 06s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/be5a3cb34480495a974f7b5fd228c98a : SUCCESS in 1h 26m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d7e9c607acc7c59f20444bdffc31c3fa29216789","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 11:12:30.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2cac5a13ba3435b229d948c199b346614d0d67c7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 12:39:37.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"87f3abccfc0024b3e9462e018f7983e702199da6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-04 12:44:04.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2524a5274a24db77d02c5139ac91b59b002ad457","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 12:44:25.000000000","message":"Patch Set 9:\n\nfsck, centos-release-ussuri brings in also ceph-nautilus which is a clone of centos-ceph-nautilus repo ararar","accounts_in_message":[],"_revision_number":9},{"id":"67862c8fe67155657344ae3187095f0c706de1c1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 12:44:32.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"fe0a818e0329f5433eb126d619fd4626df384a9f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 13:22:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d2334157abb6e56ff56550048b50dbdd66313aac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 13:23:58.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6105f2e2080f070e0d0eec568bd679adf030d283","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 13:26:46.000000000","message":"Patch Set 9:\n\nI\u0027ll redo this part","accounts_in_message":[],"_revision_number":9},{"id":"9963ff9b15d6ca07378f661b314638c1d3daca5e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 13:27:11.000000000","message":"Patch Set 9:\n\nYatin helped me handle the repo case https://review.rdoproject.org/r/#/c/27933/","accounts_in_message":[],"_revision_number":9},{"id":"954b638d79a770ac033a202ce2769c2a66fc70ff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 14:12:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d20c928ab9156805a3686d9e99705c8cd62786eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 14:14:02.000000000","message":"Patch Set 10: Code-Review+2 Review-Priority+1\n\nok, I restored the base structure which looks better","accounts_in_message":[],"_revision_number":10},{"id":"04ba1eb7bf432830c03542a1b672a5af17f740d9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-04 15:35:20.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a25a4bb8422653a0fa56cae2e3c3df51a2fa600b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:23:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/88bd9f5a3ce44e6f9c6868abfd207362 : SUCCESS in 2h 24m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/865d964e5c374f76a5a19ba744ed612c : SUCCESS in 2h 11m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03f85560a7004be8aa4ff1900dba9501 : TIMED_OUT in 3h 05m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cfa312d27e649268db8c1858b18c089 : SUCCESS in 2h 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9e9fbec101aabe08d819344f598323c760c23594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 18:22:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b3e572999b314db49a17b953b70ef48d : SUCCESS in 21m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d72fbf2cdefa4ec598916be515f14aba : SUCCESS in 27m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4a890439836b48ee9ea061fa92a25cf1 : SUCCESS in 1h 22m 12s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fed3d4d450ba4e8cb5db007302995fb1 : SUCCESS in 1h 07m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5a52a1f86df543c78da8d6f670b5f11b : SUCCESS in 1h 31m 56s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/db7bcc18b0f44086aca6fa98caf1fe58 : FAILURE in 18m 15s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/0943f83f1e974774898091d25dd301c9 : SUCCESS in 32m 31s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5ecae97d4da149db955b51e4503a4370 : SUCCESS in 2h 08m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/762d67321f8842d0b8f4c514bf85d71d : TIMED_OUT in 2h 38m 40s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/b60b5be1fd814ddfbae9611002938e8a : SUCCESS in 2h 00m 45s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d43a1ad0caef4f1ca13b1a8d26ab4ba0 : SUCCESS in 1h 57m 28s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/ef07f7b11a6142c38121bf9b5fc2edec : SUCCESS in 1h 19m 43s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/a81585f3e46145ac9cc605817da4a20e : FAILURE in 1h 16m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3ecacf0cfef84f958e2cfed83c4b2850 : SUCCESS in 21m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ab64c826e1914f15b738016f5320a33b : SUCCESS in 18m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/560e88d3a6a14c4a9ec8b8c44a6e7750 : SUCCESS in 1h 16m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39d76c9c8b834b1c843eb64d8f5fbf1e : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fa71cb01d1d47c2b364e1e566ded382 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d2facf4b02044399e814d7a968e3203 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2188324373744922959f9ce8b0aee28f : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d663a47ac0245d19284b99a464dd02a : SUCCESS in 8m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f075d237faf41a6976a3237c22ece6f : SUCCESS in 29m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7eb012b7e8814dbcb49557e5fdb11376 : SUCCESS in 25m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef3192c755c74cf8be74ba49422c5df1 : SUCCESS in 1h 10m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3976d564cdf64883bfb3c81990ebd9ef : SUCCESS in 57m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/15f4d7f2891a4aa0937227008b6871d2 : SUCCESS in 1h 22m 25s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5608252b24ce4dfb94d86eafa5333cb8 : SUCCESS in 1h 31m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/4cb70b8b967c4d03988cb4e110b6d2ee : SUCCESS in 1h 50m 48s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/5425716ed4b04ccf9c538e14ab19900f : SUCCESS in 1h 57m 42s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b397b32fe8a84c47937eab8715d5b923 : SUCCESS in 1h 12m 40s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/e1a668c838414fb2bbcb0891161cd708 : SUCCESS in 1h 38m 33s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d0173d1ce4997427b248ee7a199b94b3fa698a25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 18:22:14.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"56140897a0c8b8037eecc0c7f35c25f397c81008","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 19:57:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"cae3b359b70035fdf9bf74ae3fe5e6d0a0bd884b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 19:57:20.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ede95e38bee847bc86e9f9fcd1b800ed : SUCCESS in 36m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9cb7143274ab4b8d9ac2502caf92ae81 : SUCCESS in 1h 31m 00s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/34183ce1313e43dfb6ee7824516075e1 : SUCCESS in 1h 32m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f9be890401d4f3c878351edf1bd6ad0 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c3436a41a8e43e8847ef3c34fcea56a : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f02b9f27d33a436d9cd4df15a100926e : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/541cdf347bd3487594e0b9992ea16e31 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efadd89c4df6400082c8ff8b59bde032 : SUCCESS in 7m 19s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1692c5419aec4e718e6e925289a0b887 : SUCCESS in 37m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c193fc246a2741e893e0d9412aa801cb : SUCCESS in 1h 04m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/abec9003aed148568d861fed057fc1fd : SUCCESS in 1h 10m 42s","accounts_in_message":[],"_revision_number":10},{"id":"737c6475d4ae7db354104cb149186a03fb23f885","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 19:59:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5733bd645556405fa7062d2bbd791f9e : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":10},{"id":"798fad2fd4f3601334ac6473ca9b52b37d0d6758","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-05 07:49:00.000000000","message":"Patch Set 10:\n\nGotta be careful applying +A without seeing results. Tripleo is now broken.\n\nhttps://936dffb68330292e728e-2de10bdd194d323966e80d1fe3d10503.ssl.cf2.rackcdn.com/731798/10/check/tripleo-build-containers-centos-8/db7bcc1/logs/buildah-builds/kolla-builds/07bf003d-a3c9-47bb-94bc-287a47b1b42a/docker/base/base-build.log","accounts_in_message":[],"_revision_number":10},{"id":"34ee0a937fbaaa8e8438b971d5a59d0ec3b0a20e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-05 09:28:52.000000000","message":"Patch Set 10:\n\n\u003e Gotta be careful applying +A without seeing results. Tripleo is now\n \u003e broken.\n \u003e \n \u003e https://936dffb68330292e728e-2de10bdd194d323966e80d1fe3d10503.ssl.cf2.rackcdn.com/731798/10/check/tripleo-build-containers-centos-8/db7bcc1/logs/buildah-builds/kolla-builds/07bf003d-a3c9-47bb-94bc-287a47b1b42a/docker/base/base-build.log\n\nhttps://review.opendev.org/#/c/733790/ should fix it.","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"4d0972d4971a5d590c03598d401c30bb97576273","revisions":{"743538824d032e16ccf2375552fea1458a5df489":{"kind":"REWORK","_number":1,"created":"2020-05-29 16:18:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/731798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/1"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:22.000000000","tz":60},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/743538824d032e16ccf2375552fea1458a5df489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/743538824d032e16ccf2375552fea1458a5df489"}]},"branch":"refs/heads/stable/ussuri"},"8d746ed0c402316f265f26da7904aa51d30ac9ad":{"kind":"REWORK","_number":2,"created":"2020-06-01 09:38:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/731798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/2"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-01 09:38:10.000000000","tz":60},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d746ed0c402316f265f26da7904aa51d30ac9ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d746ed0c402316f265f26da7904aa51d30ac9ad"}]},"branch":"refs/heads/stable/ussuri"},"017507f51446c48a0845541a864132dc1250ed76":{"kind":"REWORK","_number":3,"created":"2020-06-01 12:53:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/98/731798/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/3"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-01 12:53:43.000000000","tz":0},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/017507f51446c48a0845541a864132dc1250ed76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/017507f51446c48a0845541a864132dc1250ed76"}]},"branch":"refs/heads/stable/ussuri"},"ffca03d18ce8fe5fb1de9e3c145a213a962cf1f5":{"kind":"REWORK","_number":4,"created":"2020-06-01 12:55:21.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/98/731798/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/4"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-01 12:55:18.000000000","tz":0},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffca03d18ce8fe5fb1de9e3c145a213a962cf1f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffca03d18ce8fe5fb1de9e3c145a213a962cf1f5"}]},"branch":"refs/heads/stable/ussuri"},"7df53bdbb80c4b7b29f71d97f358b90cdb519884":{"kind":"REWORK","_number":5,"created":"2020-06-01 12:56:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/731798/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/5"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-01 12:55:58.000000000","tz":0},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7df53bdbb80c4b7b29f71d97f358b90cdb519884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7df53bdbb80c4b7b29f71d97f358b90cdb519884"}]},"branch":"refs/heads/stable/ussuri"},"4ab99dc139ac86c0a48d55c35eecf9aea12d3eb9":{"kind":"REWORK","_number":6,"created":"2020-06-01 12:56:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/98/731798/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/6"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-01 12:56:36.000000000","tz":0},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ab99dc139ac86c0a48d55c35eecf9aea12d3eb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ab99dc139ac86c0a48d55c35eecf9aea12d3eb9"}]},"branch":"refs/heads/stable/ussuri"},"825a3670ab48ceb8ce1ecb9cb5a062ff4e1f3904":{"kind":"REWORK","_number":7,"created":"2020-06-04 08:45:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/731798/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/7"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 08:44:53.000000000","tz":0},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/825a3670ab48ceb8ce1ecb9cb5a062ff4e1f3904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/825a3670ab48ceb8ce1ecb9cb5a062ff4e1f3904"}]},"branch":"refs/heads/stable/ussuri"},"4faa51ce923702b94e096310ce37992d9f1d461b":{"kind":"REWORK","_number":8,"created":"2020-06-04 11:12:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/731798/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/8"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 11:12:27.000000000","tz":0},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4faa51ce923702b94e096310ce37992d9f1d461b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4faa51ce923702b94e096310ce37992d9f1d461b"}]},"branch":"refs/heads/stable/ussuri"},"0f63b64b3030a8f9ef845e4e9f086d5c45b1eb0e":{"kind":"REWORK","_number":9,"created":"2020-06-04 12:39:37.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/98/731798/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/9"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-04 12:39:33.000000000","tz":0},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f63b64b3030a8f9ef845e4e9f086d5c45b1eb0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f63b64b3030a8f9ef845e4e9f086d5c45b1eb0e"}]},"branch":"refs/heads/stable/ussuri"},"4d0972d4971a5d590c03598d401c30bb97576273":{"kind":"REWORK","_number":10,"created":"2020-06-04 14:12:28.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/98/731798/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/731798/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/731798/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/731798/10"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 16:17:19.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-04 14:12:13.000000000","tz":120},"subject":"Switch to RDO Ussuri release on CentOS","message":"Switch to RDO Ussuri release on CentOS\n\nThis change switches from the Delorean yum repositories to the RDO\nUssuri release repository.\n\nChange-Id: Ifa0614512a314a114f06ac2a39f646fddf839e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d0972d4971a5d590c03598d401c30bb97576273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d0972d4971a5d590c03598d401c30bb97576273"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
