)]}'
{"id":"openstack%2Fkolla~731862","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I12d74b53b64268d0358220af35f7fd719d82b397","project":"openstack/kolla","branch":"stable/ussuri","topic":"retire-congress-stable/ussuri","hashtags":[],"change_id":"I12d74b53b64268d0358220af35f7fd719d82b397","subject":"Add releasenote for congress deprecation","status":"MERGED","created":"2020-05-30 00:03:53.000000000","updated":"2020-06-09 10:07:49.000000000","submitted":"2020-06-09 10:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731862-1591697121344-00e32a87","meta_rev_id":"92d794319490cc143a1a51d51e68f1560017e864","_number":731862,"virtual_id_number":731862,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 10:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2020-06-09 08:48:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-09 08:49:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-06-09 08:49:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-30 00:03:53.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2020-06-09 08:48:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-09 08:49:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-09 10:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"962e81e3dd5f87bb9926713c976de19f013ed00b","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-30 00:03:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5529a903c2f49df8591962659d8a005190128b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 00:09:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1fe9477f8444e5fbb0d0bc8206c8bf4 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d88f2c8e7484c2abea32ab52fd360c8 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e182cd2f3674e1b951aafd9f2d349ad : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"35d1ae2a456a474959749189d03380282422dcec","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-30 14:39:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2144f33579a7065fea55564dcf2ece7e165421c3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-09 08:41:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"34cc94751f4b87c7dfeff32d8a2524b7710b4cda","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-09 08:42:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"32f61f8f9c69888f7cfde2693cbb4318bfa49de1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-09 08:42:34.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"33d16824e59fbbec5c06a6edd478b9f90e1b510d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-09 08:48:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0629de937552fb99a13f9757c8910cf6f1349e48","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-09 08:49:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bf0bc9c00289f1248589c29ff1678435209d251f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 08:51:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d9edd763a14240abf77b287fccc1fb : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c6251ea91e483c8c6fac5208e26920 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5e1c142c38342acbd0b9d41d7938114 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":4},{"id":"1829e442c4f5a26994a713596bc23da621dc70a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 08:52:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bcceb1d4c43a9be02aadaaeb98b5d4f04d092640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:05:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6f150def9040bf8355d020ca11e8f0 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2577c6bb6c5a4586bd4544e0c06585ef : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/670e1cb74a094bbe8a2f708789797ec2 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4},{"id":"aced73f9f50e0956f4b555d411f3b945df3d7496","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:05:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8d272bf97070999f32ae597ea70035a1a5264a1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:07:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d6fde18a066458f9a8e66ce4478e181 : SUCCESS in 2m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e38d1eee87843238f1386729dfa0c48 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"97298440a2ac4dbacdd88a900e0f2026606ef644","revisions":{"2542cad41cabb54d02e3667819d54e3674c9662d":{"kind":"REWORK","_number":1,"created":"2020-05-30 00:03:53.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/62/731862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/731862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/731862/1"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-29 13:33:33.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yingjunz@gmail.com","date":"2020-05-30 00:03:53.000000000","tz":0},"subject":"Victoria: Add releasenote for remove congress images","message":"Victoria: Add releasenote for remove congress images\n\nThe congress project is no longer maintained. This has been\nretired since Victoria[1] and has not been used by other OpenStack\nservices since.\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2020-April/014292.html\n\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I12d74b53b64268d0358220af35f7fd719d82b397\n(cherry picked from commit 6815f917077a3494be04420e3906dde1d5fe68fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2542cad41cabb54d02e3667819d54e3674c9662d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2542cad41cabb54d02e3667819d54e3674c9662d"}]},"branch":"refs/heads/stable/ussuri"},"93fd598941aa184a2464069a4fde67dd7feb038e":{"kind":"REWORK","_number":2,"created":"2020-06-09 08:41:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/62/731862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/731862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/731862/2"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-29 13:33:33.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-09 08:41:13.000000000","tz":0},"subject":"Victoria: Add releasenote for remove congress images","message":"Victoria: Add releasenote for remove congress images\n\nThe congress project is no longer maintained. This has been\nretired since Victoria[1] and has not been used by other OpenStack\nservices since.\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2020-April/014292.html\n\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I12d74b53b64268d0358220af35f7fd719d82b397\n(cherry picked from commit 6815f917077a3494be04420e3906dde1d5fe68fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93fd598941aa184a2464069a4fde67dd7feb038e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93fd598941aa184a2464069a4fde67dd7feb038e"}]},"branch":"refs/heads/stable/ussuri"},"1529b38f058665262a9fbc7b6db9a69d10893ab1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-09 08:42:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/62/731862/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/731862/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/731862/3"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-29 13:33:33.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-09 08:41:57.000000000","tz":0},"subject":"Add releasenote for congress deprecation","message":"Add releasenote for congress deprecation\n\nCongress has been retired. See [1].\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2020-April/014292.html\n\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I12d74b53b64268d0358220af35f7fd719d82b397\n(cherry picked from commit 6815f917077a3494be04420e3906dde1d5fe68fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1529b38f058665262a9fbc7b6db9a69d10893ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1529b38f058665262a9fbc7b6db9a69d10893ab1"}]},"branch":"refs/heads/stable/ussuri"},"97298440a2ac4dbacdd88a900e0f2026606ef644":{"kind":"REWORK","_number":4,"created":"2020-06-09 08:42:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/62/731862/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/731862/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/731862/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/731862/4"}}},"commit":{"parents":[{"commit":"98c342da5d7ae17eac075b47e27f2c472c374b28","subject":"Merge \"[Community goal] Update the contributor guide\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c342da5d7ae17eac075b47e27f2c472c374b28"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-29 13:33:33.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-09 08:42:33.000000000","tz":0},"subject":"Add releasenote for congress deprecation","message":"Add releasenote for congress deprecation\n\nCongress has been retired. See [1].\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2020-April/014292.html\n\nDepends-On: I21c9ab9820f78cf76adf11c5f0591c60f76372a8\nChange-Id: I12d74b53b64268d0358220af35f7fd719d82b397\n(cherry picked from commit 6815f917077a3494be04420e3906dde1d5fe68fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97298440a2ac4dbacdd88a900e0f2026606ef644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97298440a2ac4dbacdd88a900e0f2026606ef644"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
