)]}'
{"id":"openstack%2Fkolla~732153","triplet_id":"openstack%2Fkolla~master~I93276b2dc53e729f918fb3e680dd78a6013c109e","project":"openstack/kolla","branch":"master","topic":"to-revert","hashtags":[],"change_id":"I93276b2dc53e729f918fb3e680dd78a6013c109e","subject":"[to-revert] Pin ovsdbapp in neutron to 1.1.0","status":"MERGED","created":"2020-06-01 10:40:53.000000000","updated":"2020-06-01 14:30:55.000000000","submitted":"2020-06-01 14:18:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732153-1591021087085-c0d76a93","meta_rev_id":"edc8a95ba833093ed3fd3b1aa33875332ec8a7b7","_number":732153,"virtual_id_number":732153,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 14:18:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-01 11:36:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-06-01 11:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-01 12:39:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-01 12:39:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-06-01 10:41:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-01 10:41:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 10:41:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-01 11:08:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-01 11:36:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-01 14:18:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ae13b601b22082877d552daaf7502e73b8597f0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-01 10:40:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d71bec44550b1f641239852b160114b7ee62fd11","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-01 10:41:08.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"7a201aea69be49e0cd0cafbc4cb19caa435ef054","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-01 11:08:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a735ff8685bc3ca9f223a129d672502ce8b0d85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 11:10:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9cb07407f67747698d965e87c5dbd200 : FAILURE in 25m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1110bcbeb13749769107fd8c74028556 : FAILURE in 21m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a2ef60e69a640aeb296070e09ebfc99 : FAILURE in 25m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/79c6db94387b4913a1bc507c0227115a : FAILURE in 22m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51fe5ace2d88885c73569ad47f8678315ce54148","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-01 11:36:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7942072f9619468694bbea35fcc419a7 : SUCCESS in 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0477f42d1f597db695246c22279b59a0be89597a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 12:21:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bfa35ed940574a5b9bc3a763e6d6024e : SUCCESS in 1h 16m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6a163c4fca53469b967146dfd189ab35 : SUCCESS in 26m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bf20ba32164f456997a4439fb9a4fbcc : SUCCESS in 1h 34m 10s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/091147ce3aab41dabce3b0250ff967e3 : FAILURE in 48m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cd5be4e3750442e791df807068708de6 : SUCCESS in 1h 38m 30s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f3a0ec9ef262406f9081f23991486903 : SUCCESS in 58m 15s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/176403230cc14d4985b05d9d79c45154 : SUCCESS in 1h 07m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/077bae90658744c1a0ce5ebf76d90c70 : SUCCESS in 17m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6a2a2d6423be447885ae6833ca57ddf6 : SUCCESS in 1h 00m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb2d05a1baae4fa589bd76673b2aaf01 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00397b0076034f47bce8f4641c5e8a47 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18bc53714ca84a77a73507d2862355a7 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4148bf9cf0c4bb99c8882ac7593491d : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b33be65ab746dd9d05bc8e197a4291 : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f661399d924c4c5eb32c671e13c3ba7b : SUCCESS in 1h 02m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9765d4918a9344f8b20247c3669fb7a6 : SUCCESS in 24m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3eb383d3467463786d73f612ced5158 : SUCCESS in 1h 01m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1caa319be8b14b1fb867046da736febe : SUCCESS in 1h 30m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9720c4ec2d90449f8be066601a9a944e : SUCCESS in 1h 28m 32s","accounts_in_message":[],"_revision_number":1},{"id":"446b950c0d6471e6f194634aa1b5de645db2af07","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-01 12:39:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\njust fix that gate (fingers crossed for publishing tomorrow morning)","accounts_in_message":[],"_revision_number":1},{"id":"20e38b8c47f3a6ec36412563e54cacbfc48b8c10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 12:39:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3bb23569c7da08ea2c15beebe44eeaf498f67c4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:18:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2d17e01dc6bc4084bdeb74f813d04c61 : SUCCESS in 49m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/46491c73478f4fe59ecff724391e45d7 : SUCCESS in 1h 29m 16s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3507520a9ca34e95858f12c03de5f1fe : SUCCESS in 1h 37m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efc727ecace74972afd5a6a4b163086f : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2410b95aaa58430abbc6dfbdfbac9992 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ef4c71251894c35bb1a407ad9a2ae7a : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a26cd4cc92d4c6bbfeedc49a216fc24 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97757b5c641a4f6780666224d82f768c : SUCCESS in 7m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/505b99093afb4491b72210509b2eebea : SUCCESS in 56m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f78d5007c1042f0ab517f1fb286b3f9 : SUCCESS in 1h 05m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/65d64d167a18493cb6a58a317e4895ff : SUCCESS in 1h 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b44168dc81744876c1a2cbee50771a6f908ff905","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:18:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d2431d3580a3e5604695b9a16a4ac4879d568b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:20:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df482b6d28f8427ca0382771d7610de9 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1db6d11dfab14dc907a17729a888149f941827b6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-01 14:30:55.000000000","message":"Created a revert of this change as I75a17bab92c9c43c4429fe4283cc63d530c99bef","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83ef520d640783b7745368ab1654aec59afa6461","revisions":{"83ef520d640783b7745368ab1654aec59afa6461":{"kind":"REWORK","_number":1,"created":"2020-06-01 10:40:53.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/53/732153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/732153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/732153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/732153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/732153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/732153/1"}}},"commit":{"parents":[{"commit":"6815f917077a3494be04420e3906dde1d5fe68fd","subject":"Victoria: Add releasenote for remove congress images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6815f917077a3494be04420e3906dde1d5fe68fd"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-05-30 17:29:09.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-01 10:39:24.000000000","tz":120},"subject":"[to-revert] Pin ovsdbapp in neutron to 1.1.0","message":"[to-revert] Pin ovsdbapp in neutron to 1.1.0\n\nThis fixes the related bug for us while we wait for common\nrelease.\n\nChange-Id: I93276b2dc53e729f918fb3e680dd78a6013c109e\nRelated-Bug: #1881424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83ef520d640783b7745368ab1654aec59afa6461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83ef520d640783b7745368ab1654aec59afa6461"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
