)]}'
{"id":"openstack%2Fkolla~732495","triplet_id":"openstack%2Fkolla~master~I67dc11efe667e2cc67ea1a731edc5c4a03ebd1e7","project":"openstack/kolla","branch":"master","topic":"bug/1778146","hashtags":[],"change_id":"I67dc11efe667e2cc67ea1a731edc5c4a03ebd1e7","subject":"Fix pre-commit-hook","status":"MERGED","created":"2020-06-02 08:31:40.000000000","updated":"2020-06-04 14:20:20.000000000","submitted":"2020-06-02 21:23:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732495-1591133022762-edd24659","meta_rev_id":"a00d211962cb713b37b07936822edb291c98263f","_number":732495,"virtual_id_number":732495,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 21:23:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-02 10:08:48.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-02 19:49:40.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-02 19:49:40.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 10:08:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-02 19:49:40.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-06-02 21:23:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"607488fcf708c483ca7695e7ebc77ee0d4b55175","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-02 08:31:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6407dad85931f40ccba4bc31accd090a27ed7d66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 08:59:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d425cde8c50496f925a64f313f09dad : FAILURE in 13m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6e08c2831a2432d8c5258b58560695e : FAILURE in 18m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1595773bb8284e53ae401a9c5727a39c : FAILURE in 22m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b77ef2a793504cfe808913d7ed99f38f : FAILURE in 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"442697055ca7eeec82d7d3ccddde90a142c535ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 10:00:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2ebff393e149448c8fc4573639d357d0 : SUCCESS in 52m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/251aba2b4ac7445b85236dce27834748 : SUCCESS in 42m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/33c26fb79c994ab3bb8b79d17af03dd5 : SUCCESS in 1h 26m 00s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3a5d764df1da416e8aa7d495007c2ebe : FAILURE in 55m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/58139ff8caf84f5ea608644c6479ff2a : SUCCESS in 1h 23m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4edd9f5dac264bfda0b6b9429e1244e1 : SUCCESS in 33m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4f08f438d852461b904c204b8618562f : SUCCESS in 27m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/120a0b5b680642e4b285bb41eab58ac0 : SUCCESS in 1h 07m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4684b400859f4dfa937ce72d34815386 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba978d88227d4e389e9ec6c95b45ed12 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ee79d5e6a5340339553ba03a05e6d84 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20e72ac785b94c9cb2b408110a98666f : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1ac9010dbf480f955450f0ea15aee0 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47ee923e57264da3bf8c6337b27a04fd : SUCCESS in 21m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a666035f8bf945108a260db273656029 : SUCCESS in 33m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6332745f20104da390ba99eea43d9043 : SUCCESS in 1h 19m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc05c82bd9ca458ca289700f378663af : SUCCESS in 1h 11m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d0fa3a6f52c4edca727741574da5032 : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":1},{"id":"bae34b8b15ebaa005cca566e2e5ca2efe55de3ea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-02 10:08:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce90b34f5ea68b403b94cb6cf35a8a4505e74765","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-06-02 19:49:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e693be2ef9cb478eca446c23dfce650014b8be7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 19:49:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"456535af68d1797de07177c10ac5cd3e45d4ebcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 21:23:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/193a66220f3844f7a761b9353662b2e9 : SUCCESS in 40m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b9314cbacb46481490c7c0a0e3027e0f : SUCCESS in 1h 21m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/57bc59aba00a4896af154af5b16bff8a : SUCCESS in 1h 32m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d291d09149474332accba4bb5edf8de5 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/879933ad348c498b99976f78f5f2f1d6 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81b9c4ec45bd472cb9b4f85b5555401e : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe3ad50f230242d6ac9f29ebd5b2db35 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad3fae6e23c4dae95bcd578afc76842 : SUCCESS in 6m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd1d05df169e4df3aa7986dafdab6493 : SUCCESS in 23m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/228f960c20564e4b988ce8ae86412383 : SUCCESS in 1h 04m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/100828119039437eae57dda5429db33f : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2bc7d575fa792a99aee291f71129f1c9eaa8f5cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 21:23:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"037f59d5338894495dc2094920e6f5c177a4e179","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 21:25:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c52c13dd80498aa02ff5fcb20e9d83 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1},{"id":"addc1128abd9c77139ef03e1724e79907565648d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 14:20:20.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit e246e80d23a2b0bb118a1cc89d5542935fba0ed7","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"712b258f2d1870b3f07fb67b043769999edb3ef5","revisions":{"712b258f2d1870b3f07fb67b043769999edb3ef5":{"kind":"REWORK","_number":1,"created":"2020-06-02 08:31:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/95/732495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/732495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/732495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/732495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/732495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/732495/1"}}},"commit":{"parents":[{"commit":"6d223d3005bab377339f2d91e53458f1037c4bf9","subject":"Merge \"README: mark kolla-cli as deprecated \u0026 add kayobe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d223d3005bab377339f2d91e53458f1037c4bf9"}]}],"author":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-02 08:16:52.000000000","tz":0},"committer":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-02 08:31:33.000000000","tz":0},"subject":"Fix pre-commit-hook","message":"Fix pre-commit-hook\n\nCurrently pre-commit-hook is failing due to yaml validations using\nunversioned python and using non-existent json validations.\n\nChange-Id: I67dc11efe667e2cc67ea1a731edc5c4a03ebd1e7\nCloses-Bug: #1778146\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/712b258f2d1870b3f07fb67b043769999edb3ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/712b258f2d1870b3f07fb67b043769999edb3ef5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
