)]}'
{"id":"openstack%2Fkolla~733144","triplet_id":"openstack%2Fkolla~master~Ie2e17bba37494b320986c17b30b15b62dd71ba22","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie2e17bba37494b320986c17b30b15b62dd71ba22","subject":"Remove shade and kolla_sanity.py from kolla-toolbox","status":"MERGED","created":"2020-06-03 13:41:36.000000000","updated":"2021-01-02 20:43:31.000000000","submitted":"2021-01-02 20:42:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733144","meta_rev_id":"c3de8b45643fdac90c675881f0b6c71aea84905b","_number":733144,"virtual_id_number":733144,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-02 20:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-03 15:57:32.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-04 10:05:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-04 14:24:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-04 14:24:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 15:57:32.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-03 16:47:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 10:05:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-04 14:24:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"1eb768a28e1681e87581e7e9d2fb370c63b5655b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-03 13:41:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54afd5d69478494abbd37eb674ce92e979ef493e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-03 13:45:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c2fbec8a034b4a36d3cc48ed21861150b7d1fb3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 14:44:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4c5be4a12c1f4437b2151a7b51723331 : FAILURE in 38m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/41a209983b3449939f0b93cca16a7bcb : FAILURE in 42m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/42b0e93e37ca47829f3cf0808f8b3ebd : FAILURE in 26m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/00a7afdf5d854ab5933449274b316cbf : FAILURE in 21m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/67dcba8f288d48178228b9eed7521f0d : FAILURE in 24m 33s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/35ca7f03b3c04ae3bfe4bea8fc83e0bd : SUCCESS in 58m 05s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d463b4005ed24cc1823e17b944242067 : FAILURE in 23m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f6eceb80e62e493d8162f86e6c84eee8 : FAILURE in 19m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4d125a8c7ba04a93aa6173c1e48ce704 : FAILURE in 19m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec0ca110a29045b1a2bac5bd6947a2eb : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/061eb537dd0e4b3889cd64ef656c04f0 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b199d5d43874b97b3f1ca432ae83e8f : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f43c622b9e97445f91824151a2e1dc33 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7310dc9a1028436d9d9de9e110aec691 : SUCCESS in 6m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c38cc62eb3a04bd39bd298e30f21ecac : FAILURE in 26m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ff8e48e939b49d3a3be8a6a532435be : FAILURE in 25m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/802c0306ce0a40fc9a99b0d59e94a361 : FAILURE in 22m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c099d2fcc16d41ac913baf33e54fe34e : FAILURE in 18m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/54a2ccfafa074b6b890962680aa0092d : FAILURE in 20m 20s","accounts_in_message":[],"_revision_number":2},{"id":"353516da76492778ce3c25aa916bd0e682d79a20","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-03 14:45:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2815aec0e16a1b9b8fbc8a6aaaf5a77a0b94e56f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-03 15:57:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/f4235d4295794e75a080a4c6c330f509 : SUCCESS in 30m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6b417dc963df4187e54c072f086162eb2de3189","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:39:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/76b8bc921c154f5593f8ea82cfda781f : FAILURE in 59m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b230f09940e4c43a389a7c36ebac3b7 : SUCCESS in 56m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/67de538d7bd7440fa16a7a320608b2cd : SUCCESS in 1h 43m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/707e2dbd38fc4aa9a13ea144dc8decdd : FAILURE in 54m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"780e656fc2477f697293fb063063fd457ad34c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:47:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/91d72aa397a743aebe30bff0fd62bf45 : SUCCESS in 23m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0e471d47b25a4d5ab72576c595b1c219 : SUCCESS in 31m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a3bafb29b66b444da1496b70017bf60d : SUCCESS in 1h 07m 27s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f7b63580cdd845e981b9d01e77b566e2 : FAILURE in 53m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e304e8612b3748889d84bf51c9a10f26 : SUCCESS in 1h 31m 56s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f2496017d0144970a46315271cbb9f69 : FAILURE in 1h 33m 07s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6430efe613cb4e9abf20d0c24c4ad976 : SUCCESS in 1h 27m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e3dddf5595cd42c1bd2d52b4d1ed5dc1 : SUCCESS in 19m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d2b31c10ae974ed0a51fba7f679abfb8 : SUCCESS in 2h 00m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5dd05d054194411b9ff908881b28bcd : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9fd4c4d1ec458bac249f009ebf4960 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38c2b6f73ee647bf8f796f1c4ab35abb : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f507520ed4f4d01a7b815dbee1ad158 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c799b8e5c9c3461f99777e790180ea83 : SUCCESS in 7m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1eaf52c5bca945ec94de4630f9f6268e : SUCCESS in 56m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc779455060c45d2bce7646aa8e325c1 : SUCCESS in 28m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ddd366d8f2746ccb4b3b32763eeb314 : SUCCESS in 1h 05m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/df0ae780f81b47478368db6d503f1e69 : SUCCESS in 1h 04m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/afa7696f00fd49d7a080422a91a49968 : SUCCESS in 1h 09m 10s","accounts_in_message":[],"_revision_number":3},{"id":"e0c6127feb8de60aab9e39454a2f05eed61dff0c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 10:05:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b49913fcf3094796fc099430bbb636a160dfd513","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 14:24:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"02eba1ed5b58c83834f60a5213b4b03f0ed83669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 19:29:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d8a8943927c9d068dfb76afa523d4302e0c9af0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 20:42:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/316fdab5454d4664b4f8cc58664a74ef : SUCCESS in 37m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3d6a3c31dc384604a14a88a02362c0c7 : SUCCESS in 1h 02m 05s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/eb471280cc2642efa3f4791a6d2cccdf : SUCCESS in 1h 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef2b40a2dc42404a821e4addd83370f0 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b90d21c0c41942708236125de07a2135 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8babf0acd90c42ceac666fb179218436 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e987d5b3e0774576aaf1d3fe934efc64 : SUCCESS in 5m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7040eb4e15134019b5aa8ad9f05f37e6 : SUCCESS in 45m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de61544db4074bdea3316803f55a731e : SUCCESS in 48m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/73357aa91c9b43d3ab954bee11552d40 : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":3},{"id":"123bc52d4bcf9460d96595debf10231a09b2f031","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 20:42:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c3de8b45643fdac90c675881f0b6c71aea84905b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 20:43:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0d4d1a1995f43e68347a3ef8206d78f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"231ef23ce1125dbe17cc2568d143c0d7da2c2141","revisions":{"74d704a2844084482ebaff126f9d41784824d229":{"kind":"REWORK","_number":1,"created":"2020-06-03 13:41:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/733144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/733144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/733144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/733144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/733144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/733144/1"}}},"commit":{"parents":[{"commit":"fa5abde6c60c92b6996f051095a0a9cbd08d8227","subject":"Merge \"Add ceph releases per-distro to support matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa5abde6c60c92b6996f051095a0a9cbd08d8227"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-03 13:18:50.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-03 13:41:32.000000000","tz":120},"subject":"Remove shade and kolla_sanity.py from kolla-toolbox","message":"Remove shade and kolla_sanity.py from kolla-toolbox\n\nShade is deprecated long time, and only kolla_sanity.py utilised by k-a swift\nrole is using that.\n\nDepends-On: https://review.opendev.org/733143\nChange-Id: Ie2e17bba37494b320986c17b30b15b62dd71ba22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74d704a2844084482ebaff126f9d41784824d229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74d704a2844084482ebaff126f9d41784824d229"}]},"branch":"refs/heads/master"},"7d514e598d1d3cfcf69b6ebbb8634e45f8f4e6cd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-03 13:45:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/733144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/733144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/733144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/733144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/733144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/733144/2"}}},"commit":{"parents":[{"commit":"fa5abde6c60c92b6996f051095a0a9cbd08d8227","subject":"Merge \"Add ceph releases per-distro to support matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa5abde6c60c92b6996f051095a0a9cbd08d8227"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-03 13:18:50.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-03 13:44:56.000000000","tz":120},"subject":"Remove shade and kolla_sanity.py from kolla-toolbox","message":"Remove shade and kolla_sanity.py from kolla-toolbox\n\nShade is deprecated long time, and only kolla_sanity.py utilised by k-a swift\nrole is using that.\n\nCloses-Bug: #1760517\nDepends-On: https://review.opendev.org/733143\nChange-Id: Ie2e17bba37494b320986c17b30b15b62dd71ba22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d514e598d1d3cfcf69b6ebbb8634e45f8f4e6cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d514e598d1d3cfcf69b6ebbb8634e45f8f4e6cd"}]},"branch":"refs/heads/master"},"231ef23ce1125dbe17cc2568d143c0d7da2c2141":{"kind":"REWORK","_number":3,"created":"2020-06-03 14:45:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/733144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/733144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/733144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/733144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/733144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/733144/3"}}},"commit":{"parents":[{"commit":"fa5abde6c60c92b6996f051095a0a9cbd08d8227","subject":"Merge \"Add ceph releases per-distro to support matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa5abde6c60c92b6996f051095a0a9cbd08d8227"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-03 13:18:50.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-03 14:45:15.000000000","tz":120},"subject":"Remove shade and kolla_sanity.py from kolla-toolbox","message":"Remove shade and kolla_sanity.py from kolla-toolbox\n\nShade is deprecated long time, and only kolla_sanity.py utilised by k-a swift\nrole is using that.\n\nCloses-Bug: #1760517\nDepends-On: https://review.opendev.org/733143\nChange-Id: Ie2e17bba37494b320986c17b30b15b62dd71ba22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/231ef23ce1125dbe17cc2568d143c0d7da2c2141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/231ef23ce1125dbe17cc2568d143c0d7da2c2141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
