)]}'
{"id":"openstack%2Fkolla~733498","triplet_id":"openstack%2Fkolla~master~Ic6785e93dfae45b81a4cb3bf9156a822de835fde","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ic6785e93dfae45b81a4cb3bf9156a822de835fde","subject":"Remove centos-nfs-ganesha28 from repos.yaml","status":"MERGED","created":"2020-06-04 09:08:05.000000000","updated":"2020-06-04 14:29:29.000000000","submitted":"2020-06-04 14:26:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733498-1591280780913-1788e675","meta_rev_id":"35be219b2bcdbb152f81d8b017ff34f3f020a327","_number":733498,"virtual_id_number":733498,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 14:26:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-04 11:10:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"recommended":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2020-06-04 09:21:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2020-06-04 12:36:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2020-06-04 12:36:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 09:21:26.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-06-04 11:10:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 12:36:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-04 14:26:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93adf1165d6343ebbb424772fcf73a3d56fdb6d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 09:08:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fdfc1e794c49c9a3dfa170ee8180b0b59cea5ca","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-06-04 09:21:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2fe5a2031720a14222e22ae0d4fd9eee581ead6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:36:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a462743d63af4d95b7443287e12ee2ed : SUCCESS in 52m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/969d5aeb0ac24f429d72d097f13f2ab5 : SUCCESS in 1h 23m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6d12b2d53f524406922f3c4a6c017094 : SUCCESS in 1h 16m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3615342d6ddb45f4b7f29a2bf4c6ba66 : FAILURE in 54m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5198132e8b5b49b5abf621d2853aa697 : SUCCESS in 1h 24m 53s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3bed68f9232146738f89bcded909cd9c : SUCCESS in 36m 59s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d25cf98155694de9a561525dde06f58d : SUCCESS in 33m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f5665e75bc124115af1fd79b06865076 : SUCCESS in 25m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3aa49995ae8c481b9365ed7eae15acb9 : SUCCESS in 1h 13m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6b2a08bbf2b4db98e39e56e71aa7234 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f28a8c433fc4cdda561811c294fbcdb : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5e81f685b25447798b3b13ec59b4cd6 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c81105c1632f46528d04ad7fb08ce796 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79386413e4fe482ba5fcb90082a21853 : SUCCESS in 6m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0d463320a0b349f4913fad664e2f126e : SUCCESS in 21m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0054d77bd734c6d9ed1b66f01659bb9 : SUCCESS in 22m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/703a3b9a23c443c5b9c39b4ed93eba15 : SUCCESS in 1h 03m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8dec8fb6905a4e618aa4d0ff28ed1aed : SUCCESS in 1h 01m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1d2c9836217b4a2a816a343721008d0a : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"995efd52491bb76a768e0a984ff55393d38d9dcd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-04 11:10:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b9b55d389b3f4c9d9ac2692f6e33dd86 : SUCCESS in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4551e4ce112ca0b067a68073fdde2d887acdb7d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:47:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ec67bafad744784b12fc6d1e0649966 : SUCCESS in 1h 14m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2205303b56724f528085fe6316e10759 : SUCCESS in 1h 21m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3408336bed23425faa9bd1334070eebb : SUCCESS in 2h 32m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/331bdd8041d94755bdf8450cf5b9fac0 : SUCCESS in 1h 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80a8193b795bac397cf386e7070fa3dcfa7beb58","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 12:36:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d115f829d135f818199707cb18c7e82ace0a334a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:36:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b9d2546ffb7852848b026908b3fbcc46cbbed78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:26:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3182aeba146609a425a13ccec6607a82a2070f1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:26:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/620df4dfba9f4166b9b48059e9bdf78c : SUCCESS in 29m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fb65c0b7342b47bca67459ae3b3783ae : SUCCESS in 1h 26m 06s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a8677609794e4ee5a9d7b96100813b5e : SUCCESS in 1h 22m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32f32bbe23c44623a29649ec65008805 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6063e1b60b44417080c75096c40994b1 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73751ee5014b4f64b46afc57a25152c4 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdfa27a570124c06bd43be9c29fdbe6a : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c5e71498a864a43948651f8b447adee : SUCCESS in 8m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a8e395500a9413c98bfd02a41919c38 : SUCCESS in 56m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/210e4d32a9b2416b97fe680c8da187f5 : SUCCESS in 1h 43m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f0d5eea434f449b2a47e6405675983da : SUCCESS in 1h 48m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6f401dfa3036e2b87ebd5867ee0175de219d6bc9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 14:27:04.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit c001a712b0c7140a2b79c67f1b221845eb5f5063","accounts_in_message":[],"_revision_number":1},{"id":"35be219b2bcdbb152f81d8b017ff34f3f020a327","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:29:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd710c40ccd42b69df9833af819bfed : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d82ad8af979002de84a3808587b5e979929e4b42","revisions":{"d82ad8af979002de84a3808587b5e979929e4b42":{"kind":"REWORK","_number":1,"created":"2020-06-04 09:08:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/733498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/733498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/733498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/733498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/733498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/733498/1"}}},"commit":{"parents":[{"commit":"e37e876c82a668f763a8a9261c36f7c8a0f6b85f","subject":"Use CentOS official mirrors for Ceph packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e37e876c82a668f763a8a9261c36f7c8a0f6b85f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 09:07:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 09:07:51.000000000","tz":60},"subject":"Remove centos-nfs-ganesha28 from repos.yaml","message":"Remove centos-nfs-ganesha28 from repos.yaml\n\nFollow up to I913efffe6a1d8a0210b1158261c77d0d45ac3147.\n\nChange-Id: Ic6785e93dfae45b81a4cb3bf9156a822de835fde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d82ad8af979002de84a3808587b5e979929e4b42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d82ad8af979002de84a3808587b5e979929e4b42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
