)]}'
{"id":"openstack%2Fkolla~733557","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ida7e20833360bbca69c6aafc4f5d7cb375bb106f","project":"openstack/kolla","branch":"stable/train","topic":"bug/1882070","hashtags":[],"change_id":"Ida7e20833360bbca69c6aafc4f5d7cb375bb106f","subject":"Fix multiple issues","status":"MERGED","created":"2020-06-04 11:08:07.000000000","updated":"2020-06-08 18:39:32.000000000","submitted":"2020-06-08 18:38:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733557-1591641486880-c28173df","meta_rev_id":"8022586d4be0d46ff04d42b17f76687ffa9671af","_number":733557,"virtual_id_number":733557,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 18:38:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-06-08 14:53:57.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-06-08 15:34:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-08 14:52:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:32:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-08 14:53:57.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-08 18:38:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b67ffe022124d7660ddef5c2f4d0c91d702e0f0b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 11:08:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea61c14420b43014c3fdc8eec10253219f4c0b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:04:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/58d380125d944f409ab87ad59095b997 : SUCCESS in 53m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f4fc14a789e340fa9f52b3f76d55d28c : FAILURE in 1h 14m 42s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f8f92f7e84b64a778b71d3c278e81691 : SUCCESS in 24m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ce7e8a87349944fc8f75b24f1903b3b0 : SUCCESS in 28m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5b26fabf8f3e4b77a2e6ef790ee9b8a0 : SUCCESS in 55m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/df787bb7b86e4c2ab0b7a9772c9b26b6 : SUCCESS in 1h 07m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/33190fb7e4914e608fdb13519bd4ac1e : SUCCESS in 1h 26m 21s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/008b8489b9304a4f9c8954a647fe1e51 : SUCCESS in 1h 09m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8d18559699ac4768badbc4dcaa79af0f : SUCCESS in 1h 49m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/53d68d8989774048bbd9f90410c01b74 : SUCCESS in 55m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7fc5f258e8c14036af103563690000ac : SUCCESS in 27m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/55ed940733dd4c739e3f9a29135b0e25 : SUCCESS in 1h 41m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2b8b047e08674f3d9a611c855b42c69e : SUCCESS in 1h 38m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c2c0ea613ee4b5f8a84012bc4308f7f : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c6671f4bdb43da99a43587f03789cf : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7972eae6685a40929715a5ef88971267 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f11bdf2843f4afc8a90e6df4726987e : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e308c8bf252b42c782ace9683f67de1d : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5d9a55695d47f3a2c8ca619accf00b : SUCCESS in 9m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8c28df6315fd4491bce6d3a449c57bdf : SUCCESS in 1h 14m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e14451e12d040419bdad498a27c028a : SUCCESS in 32m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ea0eff44b3f481eb6280acfee6382c0 : SUCCESS in 40m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dcd465ee9f944b4f80e7f7d6dcef1334 : SUCCESS in 1h 04m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f906460557c547d28951a5a971f3316d : SUCCESS in 1h 05m 18s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"4e14e12faceea3d7ddcf2ba585b6d8f372622263","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 13:21:31.000000000","message":"Patch Set 1: Workflow-1\n\nBifrost needs a release.","accounts_in_message":[],"_revision_number":1},{"id":"231055001caa85e4ddfc925fcc54bcf01eee6d8e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 13:32:03.000000000","message":"Patch Set 1:\n\n\u003e Bifrost needs a release.\n\nBifrost screams: \"tier me\"","accounts_in_message":[],"_revision_number":1},{"id":"ef51515a0f15eeeaca1993a468f7c90f23f435ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 09:53:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e54387a38de6e67ea2f06b89f5d7ae77c2d00f87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 11:57:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/86b5ff05b06b4afc855734969d335e00 : SUCCESS in 46m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6df4e98e70c147788054556582bc30fc : SUCCESS in 48m 57s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8fb8861703c442068b1692a153bd526a : SUCCESS in 24m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/45ea0599508f4384af7bae3fb7fb2a20 : SUCCESS in 33m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e895c95f6135448aaa26e70bc7ba1494 : SUCCESS in 1h 34m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c95764d097104fa4a839a4880f7d9869 : SUCCESS in 1h 54m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/33d5a3d61923471b950079316aff994a : TIMED_OUT in 2h 03m 03s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/61c63da6f121431faddd38a79bf67977 : SUCCESS in 1h 05m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/269c8be0c5d4437f917ae49cdc2f9ef4 : SUCCESS in 1h 27m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9412bfa4c10041659017c73324d5e436 : SUCCESS in 1h 00m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7b3f632d8ed64201b8047026ad127581 : SUCCESS in 28m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ca20b2a46e434076890c9d93c9ae3736 : SUCCESS in 19m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d31b811eb99d4416bf1d35a9388da40c : SUCCESS in 57m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/263affcb041f42db9bd6cd59efa59438 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b48e7472bd347e29865fd77bd2b77f8 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/128c7c27583549bfacc65f3439ebfd96 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c2e8259253d4a8b9b8e842ded29cd79 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59dfb816e0484de5a884dda544dfc22f : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7a11a34cc2348b5be2cc7d0bed6b795 : SUCCESS in 7m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6afd1cb69b43423cbfc8285088a14898 : SUCCESS in 24m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10de05caadd346f6bd0c8b61d8a0b07b : SUCCESS in 37m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1789961aa47745b9ae32adffe0cb336d : SUCCESS in 55m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18d9f8d70cea4a0d8f29ab033ffe9e9d : SUCCESS in 53m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d9f66f6c93984e1e974c16ba9a5101b4 : SUCCESS in 1h 12m 04s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"8c0f8ae78f0a1d395c9bb75c34f86d0417215e31","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 13:30:37.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"353fb4d976ccdc7433906ccfc90ad62490eb8860","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-08 14:15:06.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nadd note which stable branches it applies to","accounts_in_message":[],"_revision_number":2},{"id":"57781df6ac7f97f5e58ef6de3be7ce58877d096f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 14:52:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c3d26b87cdd84ca22b316a1605e248883fb1765e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-08 14:53:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ea9124fe2f7c119e43af5b1eb32949e0589e8733","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 15:34:13.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e932b4bfe313da6a082ddafb8c92287788f9f274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 16:42:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f3cb2920b1664f48b513e529f96586e0 : SUCCESS in 42m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b27dba4c6a8b4813b46cfb0958da7f17 : SUCCESS in 1h 00m 32s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0be51591f80943b48776d16e27e771cc : SUCCESS in 31m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/57cd66ca5c9e4e8e9cc628d6da12ab73 : SUCCESS in 28m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/544305b7f262413e82854ee1b923d4ba : SUCCESS in 1h 26m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fb237ca90b0c4d8db579c9a21d8482d8 : SUCCESS in 1h 21m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3968f95bc10d4eb9823f8edf7623f6c6 : SUCCESS in 1h 08m 05s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7e45f64107b04b28a6ad445b1ada1244 : SUCCESS in 57m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3c7717126929464196e463a067db06d5 : SUCCESS in 1h 01m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5b19f45851444b4e86a899cfaa346c8c : SUCCESS in 32m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/80194e293ed5480f8dc8337e37862090 : SUCCESS in 21m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7e408ae76d08423cbf562ecf8bf120fd : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3f3ce9b00238479e8b05b7ee96e85eef : SUCCESS in 1h 06m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16e50b1f517b4af9ba12017dfce3c08a : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f768ff85a21425a9dc23073c0590202 : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8e35e2ebfae46dcb732bf79efb1f995 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0af68c6d29de48798ab5009f3d37dec4 : SUCCESS in 8m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56b9ab70a6fb4d8e81655847693618da : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda354192d5b4062bccd0a46e1315920 : SUCCESS in 6m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bcf18fa3aea84ab89b1804d04f8ce1c5 : SUCCESS in 26m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dac3f3bcbe954b91801f2c3a18a7c4ae : SUCCESS in 27m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/637b0a0b8e4443c2b91ff282c395cb39 : SUCCESS in 58m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9aa9aec1bbd549b18d5cbc2a1b1f8600 : SUCCESS in 1h 18m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/20c12a24afd74e7c89a4176640ca7e1a : SUCCESS in 1h 29m 23s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"151ea806a46861de7721aec2b78f10b502a5ce63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 16:42:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c050dec3e5af53513b72ac9e11fca1ebdb1be7c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:38:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"17d25edf32c6a5537bda215fa7a3e029fcf1d826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:38:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/18a2b9309d3f4afdac1164c47a0b6b8c : SUCCESS in 54m 56s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/395400ab0d8642aeaba0f2129e5d4346 : SUCCESS in 38m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6b77f69f1ee74130b49ff4d9146a9b95 : SUCCESS in 1h 29m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/440902c08c3e4af69a977d482138f997 : SUCCESS in 1h 12m 44s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6f1be6e76e0d4ca384bb8990442783d2 : SUCCESS in 1h 50m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c025ad8fca054b66adc982b36b9fb059 : SUCCESS in 52m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dcfe5bd19794437ae79c08bce453644 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b259baeff04c84a00dd970fa6b2dd7 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09a35e74ce4e452388783add87303dd5 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bad7abe4bd64f1a8bc34c61cd27eb1f : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02360c1d124843ef99f2babdaf1fb161 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053b8db841c34efdbdb2ea1fbc41e323 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f7925c5b1e0437eadb1cb00fccd1250 : SUCCESS in 1h 01m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c8dffcc6208421ca47cd5e8c149eace : SUCCESS in 56m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/012de2f6bb794fb997453118e5ac1ce4 : SUCCESS in 1h 15m 16s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"a756aed7e9f32ac68fcb1252699d770ae6df415f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:39:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8629aeb7740b4a4ab06819038dd7f48c : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3784d32d9ceacf0dad8057092752f1ae7be37ed3","revisions":{"5097a487109ac3975dfea297d5e5109329c1e8b2":{"kind":"REWORK","_number":1,"created":"2020-06-04 11:08:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/57/733557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/733557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/733557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/733557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/733557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/733557/1"}}},"commit":{"parents":[{"commit":"95f2b88a8aab6650cb1e12b491ed00692e6e7187","subject":"Merge \"nova-compute: add ndctl to expose NVDIMMs to guests\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95f2b88a8aab6650cb1e12b491ed00692e6e7187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 11:05:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 11:05:12.000000000","tz":60},"subject":"Fix monasca-grafana by pinning more Ruby gems","message":"Fix monasca-grafana by pinning more Ruby gems\n\nchildprocess 3.0.0 and ffi 1.13.0 depend on Ruby 2.3.\n\nChange-Id: Ida7e20833360bbca69c6aafc4f5d7cb375bb106f\nCloses-Bug: #1882070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5097a487109ac3975dfea297d5e5109329c1e8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5097a487109ac3975dfea297d5e5109329c1e8b2"}]},"branch":"refs/heads/stable/train"},"4403987f69032a507b4792f4f916671ee6d932b1":{"kind":"REWORK","_number":2,"created":"2020-06-08 09:53:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/57/733557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/733557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/733557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/733557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/733557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/733557/2"}}},"commit":{"parents":[{"commit":"95f2b88a8aab6650cb1e12b491ed00692e6e7187","subject":"Merge \"nova-compute: add ndctl to expose NVDIMMs to guests\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95f2b88a8aab6650cb1e12b491ed00692e6e7187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 11:05:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 09:51:52.000000000","tz":60},"subject":"Fix multiple issues","message":"Fix multiple issues\n\n1. Fix monasca-grafana by pinning more Ruby gems\n\nchildprocess 3.0.0 and ffi 1.13.0 depend on Ruby 2.3.\n\n2. Bump bifrost to 7.2.1\n\nBifrost had a python 2 fix to stop using the master branch of DIB.\n\nChange-Id: Ida7e20833360bbca69c6aafc4f5d7cb375bb106f\nCloses-Bug: #1882070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4403987f69032a507b4792f4f916671ee6d932b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4403987f69032a507b4792f4f916671ee6d932b1"}]},"branch":"refs/heads/stable/train"},"3784d32d9ceacf0dad8057092752f1ae7be37ed3":{"kind":"REWORK","_number":3,"created":"2020-06-08 14:52:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/57/733557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/733557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/733557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/733557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/733557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/733557/3"}}},"commit":{"parents":[{"commit":"95f2b88a8aab6650cb1e12b491ed00692e6e7187","subject":"Merge \"nova-compute: add ndctl to expose NVDIMMs to guests\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95f2b88a8aab6650cb1e12b491ed00692e6e7187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 11:05:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 14:51:57.000000000","tz":0},"subject":"Fix multiple issues","message":"Fix multiple issues\n\nBoth issues affect Train and earlier releases.\n\n1. Fix monasca-grafana by pinning more Ruby gems\n\nchildprocess 3.0.0 and ffi 1.13.0 depend on Ruby 2.3.\n\n2. Bump bifrost to 7.2.1\n\nBifrost had a fix to stop using the master branch of DIB which dropped\nsupport for Python 2.\n\nChange-Id: Ida7e20833360bbca69c6aafc4f5d7cb375bb106f\nCloses-Bug: #1882070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3784d32d9ceacf0dad8057092752f1ae7be37ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3784d32d9ceacf0dad8057092752f1ae7be37ed3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
