)]}'
{"id":"openstack%2Fkolla~733619","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I67dc11efe667e2cc67ea1a731edc5c4a03ebd1e7","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1778146-stable/ussuri","hashtags":[],"change_id":"I67dc11efe667e2cc67ea1a731edc5c4a03ebd1e7","subject":"Fix pre-commit-hook","status":"MERGED","created":"2020-06-04 14:20:20.000000000","updated":"2020-06-04 17:36:32.000000000","submitted":"2020-06-04 17:34:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733619-1591292086419-1728319f","meta_rev_id":"a0f6a3c61b2caceeb3e69369b7bc74cced3a81c9","_number":733619,"virtual_id_number":733619,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:34:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-06-04 15:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-06-04 15:05:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-06-04 15:10:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 15:05:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-04 15:10:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-04 17:34:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"edaaa0acb0feafda2ba0eded7033b6a8858d4702","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-04 14:20:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e04858df9829f143698be86cc5f2e82720427946","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 15:05:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e07e01b0af46110ded55ab8a42b7940cd220b91b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-04 15:10:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d41baaeeeacd8c1925d86c0f6238dccd1b6a1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:03:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4b869ec6822e4b13ba6fa5f851fed235 : SUCCESS in 23m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f307985d55e34ec6bc6ca6d01caa6153 : SUCCESS in 37m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/76003fb4f9d04869b7c595b46371c392 : SUCCESS in 1h 24m 37s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c25142f8c97942a4947cafabeb2bcd0d : SUCCESS in 1h 15m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/19c6430e159f47d591b893d37b1f6daf : SUCCESS in 1h 41m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/42a997ac050444378daec279ce565fc7 : SUCCESS in 17m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fda0c056cf31463e9be7f470d5999436 : SUCCESS in 49m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0a9ad933f6304200ab674ce072ddd542 : SUCCESS in 1h 27m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bb6c90efe29412c8892500858ad9514 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb19fdab87974b11b475ab9b4b2cccea : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34e2a3258af848018ae47a2777a843bb : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d364d98945b430f80a29c2559f7aba1 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7794232433f406baebbecbfe97f5ea4 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d62169676a24f8786662b9b0c951bb8 : SUCCESS in 27m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32103d71f6ea4edfa767b00ba3b0c1e8 : SUCCESS in 28m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8068c0c467ba4d8f931a80834b3c2728 : SUCCESS in 1h 05m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8171cc4b0f1f419bae5b4d708e0bb6c1 : SUCCESS in 1h 08m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/992473e2458945b3baaf86f24a676f0b : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"bc1016bc4cb232c827eb51fd5198d11c2aee02d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:03:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8683c46f3040d6b0938b3aa146d3040a56178d1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:31:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/27e11e4f2cb44346bcb69583c1eb975f : SUCCESS in 2h 22m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d80d194559c4d3f9a35e7c143ff24af : SUCCESS in 2h 14m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5b2f1857cd148f38e2973a328ef4eaf : TIMED_OUT in 3h 03m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1daa5c67064840c1af916d3a1a139b35 : SUCCESS in 2h 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fa4e70c8d0cf1cde9e35f30703909d12d79e595","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:34:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9bb3092fbbf100fb97ba6bbc805397941fd7d988","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:34:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b55b2422ded9480f91010f8fa4bd16c1 : SUCCESS in 1h 27m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b8e04cc0a7f54b3681a0e21dd48bf252 : SUCCESS in 1h 20m 58s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bb20bd7af3ae4057a08571dd9f7d4e83 : SUCCESS in 1h 29m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52a4aba6d6d64b97a9eaf8e4b48c1962 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95f947d78eaf47e2b54e7c5532a64802 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed27dae2fa624d2caafcb84e64d3a973 : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1aad1b1db5a641da9afbaa1960f71f1c : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcb9ed8601548358cb266ca5e4b5e92 : SUCCESS in 9m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5354ec5241d04021b84f80052ad1d23f : SUCCESS in 27m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c449524be3d7465895cde04ebedfeebf : SUCCESS in 1h 03m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8e6dfbdfebb442d39a42f9398153e6ae : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d7a345885a62bf96954dfa2d124b5868143e4d93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:36:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d90ff72b434d4c95a4e67c65941295 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e246e80d23a2b0bb118a1cc89d5542935fba0ed7","revisions":{"e246e80d23a2b0bb118a1cc89d5542935fba0ed7":{"kind":"REWORK","_number":1,"created":"2020-06-04 14:20:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/19/733619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/733619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/733619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/733619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/733619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/733619/1"}}},"commit":{"parents":[{"commit":"9655fa1f547a427e52a68737e21b007c7c339f67","subject":"Add ceph releases per-distro to support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9655fa1f547a427e52a68737e21b007c7c339f67"}]}],"author":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-02 08:16:52.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-04 14:20:20.000000000","tz":0},"subject":"Fix pre-commit-hook","message":"Fix pre-commit-hook\n\nCurrently pre-commit-hook is failing due to yaml validations using\nunversioned python and using non-existent json validations.\n\nChange-Id: I67dc11efe667e2cc67ea1a731edc5c4a03ebd1e7\nCloses-Bug: #1778146\n(cherry picked from commit 712b258f2d1870b3f07fb67b043769999edb3ef5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e246e80d23a2b0bb118a1cc89d5542935fba0ed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e246e80d23a2b0bb118a1cc89d5542935fba0ed7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
