)]}'
{"id":"openstack%2Fkolla~734131","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ida7e20833360bbca69c6aafc4f5d7cb375bb106f","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1882070","hashtags":[],"change_id":"Ida7e20833360bbca69c6aafc4f5d7cb375bb106f","subject":"Fix multiple issues","status":"MERGED","created":"2020-06-08 15:42:23.000000000","updated":"2020-06-09 11:32:02.000000000","submitted":"2020-06-09 11:30:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734131-1591702213983-de1c56f5","meta_rev_id":"916b432d9bcd80642dbc865c78012c8f896419d4","_number":734131,"virtual_id_number":734131,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 11:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-09 08:34:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-09 08:34:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-09 08:34:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 15:43:21.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-09 06:53:29.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-06-09 08:34:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-09 11:30:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eea5f372fa65bf1402e3eb719bf35fae2a27c905","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 15:42:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e4fbc7c6bb1d796c0d19a7d1b8f2f43057a3c61","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-08 15:43:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11c0183a5fc313462fdcd92bf1c7ef5bf5c80a15","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 15:45:40.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"1d7ee6b33694d0b4504039ad36981f06aab63526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:54:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8b3e42d3c2864859a3b4e02f49d5dc28 : SUCCESS in 46m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/07342676b2c9431998a63dbeb62c6b80 : SUCCESS in 55m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/842b5731d7ef42a1a1fa6f9424fda69e : SUCCESS in 1h 18m 06s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e40d37e3f5764223a5eb0a0b90e670b4 : SUCCESS in 1h 08m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a4c36c6bdae24e45b35366fcb23f5bda : SUCCESS in 1h 23m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2513ad4bbb5e46a8b6f1e528e0dd02d4 : SUCCESS in 54m 31s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/427c4a48234a4e6b91cc91ecfbb38fdc : FAILURE in 54m 42s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/393f86a006804cea9bff33ab048c20f4 : FAILURE in 57m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d8a198f64ee4e739c3ae92e37e4daf8 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91f563e4bfc544538c7f7de20f4800b1 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e51a66ab25d94347b4743928ce393bf5 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/332377868a5e44cea1d414cada05fe20 : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7283dd549be47989a0a2289554a12db : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2107b15d3bd14be385832502ed7dfcd1 : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cea1344a90c64c95acd63a5aa1c2f816 : SUCCESS in 26m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9daf5566af0406c89db16f1ba795b0f : TIMED_OUT in 3h 01m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fd8123dc6405430ca04eddc76a4a1168 : SUCCESS in 1h 03m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/296ace94d36d43868663a3222e40a8e2 : FAILURE in 33m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/87b84dbda5a94b069fe5cc936d7023e7 : SUCCESS in 1h 06m 46s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"b2565f149d3591ccf4dbcf6ea72c5ad99a07627e","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-06-09 06:53:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6019d3d9de261a15c9c61dc73a64ed7db4407e9c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-09 07:44:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8f53ca51dafda61f62f314feccf66d20acece50e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-09 08:34:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4f29f6af77a06568d95fa9a878894d7b22e4ab1b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-09 08:34:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nwe know we need this one","accounts_in_message":[],"_revision_number":3},{"id":"076f7083c33ee6e63f4499a0a00e7dfacc4d8038","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-09 09:50:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a58e876d2d1a2940c3720937f3a1c62f0f0d308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:24:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cc0cf68375dd4142a971d67c4c29e93a : SUCCESS in 28m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/09d0ba707551467e834265266a8d3bfe : SUCCESS in 58m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c8e548b557dc448cb93b0fdd34e8df55 : SUCCESS in 1h 18m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a677ed7e90e04361ad86a7d2a25f61a5 : SUCCESS in 1h 38m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/62fb2a0afcef4346aaaba4c861a1d22f : SUCCESS in 1h 30m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bce2c7438d91462198770560da25c8bf : SUCCESS in 57m 47s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/be75d88c77a54d0abb48d0bbf8ed5462 : FAILURE in 50m 52s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/200e9846d0e343bc969e2e10fe99cfcb : SUCCESS in 1h 39m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec839433fb2448aa970653420f5653e1 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c481bf9849a4202b06f511012a1a1d0 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2fd01bacdae44aca8d0aa7f6aa2f03fb : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9309dd287d15426cbcc64a03d97b8f1f : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58f8e140920540a3953e0a8eed856862 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e775f8b0be427a87014805514af44f : SUCCESS in 3m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e6959f5a2b34d14a89fa83ccae9f5a5 : SUCCESS in 22m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f813c7665a748a5b86ac1f2268e5635 : SUCCESS in 41m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c2902026ea1a43838a4e8646bad8a1e1 : SUCCESS in 1h 07m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/44123aa4e1e2457abfac4db884eb9862 : FAILURE in 42m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/63109a8e692d41a6a3bdfc8940e1dc1f : SUCCESS in 1h 03m 20s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"558db5319d7da50d099cd1897ed3fa3de91e2653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:24:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a50b128143f4cd21e353503e0766fba7b3e8f313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 11:30:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/df4c690c427447a8b4239e66853858e7 : SUCCESS in 33m 06s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/39882478f2f24d5d91f92052a293d1a6 : SUCCESS in 33m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0936a6331cd649dc9659364843f21b8f : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e386db786a44b484d4a4107f6369e6 : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76b5379e2af14df6b3c3a0bd30aed8e9 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8212834f01248abb111381472929fc6 : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54b8a3abea724741aa38bc5d528fdc46 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e1706677c4a4be4876c37ebce990f51 : SUCCESS in 6m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf491dae73d94460adcd7c69a60e1b83 : SUCCESS in 40m 28s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"19ac72ca67af5e69e1f76be8de9752818a83d5b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 11:30:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"01e23d562154f7067b26211407a443bd47719ffb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 11:32:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa938286cac44e488e1e97f961b16703 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fca4a088a49bbd8ee7ed6ce8b0b128df7b62399f","revisions":{"b0436bd728497b829d03e220822ad0d39b2c1b85":{"kind":"REWORK","_number":1,"created":"2020-06-08 15:42:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/734131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/734131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/734131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/734131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/734131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/734131/1"}}},"commit":{"parents":[{"commit":"4b77682911616a6cfac96df3695f0d7896f89b50","subject":"CI: Ignore more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b77682911616a6cfac96df3695f0d7896f89b50"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 11:05:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 15:41:53.000000000","tz":60},"subject":"Fix multiple issues","message":"Fix multiple issues\n\nBoth issues affect Train and earlier releases.\n\n1. Fix monasca-grafana by pinning more Ruby gems\n\nchildprocess 3.0.0 and ffi 1.13.0 depend on Ruby 2.3.\n\n2. Bump bifrost to 6.0.4\n\nBifrost had a fix to stop using the master branch of DIB which dropped\nsupport for Python 2.\n\nChange-Id: Ida7e20833360bbca69c6aafc4f5d7cb375bb106f\nCloses-Bug: #1882070\n(cherry picked from commit 3784d32d9ceacf0dad8057092752f1ae7be37ed3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0436bd728497b829d03e220822ad0d39b2c1b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0436bd728497b829d03e220822ad0d39b2c1b85"}]},"branch":"refs/heads/stable/stein"},"cb1b2ac6aed54ccb55b60eddf64960c5778a4f01":{"kind":"REWORK","_number":2,"created":"2020-06-09 07:44:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/734131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/734131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/734131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/734131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/734131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/734131/2"}}},"commit":{"parents":[{"commit":"4b77682911616a6cfac96df3695f0d7896f89b50","subject":"CI: Ignore more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b77682911616a6cfac96df3695f0d7896f89b50"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 11:05:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-09 07:44:24.000000000","tz":0},"subject":"Fix multiple issues","message":"Fix multiple issues\n\nBoth issues affect Train and earlier releases.\n\n1. Fix monasca-grafana by pinning more Ruby gems\n\nchildprocess 3.0.0 and ffi 1.13.0 depend on Ruby 2.3.\n\n2. Bump bifrost to 6.0.4\n\nBifrost had a fix to stop using the master branch of DIB which dropped\nsupport for Python 2.\n\n3. Drop docker-client from OracleLinux sensu-client image\n\nThe package fails to install due to a missing dependency on\nsubscription-manager.\n\nChange-Id: Ida7e20833360bbca69c6aafc4f5d7cb375bb106f\nCloses-Bug: #1882070\n(cherry picked from commit 3784d32d9ceacf0dad8057092752f1ae7be37ed3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb1b2ac6aed54ccb55b60eddf64960c5778a4f01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb1b2ac6aed54ccb55b60eddf64960c5778a4f01"}]},"branch":"refs/heads/stable/stein"},"fca4a088a49bbd8ee7ed6ce8b0b128df7b62399f":{"kind":"REWORK","_number":3,"created":"2020-06-09 08:34:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/31/734131/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/734131/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/734131/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/734131/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/734131/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/734131/3"}}},"commit":{"parents":[{"commit":"4b77682911616a6cfac96df3695f0d7896f89b50","subject":"CI: Ignore more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b77682911616a6cfac96df3695f0d7896f89b50"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 11:05:12.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-09 08:33:58.000000000","tz":0},"subject":"Fix multiple issues","message":"Fix multiple issues\n\nBoth issues affect Train and earlier releases.\n\n1. Fix monasca-grafana by pinning more Ruby gems\n\nchildprocess 3.0.0 and ffi 1.13.0 depend on Ruby 2.3.\n\n2. Bump bifrost to 6.0.4\n\nBifrost had a fix to stop using the master branch of DIB which dropped\nsupport for Python 2.\n\n3. Drop docker-client from OracleLinux sensu-client image\n\nThe package fails to install due to a missing dependency on\nsubscription-manager.\n\nChange-Id: Ida7e20833360bbca69c6aafc4f5d7cb375bb106f\nCloses-Bug: #1882070\n(cherry picked from commit 3784d32d9ceacf0dad8057092752f1ae7be37ed3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fca4a088a49bbd8ee7ed6ce8b0b128df7b62399f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fca4a088a49bbd8ee7ed6ce8b0b128df7b62399f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
