)]}'
{"id":"openstack%2Fkolla~734632","triplet_id":"openstack%2Fkolla~master~I6f04ee8243747ffeccdfe61e664e1aa6b1f9c78f","project":"openstack/kolla","branch":"master","topic":"remove-sensu","hashtags":[],"change_id":"I6f04ee8243747ffeccdfe61e664e1aa6b1f9c78f","subject":"Remove sensu images","status":"ABANDONED","created":"2020-06-09 15:46:05.000000000","updated":"2020-07-08 12:13:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"26ab797fe1e3e04bdb023d7bdebf9c30b67da70d","_number":734632,"virtual_id_number":734632,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-09 18:03:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-09 19:59:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"date":"2020-06-09 19:10:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 18:03:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 19:10:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-09 19:59:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fbaea53cc173a7cee1624cc491d5f80435c859d","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-06-09 15:46:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69de665602d18d84affdce611b87578623fba2d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 17:58:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/047a39d83a5e41729599b6e7911fd6c2 : SUCCESS in 2h 02m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7719ea4c73b4a52a230cce4a961cbe2 : SUCCESS in 1h 55m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ceb10e2082b4e25b6256365f148f9dc : FAILURE in 2h 11m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab08e967b6724afd839e2c2558852068 : SUCCESS in 1h 59m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e488ad8d543d5199c63fb432c9b6c721fbbd6eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 18:03:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d9270ad9e9994d88919d912a14c2e38f : SUCCESS in 3m 38s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8c3344b8205a41a99a93e9015d53446a : SUCCESS in 37m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/60fbcef2f9f44098817cbab40d55d804 : SUCCESS in 31m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/60d2145324054e16b161070d03b48494 : SUCCESS in 1h 23m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5a584010d4cf4526a9e0e51053bfbab2 : SUCCESS in 1h 01m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4ebe679b354d4517b1871cf8361689b0 : SUCCESS in 1h 01m 30s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/cd27015126d646309647b2a0cf44752a : SUCCESS in 52m 08s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/359e3307176c4b3f8c7d4f3cc9019f66 : SUCCESS in 32m 00s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1974a467d6ea4941a8886e88ef6f3575 : SUCCESS in 2h 02m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bf6ff24631f74cedb78d2a908adafe83 : FAILURE in 30m 29s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/8380387839cb48519663688d08142501 : SUCCESS in 2h 09m 24s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4d14151267724942a00e8926aef4907d : SUCCESS in 1h 56m 41s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/04b505b1e71948399aa653c3ad7991b0 : SUCCESS in 1h 21m 31s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/444792f815cd4b29957157df86a15da8 : SUCCESS in 1h 52m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/264ccff7758345f4bc6e73c243aa2aec : SUCCESS in 25m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/29cd7040704f4eca9d74637ef1fcd927 : SUCCESS in 27m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d474c9dc80794123a1a3a09375534aaa : FAILURE in 1h 02m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1abf596f7f7422dbdb4c698bb86eb1d : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0048c91141a24e7a80ed27624041ab5d : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba9da0f2733d4a449837d078e380acc4 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b1be1baa7044ecc9534c25d167ea8f2 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a837f16bb4420993e03b26272476a5 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81bd88c3ee9e46fd877dd3ca0f3a080b : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ce5bd52cc084955876d6736705715eb : SUCCESS in 40m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/091b7a2dd8a6405696246b628c865472 : SUCCESS in 29m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23d2f327451844a08582d8b6a5ec3c29 : FAILURE in 41m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ae5cbb6cd664477b57db680ad452f0b : FAILURE in 39m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95aa8006fffc4490ae0a91f544e00ffd : FAILURE in 1h 17m 32s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1fa15a2c0dc34a31b46bb611fa4319a1 : FAILURE in 55m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3c9332255753487590a1bcc335bbbae5 : FAILURE in 1h 42m 28s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8df97cef69ad4bec926c427f9bbc9d25 : FAILURE in 48m 04s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/21d741f59b1744f98ff1ff346726a397 : FAILURE in 42m 26s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/404a531b9aa04b39a68631165d02fab0 : FAILURE in 49m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d13064a5173d3c9f700debc3db612f8e84343f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-09 19:59:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/0d957105eaf44052983a8214761a159a : SUCCESS in 30m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d509e21899b0700c46317dd4b448d9c7b8e84ec8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-08 12:13:20.000000000","message":"Abandoned\n\nSensu got removed in https://review.opendev.org/739440","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3cb466d04aac6db3ce8d9a160156571d75f769f","revisions":{"a3cb466d04aac6db3ce8d9a160156571d75f769f":{"kind":"REWORK","_number":1,"created":"2020-06-09 15:46:05.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/32/734632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/734632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/734632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/734632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/734632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/734632/1"}}},"commit":{"parents":[{"commit":"8a42c20c47e158cc96ad78f951ae3589c2afbe54","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a42c20c47e158cc96ad78f951ae3589c2afbe54"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-09 15:45:06.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-09 15:45:48.000000000","tz":120},"subject":"Remove sensu images","message":"Remove sensu images\n\nChange-Id: I6f04ee8243747ffeccdfe61e664e1aa6b1f9c78f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3cb466d04aac6db3ce8d9a160156571d75f769f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3cb466d04aac6db3ce8d9a160156571d75f769f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
