)]}'
{"id":"openstack%2Fkolla~734633","triplet_id":"openstack%2Fkolla~master~Ifde221e807df75d884a3f8cc4fc29eb0cb303472","project":"openstack/kolla","branch":"master","topic":"remove_neutron_server_ovn","hashtags":[],"change_id":"Ifde221e807df75d884a3f8cc4fc29eb0cb303472","subject":"Remove neutron-server-ovn and neutron-ovn-metadata-agent images","status":"MERGED","created":"2020-06-09 15:50:48.000000000","updated":"2020-07-27 10:56:55.000000000","submitted":"2020-07-27 10:54:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734633-1595847292441-7a5aba1f","meta_rev_id":"3e35f6c19844b013bdbacc1eb76050a234b5fd04","_number":734633,"virtual_id_number":734633,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 10:54:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 14:00:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-07-27 09:12:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-07-23 13:05:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-07-27 09:12:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"date":"2020-07-23 11:17:24.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-14 09:41:42.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-06-26 21:04:45.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-07-23 11:17:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-07-23 13:05:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-07-23 14:00:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-27 09:12:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-27 10:54:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e963fbb64459aa75f38a1375e8cb3799581cedf","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-06-09 15:50:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7a583a2379c7d75ddbcc17028849f610ad982a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 17:23:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ebf9a558e7a4406394965622b3285aa2 : SUCCESS in 27m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5a5669cb42d74ef3a92b5c4bf310efc8 : SUCCESS in 30m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7313fa4c976b429daa9ea4994513c833 : SUCCESS in 1h 26m 39s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/684d262f85ad48238225720a55e827c5 : SUCCESS in 1h 22m 10s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e8021efd87cc409ba9048b3c8f94b8c9 : SUCCESS in 1h 00m 19s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/92b64815f41041b4ad1b59ccf9f635d2 : SUCCESS in 37m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9ac2726dee4a4095bfec661fa44dcddb : SUCCESS in 16m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/975525a56dd34275b093e0843625c1d0 : SUCCESS in 20m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a0503777bce7444fb98c7335cdd9adc8 : FAILURE in 52m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eef68e87c9df4558b622e3ce0dc0a3da : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9818a1fdbc24099a215515c82d0ab75 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9393d8bb3f04cc39f387510bcb2379e : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f29a81b90b634283ae8714d774121a18 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246db7e5c406428a92cb9f363b2824d8 : SUCCESS in 7m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d6aff59cf5b9456394d722d89f4d6273 : SUCCESS in 33m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7fc6cdb43774affa8624ca9704ca18a : SUCCESS in 21m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23fbf15a876146ddaf02b93b5043f383 : FAILURE in 35m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/040f797baf2d4aeab7518a78f37b5b47 : FAILURE in 39m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/89cdbe0b01fc46d0b9dcebf0507bf765 : FAILURE in 1h 08m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c43434bbb07875ba0737e92e3c74f9b148568dd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 18:10:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f36bdef0af5423ba1337c62b76293f1 : SUCCESS in 2h 01m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f367d3dc942a456b90827c1097e9a6ca : SUCCESS in 1h 55m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9de9957fc0a4a4bb08fc30260fd0fb8 : FAILURE in 2h 13m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad7c4a90caa945fe96c75a9ee5574874 : SUCCESS in 1h 58m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b2fe8c718d0ac8d6bf4c01efe0a478088066a54","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-09 18:43:41.000000000","message":"Patch Set 1:\n\nhmm, tripleo is seemingly happy about it; would like a reno here","accounts_in_message":[],"_revision_number":1},{"id":"e1b1140d65b440d9faa8a3d866ea07073f60b8f1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-09 20:01:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/6736966211094e40a7ba384280277f9c : SUCCESS in 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ce19ec203002385ec998cdccdc90724a215342e","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-14 09:41:42.000000000","message":"Patch Set 1:\n\nChristian, could you please offer more info ahout this PS?","accounts_in_message":[],"_revision_number":1},{"id":"6654c73db4c9f03e89abdfdeab4ffd19d3af297c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-26 21:04:45.000000000","message":"Patch Set 1: Code-Review-1\n\nadd more info why it is removed","accounts_in_message":[],"_revision_number":1},{"id":"63bc91d16e213adf7fe23265dbda138cd1a4b93f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-23 11:17:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8f4b0733d1d420b26b80fded0adce9933961dc04","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-23 11:43:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/21a9fc1d76b4402284f64948615fcc87 : SUCCESS in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0ec683c28f627503abc95b8261c0e02393f6b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 12:45:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/14623576fdb8451eb2a6aad64972d686 : SUCCESS in 22m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0c33b63998aa48debfd194dacca0ef87 : SUCCESS in 59m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/791f4eb54f664665a8b73ab16960449f : SUCCESS in 1h 22m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c2f45dc0cdaf4e86bd62486fc38a1301 : SUCCESS in 1h 14m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7aa4418772764801bce127b332bb967a : POST_FAILURE in 34m 12s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/84e23f37571f4afb8eaa4a74627bcb6a : SUCCESS in 48m 40s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1d6768d484f241bd86b2ea40256bc190 : SUCCESS in 21m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d18a035db1fe4734a80c39ad2be4ee29 : SUCCESS in 18m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f693567fe8943a99ea1f5c35d22b6d1 : SUCCESS in 55m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e59f716ead3e4410a1305b7adfebb093 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b3872650e14af199b5526a97d0d784 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b61084f39cb4dd0bbcb2ec08e9c99b0 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c019be9d88664c7dabcb0afcf4c947b9 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8bbb62d50af4480b512be447c57dc45 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33efad96676246a69899194452331d9f : SUCCESS in 3m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3724ce5db1af4c86ac6bb1652f00ba14 : SUCCESS in 21m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8edb5db813c4845bac72554352fcda6 : SUCCESS in 24m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de907cc8d1e240a1a55e3dce3040b2d5 : SUCCESS in 1h 03m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a76302d4c7ea4257b3cb87ebd588d53b : SUCCESS in 57m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da24ff179fa24e998a50d5a7f491c32a : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":2},{"id":"82cd2c8472836a16e82b4d4de6e324daaf8741ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 13:00:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ae6067dafd97451781967becbf02450b : SUCCESS in 1h 07m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a111126bbec46738358e6cca456949d : SUCCESS in 50m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ceeb5b111284026b8c98df4ea34a30c : SUCCESS in 1h 42m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/52a98fa7dd33437aa03efc0e9a8df163 : SUCCESS in 51m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6bec7d248bc08bde37357b9e62d454620b273a09","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-23 13:05:00.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"89ee0b30b21a57f8c869dac005e835c18ae3dd50","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-23 14:00:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ede810dad95f46f68c78a54a5a197cde : SUCCESS in 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a2aa5bbc35404f68335f37556ebc4af22ab851b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 14:34:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/07237cbd3ff349fcbd833161c76f753d : SUCCESS in 22m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9fb27406678448fe8916da3353eb2b42 : SUCCESS in 26m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d8240342d71c481caf22d4c915c90e9f : SUCCESS in 1h 18m 44s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6b26317698ac4236b3103fbef4222324 : SUCCESS in 1h 07m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9e6562b32972420987f88a19ba5871ee : SUCCESS in 1h 09m 10s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b87c5d7538404e4e8f9ce2d6d571eef0 : SUCCESS in 43m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3a47439a9ddb4e268f87bbe473c6f4ca : SUCCESS in 19m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/836406ff403b4635a1908f28d09fbc4f : SUCCESS in 20m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/19792a1aa04949c19bc3b7ad421481a1 : SUCCESS in 1h 23m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fb8b95af3c94ab3ab609abf60787b5e : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a05855172b0423982616232eee41554 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6372996e92c944639d4ef0f071ca9406 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca574f8e26be44a7aa7f25827166815b : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a3fe16b11643c6ba79685595c81782 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27b0a5fd86344da1850ad31f26aa97b4 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6cba6ac6268945bb9ef6c81f1ec0bbe7 : SUCCESS in 34m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bfce2428ad59463680478d89ed4edc1b : SUCCESS in 23m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0cd37bb94a44c6ea2a9c9c4281b4d14 : SUCCESS in 1h 08m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a07066a7a3c47079b3a38558b13ca06 : SUCCESS in 1h 01m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9935c7f0860435b91c18ff32436ea43 : SUCCESS in 1h 10m 27s","accounts_in_message":[],"_revision_number":2},{"id":"75ca1207a4c2e4d0fb277fb575f79c5b78d80d81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 14:53:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/793460e58fa54713b81fab8a9dcffc1b : SUCCESS in 58m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/963289c9ae114a1bb13803deee51f1a5 : SUCCESS in 48m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38ebb3f278354bf79605f8a6ac6011e7 : SUCCESS in 1h 47m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/144dff72c7e7407eaf3a10c51c590cec : SUCCESS in 43m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3495fbe514825eda2beaadd063e59b3df1f3be8c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-27 09:12:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"43249072485c6c44264850f188d2958af7102f13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 09:12:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"520e62b50082acab335544a799dc1dcd16195544","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 10:54:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2b33200a0377466eacc97e37a467e45e : SUCCESS in 34m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f808cdc0faa346ad9ebae524fa2b1319 : SUCCESS in 1h 18m 26s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7a7320d82306456cb54b3970d51a959d : SUCCESS in 1h 39m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc93fb0cb844485c99a448ff5ec4acaa : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6afdb2ca8b4aa58a765d9c661d974c : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9967a084bb9d423083d611cc8f1fdab7 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27c198593b574fe4921fb97cde79de7d : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2037f5779f2a4706a2d9958b978cbd36 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c87f23757b644a87a4f6c7ae2f3b9089 : SUCCESS in 5m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f5af21f1c6b42aab68356c6ab9d02f5 : SUCCESS in 22m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/77b4ff997271418491315c1b2dfba79e : SUCCESS in 49m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4343f258f4034d4c91dad1fd8865db56 : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"bbcbf4837782978166d2c0427f97715082211e11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 10:54:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"71f8483d66e282fc4a0ef019a19ff5e75b939c21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 10:56:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d157c4dca3b45a19914e14ec3f39e16 : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97a92a36d412410480646e808b70459a : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7f92df4eb045072b95ee226e642db13a36eead3","revisions":{"72af4701bfc2884c1128cc064dae27698e9f72ad":{"kind":"REWORK","_number":1,"created":"2020-06-09 15:50:48.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/33/734633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/734633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/734633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/734633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/734633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/734633/1"}}},"commit":{"parents":[{"commit":"8a42c20c47e158cc96ad78f951ae3589c2afbe54","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a42c20c47e158cc96ad78f951ae3589c2afbe54"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-09 15:50:27.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-09 15:50:27.000000000","tz":120},"subject":"Remove Neutron OVN images","message":"Remove Neutron OVN images\n\nChange-Id: Ifde221e807df75d884a3f8cc4fc29eb0cb303472\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72af4701bfc2884c1128cc064dae27698e9f72ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72af4701bfc2884c1128cc064dae27698e9f72ad"}]},"branch":"refs/heads/master"},"a7f92df4eb045072b95ee226e642db13a36eead3":{"kind":"REWORK","_number":2,"created":"2020-07-23 11:17:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/734633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/734633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/734633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/734633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/734633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/734633/2"}}},"commit":{"parents":[{"commit":"acd0d4bf20e3b103c9a519f0f013bc9dbd67d973","subject":"Merge \"fluentd: migrate to td-agent 4 where possible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acd0d4bf20e3b103c9a519f0f013bc9dbd67d973"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-23 10:40:03.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-23 11:16:49.000000000","tz":120},"subject":"Remove neutron-server-ovn and neutron-ovn-metadata-agent images","message":"Remove neutron-server-ovn and neutron-ovn-metadata-agent images\n\nThose have been deprecated in previous cycle - see [1].\n\n[1]: https://review.opendev.org/#/c/701074/\n\nChange-Id: Ifde221e807df75d884a3f8cc4fc29eb0cb303472\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7f92df4eb045072b95ee226e642db13a36eead3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7f92df4eb045072b95ee226e642db13a36eead3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
