)]}'
{"id":"openstack%2Fkolla~735438","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I3e307efc43f934a944a91d5d131a11f607411df2","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1882863-stable/ussuri","hashtags":[],"change_id":"I3e307efc43f934a944a91d5d131a11f607411df2","subject":"add libibverbs package for openvswitch","status":"MERGED","created":"2020-06-13 16:51:38.000000000","updated":"2020-06-16 09:21:49.000000000","submitted":"2020-06-15 20:57:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735438-1592254626560-d12a0952","meta_rev_id":"85a1754078594ba697cf45402a8d7739739817ab","_number":735438,"virtual_id_number":735438,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-15 20:57:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-13 18:35:12.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-13 16:51:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-13 18:35:12.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-13 16:51:38.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-13 16:51:38.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-06-13 18:35:12.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-15 20:57:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af8f3e84853a6eb2d01be13d9ab74b15cd4c05e5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-13 16:51:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d94e0fabcab1990292a89740296bcd29f3a46f08","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-13 16:51:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b6badc9e2c7cd41697d0e72268f709f4e9e45900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 18:11:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/766d9cc00508446e8b9cbb31d8c1938e : SUCCESS in 17m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1d9b7aeb67784493bb62e17526511b4f : SUCCESS in 27m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e5f32c8aacea48a6b688e6cd4e9b3033 : SUCCESS in 1h 15m 33s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fb6864fefae4401ba151d6a5469f3be1 : SUCCESS in 1h 15m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c64114900ae346109b85b6605d29fdb0 : SUCCESS in 1h 12m 03s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/463f15f786af4b30b56d6a7d43cc5e70 : SUCCESS in 45m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/86a3cba0e1004e6d8f76fef90ae8e93c : SUCCESS in 20m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/15dcf07e4ca24fc2ac66ae5cc69a02af : SUCCESS in 20m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f4d7561f5174c76b02f835376f9372b : SUCCESS in 53m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a00c02e12351492896af286c678fbcb7 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b451a0381474bd6aff76d4063b3ef2c : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/016640efb09848bebdf29466854e5df0 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22b95d575cbf49e9bfc44f7a78b5abc8 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/141167bcd174487298f8976528ff5a6b : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4212ca1484604f869b0374a5da9d1c08 : SUCCESS in 24m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/328e75bd27b94ca9ae4b8d701576384d : SUCCESS in 23m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a4a355b2b0c48ea89f5e13bb59f1671 : SUCCESS in 1h 07m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae8c2ad904fb44e1a2fbb9b312ba8a54 : SUCCESS in 59m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37b6213828d641d08e40ffab75d70579 : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3dab359f3feddc95189ff1f2859b04de24d45e2c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-13 18:35:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50696cbbc455443e22f67c0f3243672217cf3295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 18:35:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e15ee369d1b33e15f9eb9ce5ec5de7c878d12a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 18:59:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5885b1a4dd0d407bbd5f159190a6e0d8 : SUCCESS in 1h 08m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/646332a011194670bbed2e7aa672c83e : SUCCESS in 1h 07m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d1c3d27da0344e15b94d6e64334d0058 : SUCCESS in 2h 02m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca19931f1ff64f9e8ed7275b9dd9a055 : SUCCESS in 1h 07m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c8f34e913bb3ace8795e5ab2a28ecad4b82e055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 20:45:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/33f58022ac7745a3a6486b87a09b9832 : FAILURE in 2h 08m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e660f3a8745e4a2e81f7713a219dc1c3 : SUCCESS in 1h 52m 39s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b8784a0d2a6b4ce99185c39411d7da90 : FAILURE in 1h 51m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f7c3422c5d0494f9f8f69bc91ebfa0c : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6177ddfb43b740b1ba31e515d3783c2a : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8d40b57b0bd4c9ba72c9b30dde0814e : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5dbe5ea6ebe4a8cb0857260cecb0b70 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2813432bb05e4833949eb16215bb5d64 : SUCCESS in 7m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d57391014e7a470aa2123767d3ccb064 : SUCCESS in 46m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a15cb2998ba406aacf1dd02bdcdaff2 : SUCCESS in 1h 18m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/417f198805654517b7af2ae73ec3aa1d : SUCCESS in 1h 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3e4ca8ecfdc8fa0bfe0914e46d12c40d74674678","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-14 07:32:54.000000000","message":"Patch Set 1:\n\nrecheck\n\nmirror fail","accounts_in_message":[],"_revision_number":1},{"id":"186977e6762046e8d469da5215fd40fa42c021df","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-14 07:36:01.000000000","message":"Patch Set 1:\n\ncentos mirror is gone atm","accounts_in_message":[],"_revision_number":1},{"id":"069472fb9e0876c50859a3052fc76d002ce72bf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-14 08:04:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/4fc73ca137f64e0b99e5cc76c7965b45 : RETRY_LIMIT in 2m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d575a2c87ee4433387b9db48fa752e98 : RETRY_LIMIT in 1m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea614ab4a81f4b7990b499e0a1161064 : RETRY_LIMIT in 1m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8167b6788aca4b3ab4952232cb5b237f : RETRY_LIMIT in 1m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fc11ae78b4e482b7eb23febd05cc30fd5f062d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-14 08:46:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f8e4e01d8e894ffabf2d5d13f8c3e94c : RETRY_LIMIT in 2m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e4f4f12140b14d4baa7da70c19b1d4cd : RETRY_LIMIT in 2m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/37ad8826e0fd4f31ada544c420fc9f18 : RETRY_LIMIT in 2m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/073b026f51414c7eb6b800f369873a57 : RETRY_LIMIT in 1m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/25a6a34b4655406aab204ec0035816e7 : RETRY_LIMIT in 2m 34s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2109fa766e1a462da1205d936f4f6b39 : RETRY_LIMIT in 3m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6b2509ff172141f5b48088660140a48d : SUCCESS in 25m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7e269a3f94cb4d01b597a5673e0bfcf7 : SUCCESS in 18m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad974f727ee64abfb98366ecb201c72a : SUCCESS in 1h 07m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/830569ebd04745269d93acd3f4d6dd02 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11eceb312204450d8cd8873d9d014fc4 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1451adebc83a4679b2e3fff5cdff5d9c : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acad485a73934d549ad4dc804e64a9c8 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f78dbfb650e843a389050e14f1d79f90 : SUCCESS in 5m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c45c90c14844989aa06eb42b7c23edc : SUCCESS in 24m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6fa23aae8e8141af932a7505facb1f19 : SUCCESS in 26m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/460a632719374d8bace5050baba5a18b : SUCCESS in 1h 03m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/60f8f66e2b164bdc86598120a0e986d5 : SUCCESS in 1h 06m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/03c10f3ac1d640c98c4e97694fa06882 : SUCCESS in 1h 08m 44s","accounts_in_message":[],"_revision_number":1},{"id":"da7bd4e2609fa708865fd9c882f7a3316160df57","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-15 17:24:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b067ac1149adb8a28fa5edce5e7d2be2791b947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:13:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1660910ac43d4faea12262b80f19a95c : SUCCESS in 45m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4e793b63c89c476a80b2b5b0ee1f9510 : SUCCESS in 42m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/30566e376e3e44c39bdc800f9322be26 : SUCCESS in 1h 30m 37s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/90120bb4a54f4031b1fb71d51fc25304 : SUCCESS in 1h 21m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1d2f47e82f5f43fa952f4cc738229b79 : SUCCESS in 1h 33m 57s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3b795dbac645468aaeb93c731e1683d9 : SUCCESS in 52m 52s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d6c1352091da4631a90e67058106afeb : SUCCESS in 24m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c3def4bfbfaf4429a09253b37783ff16 : SUCCESS in 27m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b20c6a4943194674a7228e85f88c990b : SUCCESS in 1h 17m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48a884692dea40b888e3d2b2c01ad15a : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a684e7200e449abf8dcb98c4f51c48 : SUCCESS in 9m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3b6e7b48ebf4d5ba919cb85d24368a1 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19ed497db08945688c4c0510df10d962 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62dd4a95cb5424b886cf177772ebfe4 : SUCCESS in 8m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2371964d9522461584c1bcc6c59a5a85 : SUCCESS in 22m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe33483dd2f34eb4a066ce47afbb8d58 : SUCCESS in 41m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3cbfe644301a49328c60144d57e01dc4 : SUCCESS in 1h 15m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf18ce60fc89413e92ddc3b54b6d7959 : SUCCESS in 1h 05m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5576d5f7eb24f04a39251d2d6e03ec4 : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":1},{"id":"40cfdab49deb865c311bbaa90d8a765dcfd653b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:13:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b2b3a5de3e2d158704232eb5bb76b96768e73a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 20:02:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a138c03d5a6444db96c24c436a23592a : SUCCESS in 1h 33m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f08be6dc0f3a497ea351dd157e32adb1 : SUCCESS in 1h 25m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c8e2f6fb1d4495487975f1ce8a1857e : SUCCESS in 2h 20m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c795090335dc49f2a68d1b1a4763e94c : SUCCESS in 1h 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a589496630b44373bfc2e30a22b0f49c2bb423f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 20:57:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e6a3aaf02d374379b44ef5d1aa976b1f : SUCCESS in 33m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/30f5baf7ecf04257aa30d977c734bd37 : SUCCESS in 1h 18m 03s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c01940f0abf04286a59ff80aa3c42897 : SUCCESS in 1h 27m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb394b77462a44459ce7ec7b8fb1891d : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a864b5845a46bf9c980c7051bb701d : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91b60d8b11bd4a0cbf06dc5a66d6a228 : SUCCESS in 8m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ead50d9c458447ed8216997c343f213b : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d07e31c8ba645199613f282c9d93596 : SUCCESS in 9m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac7223d6c7c043a8bc64449b8adf4de4 : SUCCESS in 21m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4a2a4c36ee8481697c7012aeaebd093 : SUCCESS in 59m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7837611cf0e74c62870325b4ccc355fc : SUCCESS in 1h 34m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c1f2476a81273303860bab8fa5a54dfac6874407","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 20:57:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5db99dd81fbd07fb73bd86eb7410cd6db1a0003e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 21:01:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4fe2e728644d8eb0e2a713197ca88b : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"deff1a63b75aba8ffe0bb661589a93101bee6487","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 09:21:49.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9d821722f2779c0c747a2fa317f3eca3e60dc7cb","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bd7d3fc7adde6d0f87c6dcb0ce497fd94bf0601","revisions":{"5bd7d3fc7adde6d0f87c6dcb0ce497fd94bf0601":{"kind":"REWORK","_number":1,"created":"2020-06-13 16:51:38.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/38/735438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/735438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/735438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/735438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/735438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/735438/1"}}},"commit":{"parents":[{"commit":"35c81ba99177f91c1bc5f7cbd4f353179375991a","subject":"Merge \"Add releasenote for congress deprecation\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35c81ba99177f91c1bc5f7cbd4f353179375991a"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-06-10 06:46:06.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-13 16:51:38.000000000","tz":0},"subject":"add libibverbs package for openvswitch","message":"add libibverbs package for openvswitch\n\novs images which based on centos miss libibverbs package.\nthis ps add the package to neutron-ovs-agent and ovs-base\nimages.\nCloses-Bug: 1882863\n\nChange-Id: I3e307efc43f934a944a91d5d131a11f607411df2\n(cherry picked from commit 6d71ac843e85126bfe35c9a39d64c1767d360c74)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bd7d3fc7adde6d0f87c6dcb0ce497fd94bf0601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bd7d3fc7adde6d0f87c6dcb0ce497fd94bf0601"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
