)]}'
{"id":"openstack%2Fkolla~735775","triplet_id":"openstack%2Fkolla~master~Ieb88fd60d177f19a635ada9366193e57c516f90b","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ieb88fd60d177f19a635ada9366193e57c516f90b","subject":"Add collectd-amqp to the collectd image","status":"MERGED","created":"2020-06-16 06:25:31.000000000","updated":"2020-06-18 10:47:17.000000000","submitted":"2020-06-18 10:34:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735775-1592476485489-e09c6c95","meta_rev_id":"7df8004842d8904f88115d3cb0d10ff80c3ffd8c","_number":735775,"virtual_id_number":735775,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 10:34:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-18 07:34:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-06-18 06:22:58.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-18 06:22:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2020-06-18 09:02:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 10:02:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-18 07:34:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-18 09:02:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-18 10:34:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bd7330bcec992b49df5e05c93c84ca3f78d41d1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-16 06:25:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a785f925cfbcf0dbeedbc1a6a6f09125915e516f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-16 06:47:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7f189471087144af9722c6d7899fb795 : RETRY_LIMIT in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adeaa1d62772e5b869921e02aa87575674b8dec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 08:18:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4ceae0e46bbc4de2ad03448aa124bfc6 : SUCCESS in 47m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0051861fa6bd4e1d8d745dbe5515dd13 : FAILURE in 41m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/41319ff8ff334dadbafc727f62245353 : SUCCESS in 1h 16m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2b8abdfd69224eedac672acc76a8ddf3 : SUCCESS in 1h 18m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2c2bf453597f4492920b274a30e5da44 : SUCCESS in 1h 39m 19s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6e336afb36a94d8c8f5a792b5eadfbd3 : RETRY_LIMIT in 7m 18s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f98efeb2d23a4c1ebebd900134ea96af : SUCCESS in 20m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9b0b4848e6a24ee39403f8957c5b5830 : SUCCESS in 21m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8d5772033a1d43efba46f3b02df90de4 : SUCCESS in 1h 06m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b57c6769bec464aaf2765574ca6ece8 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f56d66e4c5f4847bb4989fcdc68b3fe : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb58dbb502a44271930f258adf1972e0 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/560a4920b67a4fd8aeeae181add11be4 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a149e3e99bf34d6c82c45f8327a3455f : SUCCESS in 9m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5356b678982d4b75a220d7d7277b2413 : SUCCESS in 24m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c24cac0e19a04380b6a37443a2e4b35b : SUCCESS in 25m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/203e6fa5a2bc46a9ab3f31d11643e3f2 : SUCCESS in 1h 03m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dfbbf9eee7e44249958ebe4c20cd2db8 : SUCCESS in 1h 06m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2534f44fc098468b80833aa8584f48a1 : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ec74f6771c3796f3a92713d4474d0cf6b6c18056","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-16 08:34:11.000000000","message":"Patch Set 1:\n\ncyborg fails: \nERROR:kolla.common.utils.cyborg-api:Error\u0027d with the following message\n\nERROR:kolla.common.utils.cyborg-api:The command \u0027/bin/sh -c dnf -y install mod_ssl mod_wsgi     \u0026\u0026 dnf clean all \u0026\u0026 rm -rf /var/cache/dnf\u0027 returned a non-zero code: 1","accounts_in_message":[],"_revision_number":1},{"id":"6e7e31681aef1ab4e95000365c84079dbd51ee0e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 08:35:26.000000000","message":"Patch Set 1:\n\nFixed here: https://review.opendev.org/#/c/735800/1","accounts_in_message":[],"_revision_number":1},{"id":"33bbae60153894a058536d0f6fb95e7e4f6cb8ed","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 08:44:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"037c69137564224a0cfe56b84573cc7bf289d53d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 08:45:06.000000000","message":"Patch Set 2: Code-Review+1\n\nwhat about amqp1?","accounts_in_message":[],"_revision_number":2},{"id":"745ddd07b4abadb4c288db5107ae4b7c25744c71","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-16 09:24:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/c56d5db70cdb44ccb37643a485675719 : SUCCESS in 38m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5021b1bd533e5dec9573c3c097b7ee5d3950812f","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-16 09:53:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"66f3ae5b49978e96596ce75a12dbcdba964c2141","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-16 09:55:26.000000000","message":"Patch Set 3:\n\nah, thanks!\n\nampq refers to amqp 0.9, which is rabbit. VMWare seems to have a product being able to digest metrics via rabbitmq.\n\namqp1 is supported e.g by Apache qpid. This is e.g part of https://infrawatch.github.io/documentation/","accounts_in_message":[],"_revision_number":3},{"id":"c4de2e2edbd927b03ad92c7c548698bacc822386","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-16 09:55:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f70f5eeba9420c77bfbb5997b33c40a3e63471f5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 10:02:29.000000000","message":"Patch Set 3:\n\nthanks, Matthias","accounts_in_message":[],"_revision_number":3},{"id":"26614b7e068690b8b7ac5475ee86315be01e7b11","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 10:02:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"33db4d7c8563a728aae56363f71ce1c425c6510b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 10:42:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a00cf072e659e1ddbafa6858d334860e41bf96e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 10:47:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a40aa1b607b882a94664aaa22c43db3d08ead3e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:59:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5dc0146c849e462db94c05e4ff74861a : SUCCESS in 40m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b06254daa9dc422f8c465c2888767edf : FAILURE in 43m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f16f7384f91c468db69207b1657252d9 : SUCCESS in 1h 37m 20s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5140f4ee8d15491cb43ed575aa5c0238 : SUCCESS in 1h 32m 48s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4d068b7a40ea43389325dfc7aa77631e : SUCCESS in 1h 40m 10s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/58b0178b378d4f7d9153fc92a30b0f33 : SUCCESS in 1h 43m 58s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d90af49af7784267b5818af66bd97000 : SUCCESS in 40m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/25d7ba4a592f42f19e977cc581908cd3 : SUCCESS in 39m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/840629abfb6b49e18a4545999842469c : SUCCESS in 1h 10m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b87220ef8e5f4ba8a443d6d77bcacd91 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba22535eea7441f9222d3e0b5ec6027 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9590066a60014f1d8498f345dc991839 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b55702aed9354e9a99ade38697033d93 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62c0e0d7e454b43b89658406c538d8b : SUCCESS in 10m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3042f20264374c5da5b5cf2e9698cdfa : SUCCESS in 28m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef03a5dfdbfa41fa9f98425b4ab8ef23 : SUCCESS in 26m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f91b28146ae4d74a552769c4f740d12 : SUCCESS in 1h 07m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/732903339e5947eab1c2737da58a4d26 : SUCCESS in 1h 25m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/506027c30fd041caae021c812649032f : SUCCESS in 1h 35m 48s","accounts_in_message":[],"_revision_number":3},{"id":"693d1e04bcb261d5a01bc28534271ba351b65336","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-16 12:09:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/954efa62c8ff48bdb1e3e3e08da0cebf : SUCCESS in 1h 01m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0ce5574dfb2843c1bdf05b97c183adad1dd61366","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-18 06:22:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"fe18e3a20823e81d314b67026286fe759c18848e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:02:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ae5cc7be9f1c4704be679fe46f892c77 : FAILURE in 15m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c99fb1458e4223577ff79a911e63c79ae4f70044","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-18 07:34:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/fc20369193054487a1cb6c5f3d132a52 : SUCCESS in 26m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"30240102a40f776e2a25c4ac3139dd8ddb877e2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:33:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/76780cea8edc42babcbcf0ed5963050e : SUCCESS in 21m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8b5a691e642848fbabd0dfba06a7ca53 : SUCCESS in 52m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d6047a3d338744a9a77a5047a788b66d : SUCCESS in 1h 18m 23s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1d70bcb7946445d9b5a848d3827aaa7d : SUCCESS in 1h 12m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/91497ba652b64962b076b0d95adfaad9 : SUCCESS in 1h 23m 08s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3f3359cd0f3844d59d67469d933cef58 : SUCCESS in 52m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4fe549e12509445696211622e9150c08 : SUCCESS in 32m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/14b7bc1fcc0c4482bde682d82647ef8d : SUCCESS in 19m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0db1b8d34ea54eb4817119e209f2fa11 : SUCCESS in 1h 04m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9a52512ad904409988d7f01b1001265 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71bda80462924134b589336a71572c6d : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c5f94f68a6f491dad3deac06349ff08 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f19403d20681469e9277097e2e19540d : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c63538479941f784d7e77c2a366f92 : SUCCESS in 14m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3de342587f434b7584ca12ffd382a2df : SUCCESS in 1h 13m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3218a6216884bb5b0766d4cfa82e966 : SUCCESS in 1h 30m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7fb678a6281542439d74befd75a494ae : SUCCESS in 1h 49m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5175f4ff4d2e4c528b9d246c5e436632 : SUCCESS in 1h 23m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6432f4b248db4ccfbf4b2a7ac04252f4 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":4},{"id":"5114ee5bca6551d86acb87bdf83e88e2d606cf54","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 09:02:59.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9445e42de4742e7d854ff8aba41228cb87d1f0ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:03:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"95c1fec84a8fd7ff485f76e463d705976e3d3775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:34:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8be377d9fecc4149942f54df5a3457ce : SUCCESS in 1h 04m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f6725618b3564b50a7428367f2a0a201 : SUCCESS in 1h 19m 11s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c14a21bedff64ec3a62a8fd30be4551f : SUCCESS in 1h 26m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4613a776229447ebf91b7a5ff4a83f1 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b86fd8fad4d4e1583023c6adc7d47f1 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbbce82713c840ab9824ec2d852af973 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3d80737c100453e9edf5d5086412c4a : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab324c8940c84fdca14b0b55a01699f4 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60d026c523944b8b8e0ed37aa722c468 : SUCCESS in 41m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba469d44dbfc44e68988fa008a448573 : SUCCESS in 1h 01m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ddf153bb98549a99f12fed78786f1ca : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":4},{"id":"d2dfeb6970e1292c0be370d9718e8b49d06a7442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:34:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3cca11bba0b17badc5808ebf33b83c051ce83a79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:38:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48cd0a37d0634181a572694037567f12 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":4},{"id":"2464f51a5fc62f23b7f01cf5159cd51425380921","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-18 10:47:06.000000000","message":"Patch Set 4:\n\nthank you very much.","accounts_in_message":[],"_revision_number":4},{"id":"7df8004842d8904f88115d3cb0d10ff80c3ffd8c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-18 10:47:17.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 63d689c4806d00ab7ce0ed4ea1f0c01be14ae07d","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8bfa655007e9795ee6b4eba09cba4f403c113b48","revisions":{"5aa7fe513074ba6c63aac53ec23a86216b03f619":{"kind":"REWORK","_number":1,"created":"2020-06-16 06:25:31.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/75/735775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/735775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/735775/1"}}},"commit":{"parents":[{"commit":"c07409bfabdd8135316609160e6477d202b46b95","subject":"Merge \"Revert \"[to-revert] Pin ovsdbapp in neutron to 1.1.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c07409bfabdd8135316609160e6477d202b46b95"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-16 06:24:29.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-16 06:24:29.000000000","tz":120},"subject":"Add collectd-amqp to the collectd image","message":"Add collectd-amqp to the collectd image\n\nThis allows to write collected data via amqp.\n\nChange-Id: Ieb88fd60d177f19a635ada9366193e57c516f90b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5aa7fe513074ba6c63aac53ec23a86216b03f619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5aa7fe513074ba6c63aac53ec23a86216b03f619"}]},"branch":"refs/heads/master"},"45ca9dbe5572f179f1e2fce331e1e43344390e8e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-16 08:44:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/75/735775/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/735775/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/735775/2"}}},"commit":{"parents":[{"commit":"14ef775c6ea849964d9afa03575a415687d9b51c","subject":"cyborg-api: install proper WSGI for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14ef775c6ea849964d9afa03575a415687d9b51c"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-16 06:24:29.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-16 08:44:45.000000000","tz":0},"subject":"Add collectd-amqp to the collectd image","message":"Add collectd-amqp to the collectd image\n\nThis allows to write collected data via amqp.\n\nChange-Id: Ieb88fd60d177f19a635ada9366193e57c516f90b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ca9dbe5572f179f1e2fce331e1e43344390e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ca9dbe5572f179f1e2fce331e1e43344390e8e"}]},"branch":"refs/heads/master"},"2ce8ec26c23b51b2d59c88e8614a25325c46b067":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-16 09:53:06.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/75/735775/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/735775/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/735775/3"}}},"commit":{"parents":[{"commit":"c07409bfabdd8135316609160e6477d202b46b95","subject":"Merge \"Revert \"[to-revert] Pin ovsdbapp in neutron to 1.1.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c07409bfabdd8135316609160e6477d202b46b95"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-16 06:24:29.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-16 09:53:06.000000000","tz":0},"subject":"Add collectd-amqp to the collectd image","message":"Add collectd-amqp to the collectd image\n\nThis allows to write collected data via amqp.\n\nChange-Id: Ieb88fd60d177f19a635ada9366193e57c516f90b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ce8ec26c23b51b2d59c88e8614a25325c46b067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ce8ec26c23b51b2d59c88e8614a25325c46b067"}]},"branch":"refs/heads/master"},"8bfa655007e9795ee6b4eba09cba4f403c113b48":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-06-18 06:22:58.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/75/735775/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/735775/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/735775/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/735775/4"}}},"commit":{"parents":[{"commit":"9fc8f3e9872c4a9c5fca8419979be3a254dd893a","subject":"Merge \"Switch to tarballs.opendev.org for source images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fc8f3e9872c4a9c5fca8419979be3a254dd893a"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-16 06:24:29.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-18 06:22:58.000000000","tz":0},"subject":"Add collectd-amqp to the collectd image","message":"Add collectd-amqp to the collectd image\n\nThis allows to write collected data via amqp.\n\nChange-Id: Ieb88fd60d177f19a635ada9366193e57c516f90b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bfa655007e9795ee6b4eba09cba4f403c113b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bfa655007e9795ee6b4eba09cba4f403c113b48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
