)]}'
{"id":"openstack%2Fkolla~735800","triplet_id":"openstack%2Fkolla~master~Icc9d65c4ce22c959145a8d95fc6b823252659ddd","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Icc9d65c4ce22c959145a8d95fc6b823252659ddd","subject":"cyborg-api: install proper WSGI for CentOS","status":"MERGED","created":"2020-06-16 08:11:46.000000000","updated":"2020-06-16 13:46:35.000000000","submitted":"2020-06-16 13:42:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735800-1592314934925-54d98340","meta_rev_id":"e8f7762184e911b8e621cdce684e417ec633d809","_number":735800,"virtual_id_number":735800,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 13:42:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-16 12:11:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-16 08:32:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-16 08:34:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-16 08:34:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-16 08:34:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-16 08:34:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 08:32:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-16 08:34:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-16 12:11:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-16 13:42:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47d9f13e83db5fe895ecd73149f60b9a392e2e41","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-16 08:11:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20b9d4428ad4ca780bb9715a25eaf4424b573ce7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 08:32:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64fdafa909ff908666fdd192ebc3f78a44d84b15","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 08:34:25.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"8ba78734a28773479534ec8734cd5bb3e7d9df8f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 08:34:44.000000000","message":"Patch Set 1:\n\nI could use all switches at once ;d","accounts_in_message":[],"_revision_number":1},{"id":"5e214fd010415f7bdd389523d98a4e53c1b1ce37","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-16 09:05:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/c700c6d27b5f4f41a008179fce467751 : SUCCESS in 51m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a2a51bba1693e9d01ba61954c0298cba48b75d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 09:58:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/27bd3b16b9cc454bb60bc85f53a61c72 : SUCCESS in 32m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ba16a920f1c84f888572482911e46401 : SUCCESS in 37m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1da667504ef347c8b7a3b5fd19856a1b : SUCCESS in 1h 27m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8369a012ba83468f82025a28ee048dbb : SUCCESS in 1h 26m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4df0071f06f64b34a8854fe74a9d64bd : SUCCESS in 1h 29m 06s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/9eb52018b2b847b49f9c92a72f1032d2 : SUCCESS in 57m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/60371d7b61fa4b729265158d6c12764b : SUCCESS in 27m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d350f2c1e7e94795b87297e72cf32e07 : SUCCESS in 27m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d5a56a86195f4d338a37cd67b8f55e50 : SUCCESS in 1h 07m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b33b914b2f8b42db8716e7f06edd6460 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af44c0580fd542dfaeab6b356ec8a28a : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0c9be8a00ba499da0039e3fbaeb57b7 : SUCCESS in 10m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5938f9f6703e4301b650b2fec5a641f2 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab1e648033a846d085c61c56118891ea : FAILURE in 8m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1327e86803341d1a5f741d06b5f5645 : SUCCESS in 24m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/74efde54c5e24f18a30f292a77c9738f : SUCCESS in 27m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7087f7228d5408ea19fbdd81b756e8b : SUCCESS in 1h 07m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/308d93e5b1a94ce89bc442052b6cde67 : SUCCESS in 1h 05m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/432b74cd778643d1a965a9bd9c6056b9 : SUCCESS in 1h 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8441fbe04759341da03e6902fa2bd367b61db3f5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 10:09:23.000000000","message":"Patch Set 1:\n\nrecheck\n\nwhat the docs","accounts_in_message":[],"_revision_number":1},{"id":"c1dce7c17264343a3b5bf927f31ef7b1db6e9e55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:08:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/51d2fe7a7af1453fb86deb90040bb746 : SUCCESS in 41m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c4eb20d5f22c412eb207960a15108d5a : SUCCESS in 59m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c8dee724bd69456bbe97434579a6f10c : SUCCESS in 1h 46m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b11740bc1d4d4a71b52feb0f38a44451 : SUCCESS in 1h 32m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8ca3a77f6a7947aaa6ba508d6889cf53 : SUCCESS in 1h 43m 38s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c598a77c5e7b4815807fcfd6551389ff : SUCCESS in 1h 03m 07s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3023fb6cd66e46bc84614e1b61a3f96e : SUCCESS in 55m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/459c778be6fa41438f0a17364237bb94 : SUCCESS in 36m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/71a57ca32d3f46e1b4d10795e4092ab8 : SUCCESS in 1h 10m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08c37e0e54b047e385968b60e5da4208 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23af2121036e40e4a1674076c4bf3658 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6123cfa180e4a8b8f3dd03a63ba2d6d : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2baabf65fae74f459f4e05cf9469c855 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3325b4db9042e89b70f0a130cc854e : SUCCESS in 8m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d86a0b8196741458889155d394ad3d6 : SUCCESS in 28m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b118d46ca1ee43e9a90d0a4328677633 : SUCCESS in 33m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd63c0667d514af4a69264347e81707e : SUCCESS in 1h 08m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dadbb8d2d9c04b5193914e3812e8eb8d : SUCCESS in 1h 02m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d9549ec1ed914d26ae34a5ffc8ca3443 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"da2ad08c165ae38ae96706782dff52406454e43b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:09:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4423e38b226395eac3f3ae51dabb6f35c4ea553","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-16 12:11:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ae2a3a0b30dc4228920bc0d00500e195 : SUCCESS in 1h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4989405973b53a2ecc608f16235a4c0435e4a98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:30:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/94497b67496d4e9e85bbf02afe0324be : SUCCESS in 1h 40m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/897ed305793d43e4b84032b0898fae85 : SUCCESS in 1h 31m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ea527377de24110943e60883b3bfcc1 : SUCCESS in 2h 24m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e06184f1073459c8890213c48e122d3 : SUCCESS in 1h 38m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"353228cfbcc84138afe2e29809ef505f6a0bcd00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 13:42:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"24b3666f2d4a856e0348d6c524bc8fa53f84220f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 13:42:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9a81bca5d9e94940847f5ee5069b7322 : SUCCESS in 30m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4261d106cca5479abf8bd381124dab17 : SUCCESS in 1h 31m 56s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9d67b5608bb841cba87cebbf5b298459 : SUCCESS in 1h 26m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8675429b818e4dc09b9fefff7fee4a36 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38741e03a9e3412a83327c75c8b45d63 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a641b9c0929e406abb98faebb281ccd0 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/893afe780d2d42a6ae30f1813642ce66 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edaef6a7e264db3b01aa0eada1a5016 : SUCCESS in 9m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10fb907932bb48f0ac6007ffcb878b61 : SUCCESS in 41m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22bb4da5326d427d8f25a0fee8c37fd2 : SUCCESS in 48m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9e62d56a36f74084b796dfde03827950 : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5c95f2d265a293972bea6024297bcac9b6c19308","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 13:42:49.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 4d504c1ec9335c020a0f8f6d9a1b8335c5858844","accounts_in_message":[],"_revision_number":1},{"id":"e8f7762184e911b8e621cdce684e417ec633d809","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 13:46:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7004f4b4e694b9d8161ca110c1202fb : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14ef775c6ea849964d9afa03575a415687d9b51c","revisions":{"14ef775c6ea849964d9afa03575a415687d9b51c":{"kind":"REWORK","_number":1,"created":"2020-06-16 08:11:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/00/735800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/735800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/735800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/735800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/735800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/735800/1"}}},"commit":{"parents":[{"commit":"c07409bfabdd8135316609160e6477d202b46b95","subject":"Merge \"Revert \"[to-revert] Pin ovsdbapp in neutron to 1.1.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c07409bfabdd8135316609160e6477d202b46b95"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-16 08:11:36.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-16 08:11:36.000000000","tz":120},"subject":"cyborg-api: install proper WSGI for CentOS","message":"cyborg-api: install proper WSGI for CentOS\n\nChange-Id: Icc9d65c4ce22c959145a8d95fc6b823252659ddd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14ef775c6ea849964d9afa03575a415687d9b51c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14ef775c6ea849964d9afa03575a415687d9b51c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
