)]}'
{"id":"openstack%2Fkolla~735826","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I3e307efc43f934a944a91d5d131a11f607411df2","project":"openstack/kolla","branch":"stable/train","topic":"bug/1882863-stable/ussuri-stable/train","hashtags":[],"change_id":"I3e307efc43f934a944a91d5d131a11f607411df2","subject":"add libibverbs package for openvswitch","status":"MERGED","created":"2020-06-16 09:21:49.000000000","updated":"2020-06-29 10:09:40.000000000","submitted":"2020-06-29 10:08:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735826-1593425290013-3d5ab54e","meta_rev_id":"192f3d3e51507de4cba7f211bda6bf818fe7ba5a","_number":735826,"virtual_id_number":735826,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-17 10:37:56.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-06-18 10:59:18.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-29 10:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-29 08:26:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-16 09:32:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-06-29 08:26:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-16 09:21:49.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 09:21:49.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-06-16 09:32:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-17 10:37:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-18 10:59:18.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-29 10:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c4daa4e7359e1c339029f0d1894240dbc90619c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 09:21:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"8e240b1f749dea2dcd59986ec08fe0d5538c0ffa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 09:32:19.000000000","message":"Patch Set 1: Code-Review+2\n\nthis only really need to be centos8-constrained but fine if they build c7","accounts_in_message":[],"_revision_number":1},{"id":"4cf51bcb1b63a62717b1a41933e5039a21709a6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 10:46:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/02404c256df44b72adcedc51af7e729e : RETRY_LIMIT in 5m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bc12d1484f2b4bb08c39d7ec529383d8 : RETRY_LIMIT in 4m 40s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/149dc29e32954697b26cd8ce97c81105 : RETRY_LIMIT in 5m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/92ce810f02bf4fbdb77ef4bc5307739f : RETRY_LIMIT in 3m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/142a5000852c483ebdd9f17a5694a87d : FAILURE in 6m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0cf2243b21ed485998f536b157ea37a0 : FAILURE in 24m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9d8002dca4cd4ab6ba75fec628937b40 : FAILURE in 6m 13s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/eb4e3780873340a7b838047f8e0e2268 : FAILURE in 22m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a8a8256b941e4f86a26fc2b2c7ac5f8c : FAILURE in 9m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/37eddcd02b8c464c9d70d525fa9ca994 : FAILURE in 21m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/362733b2da89469abdbbd6b4e903856c : RETRY_LIMIT in 3m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/97649a00cdd748a48d06e81649d2d541 : RETRY_LIMIT in 4m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7365695fa19547e790693ae6dbb53100 : FAILURE in 5m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dae265d853a1439c929d0496c2b1acaa : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed377e8e7d8847a0b2c926b64d26a5e7 : SUCCESS in 7m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8521feb3ab0e4106a00abc5798c2f43d : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49762621d8d44a478023b8b03083e673 : SUCCESS in 10m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79891a771982458a8c962db385f7b59e : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6e805437794d65b0796d4ba1790530 : SUCCESS in 10m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/213c42ec05b5486bbc790d769bc406a7 : SUCCESS in 23m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c52d1cb154814294b2cd936424f01590 : SUCCESS in 27m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ac66f6c796c493698e96933afe145b9 : FAILURE in 7m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c40248bdbb3549179d4f70c50cc03f6d : FAILURE in 7m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8fa329cb1b6a422d95e8c9fa07592937 : SUCCESS in 1h 09m 27s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"74a89693882fd6dca1f31f8089dd1e780ea76248","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-16 21:53:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70df2c9f5bbcda783e1878831db9bf3192dfc45c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:18:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/13ee9a11209e4d4cb4ffa7bd1df211e2 : RETRY_LIMIT in 3m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/01ae301cf92d4715b6f1250cd0fcbcef : RETRY_LIMIT in 3m 02s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a86917debb1941c1868f182f6f975c95 : RETRY_LIMIT in 2m 50s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2310fb7a9d1e4e92b1c62f4733de0465 : RETRY_LIMIT in 2m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9291311dbbbb4644bdf34851295c982c : FAILURE in 6m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7b222ee4dd7142a09da5fd7a77ba0ead : FAILURE in 24m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1fd2fcedf4ca4f2f8cc58cf89c9804bc : FAILURE in 7m 37s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/94d89bdc19144bc9af62ff287c7f331f : FAILURE in 20m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/069cb309c5c54fd4a7de7dc3431b0dad : FAILURE in 7m 20s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9cc2ab90687145968024f03a5f06f35e : FAILURE in 14m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/482af911977349cdb76e01bdef37e52f : RETRY_LIMIT in 2m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0870e626ad254a5c82d8d39f397f84b8 : RETRY_LIMIT in 3m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d45b54fbe88842cb955796611618ec27 : SUCCESS in 1h 01m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34497994482f4da3964ed1d8e9b6791d : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8e295777f23452da7747cfd05258f6f : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57ffadd2fecf4b268a1cdde9982091fc : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67f417f700d5487aaf52c79f5e972b9f : SUCCESS in 7m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c49ae50f4ae491cb4e42e6129c014dc : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a1f4ba9bb34c4ab1edda26b1fe9c9a : SUCCESS in 8m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/55e65ffe58df4feaa396573ad9b71f8e : SUCCESS in 25m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a074a00377174ad3b45044bbbef067dd : SUCCESS in 1h 01m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8dc0700f06f04f37b0df5b22871cea2f : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f871269d5ea34837abbb4431c5eecdc4 : SUCCESS in 59m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d95ee112f644ebd9b84d1269b90bc39 : SUCCESS in 1h 12m 23s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"e8662c6de30e563d8bbd394aca4800a7b86126ef","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-17 10:37:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"946795bca5e00c44b06baa63b9cd523f781fde18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:44:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c62f9c559c394fb2bac3fd8502fc5388 : RETRY_LIMIT in 3m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2068df51f38e485a9c5168b8140ef68f : RETRY_LIMIT in 3m 17s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/281a24b2ef234949b07273d47086595d : RETRY_LIMIT in 3m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/aca197039ec74caf97687472be3ce6b5 : RETRY_LIMIT in 2m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1a522dd147b645c497121b60bb51d834 : FAILURE in 6m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1268de2d8743453ea305900923f84e9e : FAILURE in 24m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/48144979a6134da9b98f9203c6c0fa14 : FAILURE in 5m 48s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1be4f0b74a2a40e99429043dd5ff0f8f : FAILURE in 20m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7a19d36d6a364ebd9377fccc852f1e3a : FAILURE in 5m 23s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/061852780c834cbfa08bb346816cee6b : FAILURE in 17m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/85c8f3dd7bd74e7195c3fe2bbd90c350 : RETRY_LIMIT in 2m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cb5d41a847284c198f044dbd48b01413 : RETRY_LIMIT in 2m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc63516d7f3d4431966de81728ec9692 : SUCCESS in 44m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c885f2aed0a647cba088318710b0272a : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27640f412ea43b9b8f97d7d36972917 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0875697478994aecbe545ba5405651e1 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9261e46e9e164897aa7b4fe27c688956 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a08bd739256e4b57843cfca6ab6df360 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631e8981edc840e5aab037a9565ab492 : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3bf77732861946d09da332f437a560da : SUCCESS in 23m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d08b6558b59143dcb3f2d5ff3f45dda6 : SUCCESS in 26m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/add7d13acedf429b8c3da322590e0cc1 : SUCCESS in 57m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/41f6d3e97f154b89a5a9a277c4db5ebe : SUCCESS in 59m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/113429571dd8467082e8814d761bfd12 : SUCCESS in 1h 05m 20s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"4a358a57cf6048e0f3ec6f321c86d99008bb5c4b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-18 10:59:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d657f516bcd406d380fe2640768d66d40a7377fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:30:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/35bd9b5e0b5447a89ed0b5585a74c99f : RETRY_LIMIT in 5m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4734553bfb7b418ba834072c19644ea3 : RETRY_LIMIT in 5m 36s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c8dfea773e4840c688713b5a952a9cbf : RETRY_LIMIT in 5m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/bad0585a5a3b43ab8f90ff4cd44b459a : RETRY_LIMIT in 5m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cdd35856f8884cc9b2f3aca9561d2101 : FAILURE in 6m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a0afa99e7969446ba7a5be7f24e05a89 : SUCCESS in 1h 13m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b6bab04a285e438f8978e66bf2d1e920 : FAILURE in 8m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/aca3c36d1ab846bda74bc6895288cacc : SUCCESS in 1h 13m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b9f9947b37e3495c8a179e36c82d9a6a : FAILURE in 7m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f25f6324c053499fbdef54028310aed0 : SUCCESS in 1h 02m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e3c5286c279e43efb4066a6ad3c27102 : RETRY_LIMIT in 6m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/89bab03aba5f4c219f821a46512c5b78 : RETRY_LIMIT in 4m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9036981f3add4aaab977f91b55d649e8 : SUCCESS in 55m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66328d96c5be4258af1318d63583e11b : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be2d74cdc9d43d4932d1563b1ca4ee7 : SUCCESS in 7m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5cbbca2971c0421dbebbcfe77b47c418 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bfb4db7a51642e3a4fc12897a113378 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee6ad26fa7a34d20a92858da78ad7517 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ab09033a6124bc7994ed67b5b8d2b14 : SUCCESS in 11m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c5a2340f1c944b918f88114ce68c7a17 : SUCCESS in 28m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/66cbab246b7f4c5dbb104f22547a8fbc : SUCCESS in 27m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c1efa3efdbf41418739c2ad9168bf78 : SUCCESS in 1h 02m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d5fa0ded23346c9abbacca5b426b9db : SUCCESS in 57m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f7adaead645543c0b9cc7fdbd773f80b : SUCCESS in 1h 22m 15s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"b27e17e671359f440b41c6996df404e8d0c47ed5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-28 09:49:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"35398a718860a3e3c05ccc3e9bc164ead100a7ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 11:22:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e29971b782b1436eaa9e5074d97f90dc : SUCCESS in 44m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c0e433e6cb6f4e5fa400c3f10e5c7d67 : SUCCESS in 1h 08m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5d23b761f25c4c9587d64dda8aebe9f9 : SUCCESS in 34m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/97f74117116847cfab0c088ff4f6efb2 : SUCCESS in 27m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a70ed22b478b4f6086799423344b07b1 : SUCCESS in 1h 28m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/945abc7fa0f14bb8b1fb86d47ea49d36 : SUCCESS in 1h 18m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fa8279bf42114364a20a4815fff240a1 : SUCCESS in 1h 20m 17s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7e661109e0c14a3fbcecf3aba1c82b6c : SUCCESS in 1h 06m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c0a86992344e4031b8a76a9407818635 : SUCCESS in 1h 26m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3969be72b85b477db63a0395545d178a : SUCCESS in 43m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f947c6d0ae33428abe09238d807a718f : SUCCESS in 24m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/06367edde5234448a6c0e83508bed3b3 : SUCCESS in 19m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d00e7926d5324aa2bc98a69abee61b0a : SUCCESS in 56m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a886737e2c84be0a83a48dce4ee6038 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4faf2007ff694d0fac34ebe3b234bb23 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c06e01a618844d1ad1717d1a098e2a4 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3103308812c46489fb5cf45aea2d977 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30000a39dbd94e4a8987911ada23fb41 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab8572024254c2b91cd616aab3f84cf : SUCCESS in 6m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c97e07e69674c16ba832fef54bfcfe1 : SUCCESS in 51m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/833d9fd30bed4694a0909d760aba5396 : SUCCESS in 24m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/689c8665a6ec4e4c968a8eeae9f8fcc4 : SUCCESS in 51m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e5549ced3cdb43d39fff1b1248b7d5a1 : SUCCESS in 53m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/597ed7e737c045a9bc65ad9a9ec4d4bf : SUCCESS in 1h 22m 33s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"55297f1cef8be51cdbf48ae8aeadf1189e7030bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-29 08:26:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9fc4f33e87e9e29fe69197893e81e0bf7dcc71b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:26:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33fd55dc60d9d15991a223e93be666f19ceda601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 10:08:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/222d428a5afc4d678eb76053adde20b2 : SUCCESS in 57m 14s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d0426f6aaac64da6b3da8f1600e15f67 : SUCCESS in 32m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2c491bcee5b8401bbb487c828002b373 : SUCCESS in 1h 36m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6cc631325ba74b7389b4726b35131079 : SUCCESS in 1h 14m 33s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/33e17b1c3ef145898e29fcbddf083702 : SUCCESS in 1h 31m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/25ee2087c17244c08a8a80dbaa93c75d : SUCCESS in 51m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8deb1cf88c634df5be55c81f3d197756 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5076e3a3f35a40dcab9f5f7a5c1469bb : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a0e845201d349c3bbed9dbaf59f3879 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99ba64b3e9984a1888ce2c9ce3b49bb7 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55a0b969ac1440489bf1e88a9e40cd8a : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59be58a29a2479a9989924c99d223fb : SUCCESS in 7m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0dae8cccbe984bbcae2342afc93b9734 : SUCCESS in 38m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ecdcd6a68e084ee889e112c0d96cbe0e : SUCCESS in 1h 18m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b112b2f27d8b4640b1865214939308a5 : SUCCESS in 1h 09m 27s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"393072205fb4ddfc05ad077f09b39647d2750cc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 10:08:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e04541ef8f36f0c2858bb4f6e86e7b96b5fbd2fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 10:09:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd3f9e1b50540d088471db65b0e3f8b : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d821722f2779c0c747a2fa317f3eca3e60dc7cb","revisions":{"9d821722f2779c0c747a2fa317f3eca3e60dc7cb":{"kind":"REWORK","_number":1,"created":"2020-06-16 09:21:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/735826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/735826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/735826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/735826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/735826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/735826/1"}}},"commit":{"parents":[{"commit":"47df05074204b87805986c7e9ae6743e3dce28db","subject":"Bump versions for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47df05074204b87805986c7e9ae6743e3dce28db"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-06-10 06:46:06.000000000","tz":480},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 09:21:49.000000000","tz":0},"subject":"add libibverbs package for openvswitch","message":"add libibverbs package for openvswitch\n\novs images which based on centos miss libibverbs package.\nthis ps add the package to neutron-ovs-agent and ovs-base\nimages.\nCloses-Bug: 1882863\n\nChange-Id: I3e307efc43f934a944a91d5d131a11f607411df2\n(cherry picked from commit 6d71ac843e85126bfe35c9a39d64c1767d360c74)\n(cherry picked from commit 5bd7d3fc7adde6d0f87c6dcb0ce497fd94bf0601)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d821722f2779c0c747a2fa317f3eca3e60dc7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d821722f2779c0c747a2fa317f3eca3e60dc7cb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
