)]}'
{"id":"openstack%2Fkolla~735841","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I79af6c37950e156018a9204fbcc7417cd7d41012","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I79af6c37950e156018a9204fbcc7417cd7d41012","subject":"CI: use wheels from infra mirror","status":"MERGED","created":"2020-06-16 09:44:38.000000000","updated":"2020-06-19 15:36:18.000000000","submitted":"2020-06-19 15:34:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735841-1592580870596-9795058d","meta_rev_id":"8d30875a4c9510960a891933432f1c2df6a8f2fe","_number":735841,"virtual_id_number":735841,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 15:34:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-06-19 13:33:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-19 10:51:11.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-06-19 13:33:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 10:06:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-19 13:33:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-19 15:34:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21d87554d28806fe0afe806cc60f9a14099b0dc8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-16 09:44:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b6ea16e9695059815e9c3410655b3bd4d6e9a3b7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 10:06:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"447d7e77c9dc0d92d91ab08c2a4a4d8222408894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:25:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4de65893d3f043c0b14d64b7c685ff85 : RETRY_LIMIT in 4m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bb8cb1dee67d426d885dfabcb08afd4f : RETRY_LIMIT in 5m 27s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a4255c701da74e80af5ad5774951d1b4 : RETRY_LIMIT in 4m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f0ce29cb4c62458383827695b09b349a : RETRY_LIMIT in 4m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1e1a6a64c47f484e9316f6c40569bc32 : FAILURE in 8m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/57db345cbadc48bb995241c754fd571d : FAILURE in 25m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0829fdadc0f242edb194808b7778050e : FAILURE in 7m 24s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c43fcac89b454189bde69b16515429c2 : FAILURE in 21m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/33ecc2366f7c4b3ab0bd71ccb84bb82a : FAILURE in 8m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/742a5a5551eb41b99f987989b74cb4d2 : RETRY_LIMIT in 4m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2311a7996dd943cc8a66d5b6712f47f2 : RETRY_LIMIT in 4m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/806884e96d304dba84be0bc32ca0678c : FAILURE in 5m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06a4aed64c6c425c8bee2dd9aa3be8f9 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c80198c258a45b3853792027128615c : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bf365e36ad7477593e5001ffbb79b38 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2959ca885629408c897369ba17eb68be : SUCCESS in 8m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be8adb64ec074743a52b0fc2751c6bf4 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db849880a23c49089cda79f8e43dfba7 : SUCCESS in 8m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/896a9e8e8d6843ce8514bd51e6420825 : RETRY_LIMIT in 4m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6453d9563454453ba3d8d5774394e79c : SUCCESS in 23m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/55c42053e1be4414aa5976b5068bf817 : SUCCESS in 54m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ba21c246030048989f586b5ac66c2406 : SUCCESS in 59m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0a73ad92a9348f8998622da6827a344 : SUCCESS in 1h 13m 52s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"6c0bfc38e3212205a4891fa0ac65a025a6b52847","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-16 21:53:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b253d43f2da288077863d8cdcf382da7be8393d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:29:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/227c3dc8416d4f9d98ed1a90bfdad072 : RETRY_LIMIT in 2m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c5ae34727aa84d2eb744a481e5ea23e4 : RETRY_LIMIT in 2m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e239b5dc965942c3844b5668be096f50 : RETRY_LIMIT in 5m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/56cee8d431cd49b89835ae395b3017cb : RETRY_LIMIT in 2m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/96d12ab8976a4b458a0902aef0f6708a : FAILURE in 7m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7a25a0a59bef42cc863a1c156138d0b7 : FAILURE in 33m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c0fec62953cf4c61a59b1ef4b26d4361 : FAILURE in 9m 12s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3d55f9cb64cd42c8a48e93aca41d635b : FAILURE in 25m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9325a25734ee48d79abfd52f94321fef : FAILURE in 9m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9c11568006c54cdea33d2d5a5f5a6a95 : RETRY_LIMIT in 3m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ffef15e130854783aba82ade6aff26ad : RETRY_LIMIT in 3m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/daff778345024fb2b0e795ae6af1ea28 : SUCCESS in 1h 05m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0669104c3bd54e729c803d0bf43f0213 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a21e2c499b4733a924dd6ac40bd323 : SUCCESS in 9m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b031a9bcb054611929ae235f45638a4 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e2d04da5dfb44d49ba4cbafa03b9923 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/608dd566c10743c992a635a60201eaf8 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b254079fb284ee5ba8c6ec5a8a91c08 : SUCCESS in 9m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b2c2c70ef41a4a31b3c8e52e8e6f0c17 : SUCCESS in 33m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c678ffe96d3475995b577f741c0f919 : SUCCESS in 29m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef87ad6415c14d9ba1b8ba1954d8db6a : SUCCESS in 1h 05m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87b206c1a697497196e3a87b530332c1 : SUCCESS in 1h 30m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5711fc42f5f64f5da8b2229a8771d95d : SUCCESS in 1h 04m 32s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"78d484a5e0f23b4ada220fe4f677b1999f412857","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-18 10:59:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"582ec87f58cfdc15d7d0bf0bf7f16f7ef664b3a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:17:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fc719bfee65742088b4a54a5db71b375 : RETRY_LIMIT in 6m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dfcd86688dd849898180d0c3ba2d80e6 : RETRY_LIMIT in 5m 43s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9681d7904d9d4054bade2f5df3ae8fe6 : RETRY_LIMIT in 6m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/125f78f4cfe54f1cb37e7c658aa7e250 : RETRY_LIMIT in 6m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b5e5af7eb9a6435b9c2f85f5be1dcf42 : FAILURE in 6m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4364dd91335e4a61869bb79da907b21d : SUCCESS in 1h 11m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cb2e0638c48f446aafc02e730c6ac030 : FAILURE in 6m 13s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0d4c7d1f457d49b69636736a23336b41 : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b726650f023b463bbfa0394fa4f9293b : FAILURE in 6m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0853f62859674589ab7aef440427b6de : RETRY_LIMIT in 5m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8ebf663527174793933a8ac1f651057c : RETRY_LIMIT in 4m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/68671465b94648fcb64d2381c4e5865e : SUCCESS in 1h 07m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f03fab699d1541459a7e7fa2bd326563 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/989c5d6c0d2345cda4b1dcec83f788c0 : SUCCESS in 7m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3031f52ff647486aa66d5820eb58aa2d : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cabd4365dec94f9c9703d369b606255f : SUCCESS in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d81ed4336ea492fb3244037c6280288 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a7278d2a03a4223a3c4d9af591c44e2 : SUCCESS in 9m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa348c552900469eb75f21f02edc3f51 : RETRY_LIMIT in 4m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/82570f54bc2b49359b6e00f3a7677c51 : SUCCESS in 22m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fdf3f7090ab34e75927ee0b19e133ae7 : SUCCESS in 43m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d6b4837740644b6a99f77ce3ddd61bc7 : SUCCESS in 58m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/18ebeab70f094e0d97c44e04ae23559e : SUCCESS in 1h 09m 03s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"2a9bd274156150ee62f2248868f34ce9659e8d58","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-19 10:51:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"41ae70672befcd0b9b2809ca2b24c68d5d54b8ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:46:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9d859a1a82164fcc8199c1e3b47f16d5 : SUCCESS in 50m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d8bdf5b6316944b88c5fde93f9a7cdd0 : SUCCESS in 1h 01m 44s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4fee47f5b5f241229c8e2752bb5c0a5c : SUCCESS in 24m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/dd3e6f1eaf7f49d184db6e438c061dee : SUCCESS in 32m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fbbdb245a88749388bc800e4b0a65448 : SUCCESS in 1h 29m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a096aaa7b08c4a6c8533bbf36c132da6 : SUCCESS in 1h 11m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/60f46c52a33942c7938982307aaee962 : SUCCESS in 1h 22m 35s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a6fca56d751044af8acb65a79d415a3a : SUCCESS in 1h 05m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cf0dff11c9c1469693054ea6cfd7ec8b : SUCCESS in 1h 49m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7339a9175bd146f6982b30432de4dcb5 : SUCCESS in 19m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/56e80c08831b4c28b728998717197710 : SUCCESS in 21m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f2cf3b222f62480b8d42c1ccb38f5ed0 : SUCCESS in 53m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6573ddbab6b64aa583e9d537695261cc : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc7565859004c858731bb8d90471c21 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5bc32280c71491cb21b17e3ad21e039 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10e3f76e203141a4ab3ef0fbe68120a9 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c23a9a9619224a5dbf28c2e6ca1a17af : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ed30aa2a6643d99a5a764d556c303d : SUCCESS in 7m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8070d88e7f3a481590396f9cd0d2b2c4 : SUCCESS in 26m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e8b558e750346b1b9a111b042e7b212 : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c0c5faff30b4a0f998cf2c058cf1e98 : SUCCESS in 55m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dbb2b2a4e67d4ed189c1297e05e79879 : SUCCESS in 1h 01m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5819d658ac74f2ab085b35cd5fb3662 : SUCCESS in 1h 04m 03s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"517f06f4295d43dfad99bd4320ac4d0a5bd32cbc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-19 13:33:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a9c2a8b02c40c2cc6af71b61d56f130fbf0c334d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:33:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b45b1d78f4e69274deb1c13b6291503d5d8a6c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 15:34:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1a9bbfe5a45de593e4e6fe5a635fe77ffc1f826c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 15:34:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dbe35afa17e54a4f9d88ca03f7ed05fe : SUCCESS in 1h 15m 00s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/939173b9d7c84678a8b8b60a553cbc7a : SUCCESS in 57m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c4d122825b414fadb3a00cc8a76eaaa3 : SUCCESS in 1h 25m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1e7b2456e45b42bb85067533bf60e9e0 : SUCCESS in 1h 58m 20s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9f97569863f643f5b6c4a5c5d719d763 : SUCCESS in 1h 38m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f1796e1dac4429a9336c587da6acd58 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69561b237a64440598f104c1e14fe10f : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b0c5d0970c049d9a8175348075ffaae : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/555b3d7e49434146ae862b27da7ba345 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/525ed29b5f09416b9bcddb9d07c39b85 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d6f74e02c4434dac4ce5ccd39f1a46 : SUCCESS in 9m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/624b34affd814690a2461d3b02030801 : SUCCESS in 31m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76ca7608d21444c5a95cbf425479d207 : SUCCESS in 1h 04m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9741178c54cf4f01a0b5b5be14d05519 : SUCCESS in 1h 18m 19s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"8d30875a4c9510960a891933432f1c2df6a8f2fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 15:36:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7291f627a96a450eb21f1b91f4259a25 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c63cd16867c6bbfb86cc3902e0c601ad1842f3d7","revisions":{"5639e4b7b44237e82b37a6857063009944a62fea":{"kind":"REWORK","_number":1,"created":"2020-06-16 09:44:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/41/735841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/735841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/735841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/735841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/735841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/735841/1"}}},"commit":{"parents":[{"commit":"47df05074204b87805986c7e9ae6743e3dce28db","subject":"Bump versions for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47df05074204b87805986c7e9ae6743e3dce28db"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:54:52.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-16 09:44:38.000000000","tz":0},"subject":"CI: use wheels from infra mirror","message":"CI: use wheels from infra mirror\n\nIf they are available then we do not need to build them. Especially\nAArch64 jobs will speedup (once wheels are built).\n\nChange-Id: I79af6c37950e156018a9204fbcc7417cd7d41012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5639e4b7b44237e82b37a6857063009944a62fea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5639e4b7b44237e82b37a6857063009944a62fea"}]},"branch":"refs/heads/stable/train"},"c63cd16867c6bbfb86cc3902e0c601ad1842f3d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-19 10:51:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/41/735841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/735841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/735841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/735841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/735841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/735841/2"}}},"commit":{"parents":[{"commit":"729bd45077e5bbbc4c23ce873eeb6768c8cf2a40","subject":"CI: Install python deps, fix mod_wsgi in cyborg-api, disable masakari-monitors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/729bd45077e5bbbc4c23ce873eeb6768c8cf2a40"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:54:52.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-19 10:51:11.000000000","tz":0},"subject":"CI: use wheels from infra mirror","message":"CI: use wheels from infra mirror\n\nIf they are available then we do not need to build them. Especially\nAArch64 jobs will speedup (once wheels are built).\n\nChange-Id: I79af6c37950e156018a9204fbcc7417cd7d41012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63cd16867c6bbfb86cc3902e0c601ad1842f3d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63cd16867c6bbfb86cc3902e0c601ad1842f3d7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
