)]}'
{"id":"openstack%2Fkolla~735842","triplet_id":"openstack%2Fkolla~stable%2Fstein~I79af6c37950e156018a9204fbcc7417cd7d41012","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I79af6c37950e156018a9204fbcc7417cd7d41012","subject":"CI: use wheels from infra mirror","status":"MERGED","created":"2020-06-16 09:48:21.000000000","updated":"2020-06-26 15:53:29.000000000","submitted":"2020-06-26 15:51:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735842-1593186708844-c88092e0","meta_rev_id":"fdcdf21ee2ab303acd1177102b86eade29175b9e","_number":735842,"virtual_id_number":735842,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 15:51:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-26 10:50:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-26 10:19:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-26 14:51:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 10:50:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-26 14:51:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-26 15:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75b3f3403bd7c9036334a8ab9497cfb7e228bcac","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-16 09:48:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10499619899128dee35225476639fb146f2eb36e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 10:06:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a8eeeb685449e233db08e8c66a28c50f2488834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:44:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3f450097883144988425a28c01a817b8 : RETRY_LIMIT in 4m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d1f71a39122e43a58dc4290de60ed53d : RETRY_LIMIT in 4m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/132308daf15f46f1bc3cd91ecf652e37 : FAILURE in 5m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/30f93d05d8e44de78a834938eaafd956 : FAILURE in 4m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4cc54db1a725469f9c6e92706ace75c5 : FAILURE in 5m 56s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/770011d7d7d74bd7ad71da2cee04adc4 : RETRY_LIMIT in 3m 33s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f0dcf46ec6754d5d9d6ff0c9876e1d49 : RETRY_LIMIT in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b5ceaf25d83481ca46c03704ad47403 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/279e6fd2ec2f4927b365173be44cc94c : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7b11bc574f545a29251319a054e887e : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b930860980b4ed5a647f5d1e67a00b2 : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80c2f637c4f241a0b912059d5949fd0c : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d000fa1fbb2043b6863af726c1a29dee : SUCCESS in 6m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/770a912bcea54ca3934a7166d920db73 : SUCCESS in 26m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3497558acddb46b5b735111517e46e7a : SUCCESS in 23m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2ac739ca2f44d5d90d871873272758c : SUCCESS in 1h 12m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7624200107b24db99104539a44cd7777 : FAILURE in 34m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cef998e51a89454f9bdeaefe9e160922 : FAILURE in 5m 04s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"8b4098e9fa73a52e1eb6c97d585b4c07361bd0c0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-16 21:53:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8250be69a47de72daee552c3a39d0329ecc57df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:22:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/453b642d0b2843a2a1c48082d5230615 : RETRY_LIMIT in 3m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3144163b04fd4263826e8d0d71e487e0 : RETRY_LIMIT in 3m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/512964ba34d14f1794b916e48cdf0fa3 : FAILURE in 5m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/08e9ca4e7725409eb53715b59c2a8e6c : FAILURE in 5m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/210bf96f5f2041eaa8ab1394e22bbece : FAILURE in 6m 01s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/f5061177e32741108fdbff0cb37a6efa : RETRY_LIMIT in 3m 05s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/481042a1c2554cf5830df063ce6aaa36 : RETRY_LIMIT in 2m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8c5a6ee35f74a60879ecdff5990f0d0 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a626bbfde24b689b61871bca22acf2 : SUCCESS in 8m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8054e4f9a1140938ca419d821ce4678 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6597d92d40a44008b9d82ff2829b3da : SUCCESS in 11m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e30b4007aa6541cda980a23cf929a129 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d86951a11864d579a18efe4384734dd : SUCCESS in 8m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f471485e7ebe43adb9f30a60ff3bc28d : SUCCESS in 1h 11m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d631e007a3442e0a955af1f962453b9 : SUCCESS in 1h 03m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/728bd21a85d04520bd1dd5ac92ad1777 : SUCCESS in 1h 04m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80cfaacd09b949b7976f5fcb817c2644 : FAILURE in 34m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d7aa5765069c4231808ac38ecbf6d25d : SUCCESS in 1h 02m 10s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"0b0a4159dbd0ac72ca05669c6dce81f193a10c07","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-18 10:58:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"48288c72dc2a789d6590998b3be59e4548683c5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:14:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d2136401fe0c4a218cee9cecd8263a5d : RETRY_LIMIT in 3m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/246c6520fd9046168bc14f67277a29d3 : RETRY_LIMIT in 4m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f3f4f60c50724159ab1f73416fc58e83 : FAILURE in 6m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/308caf5ae41946ebb3d3a4288b1a466f : FAILURE in 6m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cad671d83752464da546989a1afb1d60 : FAILURE in 6m 41s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a4a4babc39d147e99eb36e9dc731bed6 : RETRY_LIMIT in 3m 24s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/128cf1db5dcc4558b1e1d027d5e85599 : RETRY_LIMIT in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/477161cb8c19446b9e0eb711bbeb5f69 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d050bd007c364d828b0800ce3126d83a : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d75d969c17549489e275a0f90bc03bd : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6911f955bd8c4706b63949b7e64ec272 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83e3a7b2e20d41d0b25e300cff897d11 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b1dcced4cb4b9fb00e92bb6d467fb3 : SUCCESS in 6m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f02ee9dba97743629b7f12214bfc4d68 : SUCCESS in 28m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cfa0efb9797a4cc79af598bc625a03bb : SUCCESS in 28m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d17755501874e58ab05308f5481acf6 : SUCCESS in 1h 01m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f08dc577b31842daa4568180ba43bc4d : FAILURE in 39m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f7fc79435b7c485eb4f6ad07d4584cc8 : SUCCESS in 1h 06m 49s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"555af4a8f30cd6e0aa746ea1ae2cfc2fe738d10e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-26 10:19:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"32fce4701e254376f1cbd9c5940720608c9842e8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-26 10:50:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2464effb71bbc54d933bfd12d7823cd4c4abee80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 11:45:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c4b96f03fbdb4e77a50a771c722dee98 : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b6614e468a23441e9d50d4a85e42331c : SUCCESS in 1h 11m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cea9f58d215343bbb65f382da53c8165 : SUCCESS in 1h 22m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/89799bc10b034d2680da6bb3f15b1932 : SUCCESS in 1h 03m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/68216ab8c9f14951802f73cbd43ff3f2 : SUCCESS in 1h 22m 24s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d10821318f7249b3a799bf7d579bdf34 : SUCCESS in 47m 16s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/69a373bc22d44fafa3967d07e3d996b1 : SUCCESS in 50m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e08f13a499c48ab81b0500a9c7c1fca : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db6bc4eac834c929c7c15a27f4c2715 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58a291d674b346d5868f07c39ff77e86 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3d935a13e4147f3bae2949cad0f72db : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7c8775b976345aa9bbd074841ce9d84 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8447d2d968d846c4a02a08fb95f3a355 : SUCCESS in 4m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/062f028086e04184bf258f9d8c419db6 : SUCCESS in 22m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/820a9c7cbe5c41739348f990066ca592 : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f60bf02e774f4e31b08b4b4bc86ee4f5 : SUCCESS in 1h 11m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7179047d6244e7e82f7287766071899 : FAILURE in 36m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a832850a13ae4737ac7d2aca5b0e3f0d : SUCCESS in 1h 01m 20s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"2464e00b47b9631a2d75e43d213dd8870c18aa9e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 14:51:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"65a18a3e3e560410fc4c31edd613ac79bf253d50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:51:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a73f4152d65ed91ef50172e0d42ca2aa949fde29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 15:51:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/eb3287b60b554aa293c5b67d0c430ef9 : SUCCESS in 48m 48s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7ae1114f9f11449c826d21cc2e08ece6 : SUCCESS in 54m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbcf4687a2ab4ee492466668b3608401 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f5ed0372c547a994e2c05aa6f146e6 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ac11928773a46d8be7b7e15a6325caa : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7590616dda8f48efa5441bd4d7dcfe7e : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1dd029d2e02c466f8ae3058f2bc254cf : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ea649c36e149f4af47b320e5d77294 : SUCCESS in 4m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/446903d3943c43aca8b14c338f535990 : SUCCESS in 35m 38s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"ba5d4fd55e96cdde3833eedccbc7deb95edaf58c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 15:51:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fdcdf21ee2ab303acd1177102b86eade29175b9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 15:53:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa69ae4694834ef38f4937881b21fcfd : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6604041eb365952e269a518cd7a395d6f2b418c2","revisions":{"ae17236aa0112ec7c86004dc0e90b7f2e41caaa5":{"kind":"REWORK","_number":1,"created":"2020-06-16 09:48:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/42/735842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/735842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/735842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/735842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/735842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/735842/1"}}},"commit":{"parents":[{"commit":"8cce9293a42afca7a99f32090d4f55b1cf7310a7","subject":"Bump versions for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cce9293a42afca7a99f32090d4f55b1cf7310a7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:54:52.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-16 09:47:42.000000000","tz":120},"subject":"CI: use wheels from infra mirror","message":"CI: use wheels from infra mirror\n\nIf they are available then we do not need to build them. Especially\nAArch64 jobs will speedup (once wheels are built).\n\nChange-Id: I79af6c37950e156018a9204fbcc7417cd7d41012\n(adapted by hand)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae17236aa0112ec7c86004dc0e90b7f2e41caaa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae17236aa0112ec7c86004dc0e90b7f2e41caaa5"}]},"branch":"refs/heads/stable/stein"},"6604041eb365952e269a518cd7a395d6f2b418c2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-26 10:19:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/42/735842/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/735842/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/735842/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/735842/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/735842/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/735842/2"}}},"commit":{"parents":[{"commit":"d12137ce64fab7e8436aa1c2e32e3bc6ed47c1a1","subject":"CI: Install python deps, mark oraclelinux rsyslog as skipped","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d12137ce64fab7e8436aa1c2e32e3bc6ed47c1a1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:54:52.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-26 10:19:35.000000000","tz":0},"subject":"CI: use wheels from infra mirror","message":"CI: use wheels from infra mirror\n\nIf they are available then we do not need to build them. Especially\nAArch64 jobs will speedup (once wheels are built).\n\nChange-Id: I79af6c37950e156018a9204fbcc7417cd7d41012\n(adapted by hand)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6604041eb365952e269a518cd7a395d6f2b418c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6604041eb365952e269a518cd7a395d6f2b418c2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
