)]}'
{"id":"openstack%2Fkolla~735918","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Icc9d65c4ce22c959145a8d95fc6b823252659ddd","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"Icc9d65c4ce22c959145a8d95fc6b823252659ddd","subject":"cyborg-api: install proper WSGI for CentOS","status":"MERGED","created":"2020-06-16 13:42:49.000000000","updated":"2020-06-18 21:22:28.000000000","submitted":"2020-06-18 21:19:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735918-1592515171548-7e2bdd3d","meta_rev_id":"21f243d3b61fe0be7fa9410c73f8e10fb6d47715","_number":735918,"virtual_id_number":735918,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 21:19:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-16 14:05:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-16 13:43:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-06-18 10:58:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-16 13:43:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 14:05:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-18 10:58:39.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-18 21:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bc78569f003e1caabee954e6dbae0842b0c6444","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 13:42:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6048aff3451da68164ef89cea39d57073e5feea3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 13:43:01.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"6bc075bb7d1f49ce5d29eae73b4a70421c19b6ee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 14:05:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8dcc97a67101260dd7224d9cddb895783644bd6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 14:47:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cac046cfbfc0499aa5a29c5d0a73c7a9 : SUCCESS in 52m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48a1f1270ed70b22ca1a10041432dd2a4ac03616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 15:08:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a63e4689058d48068d6f7946a3769449 : FAILURE in 20m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/581991dbc14b4f3399d84465af646340 : FAILURE in 48m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/10a81b861f7449e5be17c425fac729a2 : FAILURE in 23m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f888c7996d6041b581dcf45604b13ba5 : FAILURE in 21m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2a20d76e851446b9875490a2e57768f6 : FAILURE in 25m 46s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2026a6971299411f98de106534195385 : SUCCESS in 57m 59s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ee119abc95274f409c627b3f3a4ad227 : SUCCESS in 21m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f8df7751dcd8454188d35d6c5f7bd84a : SUCCESS in 26m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/67d3a92e89fe4d95a509425d94682fb4 : SUCCESS in 1h 19m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/613620054d154792910046d00dbd9816 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132c0304c53749e08e437bd76f76b553 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21180b9cdaad46e89d963027e7cd69c3 : SUCCESS in 7m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c1d424fe55a46fcaaec26db4b2ae177 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919b91ece6ea4ae2ad564711c993f52a : SUCCESS in 10m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30cfb32bc8734bb8bfd37cdb63b3cd98 : SUCCESS in 24m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6e09af609a94cca82b2c404455c9d3d : SUCCESS in 28m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c46756f90ba44bef98e9a4c185a9da93 : SUCCESS in 1h 10m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/33e477f6574f41bc80bbe0f62240e0fc : SUCCESS in 1h 03m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8aae4b981cc645d8af5feaf2fceb6afa : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":1},{"id":"77c7ab6760b3b0a68d1ccda4c935a52c3419e4d2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-16 21:45:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ad07c74718e0d95ea076134a37e4140b6e7d5860","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 22:43:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d9ca3dd4f5f4687bf0f3926632cd2ee : FAILURE in 53m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f2a0b107df7a98d98e7c19dacf7ba4b8d60cdf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:41:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c5b60e4cf00a4dd2b76290644519aa5b : FAILURE in 26m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/84ec104d6cef4e9f98eae5442e592079 : FAILURE in 1h 25m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6565399962794c16b25079e047d0a3dc : FAILURE in 18m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/77ba509320a24bc78e03e4461a58ea99 : FAILURE in 17m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b02aeb9064f14dcab7a7f18fe96325ba : FAILURE in 22m 31s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7c45537bf679450f952a2f3a477de184 : SUCCESS in 41m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a76c3d9c46b445b969a2a00577fb266 : SUCCESS in 19m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f9e2c2253a47453a8ac54c4230ce9fed : SUCCESS in 41m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a15fb783fd4a4184a8cf26a21d202210 : SUCCESS in 1h 21m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17563e3eb5cc452c9964a41e720d6016 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c44a1bb439fd4950a5eadd68e67d470b : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de700541d38247bd814d040572f9e498 : SUCCESS in 7m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ea96ac65f4a4394a67a75a8cc21b24a : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b970011bfe147a29d17e907f60acd44 : SUCCESS in 8m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d8b460b865b842d3af28af759e9986fb : SUCCESS in 1h 37m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/30f120412fca458da2d5e0fc48e88925 : SUCCESS in 1h 37m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1fe9953da6c4dd4a285c42073c2c0e3 : SUCCESS in 1h 50m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf8911b5e362407188f80fdd8d2c6150 : SUCCESS in 1h 50m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be0091fce5a74544bd58239d8bbf1ce6 : SUCCESS in 1h 11m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d73a32d3d1648b21e3ce51351c498027ac6d9f93","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-17 08:36:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"df0cee0619298e949562f38c6ffd4b11cb5f076f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 09:50:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3178cffe6fa24b6f8fb45e680a485a9d : FAILURE in 20m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/52f7370619644c3a98444fa99c6ef22e : FAILURE in 28m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e89c512774f24a98987f7e83fa38ecd4 : FAILURE in 24m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e95ae76d45654846923c808dc140c7c3 : FAILURE in 18m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/edff5158cffb4fe09d0cd44932a42bc5 : FAILURE in 22m 47s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2c1991fc594c4d449679ed3d035fbb13 : SUCCESS in 55m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3ceb44741094484d81b799af10c7022b : SUCCESS in 39m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a10b09da09e44b299c3957dba00b68b4 : SUCCESS in 30m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/405617c20c474851a1279e35addb5530 : SUCCESS in 50m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3c7212146c64be7b4f76d53195214a5 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da38866797be4fc79317f02fc9bdbba7 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5865c34f010b4126896f1b9552ecc7fb : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8c85de82b8445f9be56d596a1b09ebd : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6bc1abd224541e2b79bf5a783fd8dcc : SUCCESS in 11m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b36992d5cf424c7cb701f900b679ba6b : SUCCESS in 1h 02m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c7057bfade44cd8be66be8ac8f5675a : SUCCESS in 35m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6636a290081f4f43b6e76efb7fe3a58e : SUCCESS in 1h 06m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b27d9d5e708944db8e40cd2507504696 : SUCCESS in 1h 05m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/045c906a67be4efc942fada602d6c970 : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5751aead831e6c188bb009d371d94dac7aad4690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 10:00:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/535c43f5812a4834a7cc5f4c0379de44 : SUCCESS in 1h 04m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4c2ad70db90410f848ea198f139b87d : SUCCESS in 58m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3c54a8669452b7666455c39e3f4dc800cf7a68c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-18 10:58:39.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"51b31183b27ffd4233650ef09a8e77a06b6eb303","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 11:47:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af3f8c6cab490ac66be6bc7ce35d79cae03b0280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:17:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4226769f1d5e45bca2cd27f46af57cfc : FAILURE in 41m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9478bd86d7754de9ad513e5af3090cc0 : FAILURE in 29m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/77c0d60aa99942bcb2c8f079809a8b6b : FAILURE in 18m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8be8877acb2a41e88462a4cc151796eb : FAILURE in 24m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/196a452fc5ca45d7885c68da418bbfe7 : FAILURE in 22m 23s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/a124e32be76f44f28b5093f05c1d8416 : SUCCESS in 48m 55s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ddd790313be349beadebaad8eb2f7cbc : SUCCESS in 21m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8321151979c341f6a39bfb00bd97cfdc : SUCCESS in 30m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b009f98ba6584ec2a075812c83e2a0d2 : SUCCESS in 1h 12m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dc88265fafe407e9d9561226179e802 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef02a7a39d54bd38cdb1e590b02412d : SUCCESS in 10m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8247082115d94eecbe8fe9a5cfaea8de : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2f3992b6e734a0f9a10138c9039df97 : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/327cf74ee4894fa49331895f6da93014 : SUCCESS in 10m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b92a2f1857454a289c00de7de7ebdf3f : SUCCESS in 29m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e957688066be4566b423b30c649b4553 : SUCCESS in 22m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31eb5614aed7430a940af503640af0f6 : SUCCESS in 1h 10m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83800cbab0ab49b9add8a0c70d5a59f3 : SUCCESS in 1h 04m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ec2ab1ef6bd4c42be88f36b602bdff4 : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ebdf96b52914ab6cac05e3a3ba7fba385fb7ea33","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-18 18:07:34.000000000","message":"Patch Set 1:\n\nrecheck\n\nstale mirror I guess","accounts_in_message":[],"_revision_number":1},{"id":"425f8528b37c8d30b4455dc75373a0e8ff1ff4c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 18:49:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51211ecb5b1ea7e3bc44ee84d6f83369aee349da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 19:43:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/43d44211bf9d4c80b06704bd97c31267 : SUCCESS in 22m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d09b19af06de4304a4ce6fe8d324a697 : SUCCESS in 40m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1b09cffd073c4510a57b8dc275eb5359 : SUCCESS in 1h 23m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d6dc1e5cc6e24f12b0dd5c1ebaeeb4ad : SUCCESS in 1h 19m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/678eb11b9ff64a1797fc6e8485516743 : SUCCESS in 1h 28m 32s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/91a037fc28ff4f6da30c1d3973e85f96 : SUCCESS in 40m 59s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c1098eddc0df4da99761eb3019db9228 : SUCCESS in 34m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f08d92c2c4d846c18bc82d9b239ec74f : SUCCESS in 19m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/925b9814006f4d6a9b61ae47bb9332ae : SUCCESS in 1h 09m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/098c5cc5c8e6477180da7636bc0a437c : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0961290b5e87421caa8fe24cd155af4c : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71171fc921f54c7c923a95c3b84dcfd2 : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca4ad01c59b545f0afc5d80d3f06ba99 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e42e7a291624df5ad3c7dd26032f627 : SUCCESS in 7m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1a89d1fc1c6444d8f1e8da33d6c2781 : SUCCESS in 54m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5964fb1ebd5d40299aba36bd07401dc0 : SUCCESS in 25m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0c2b80d5a514337be8be614a1e00c6b : SUCCESS in 1h 07m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eb024aa9b92a46ffb02f514c43e3b1b7 : SUCCESS in 1h 02m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2886ab18e0254a80a39f0566911f9199 : SUCCESS in 1h 33m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a5d9796faa0d4f1643f548b6d00a6e5764fb0449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 19:43:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"883f8d8c47aa6995ff2c01b6cfc59c69406537c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 21:19:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6ec7b1e310ee41e69058ea4d24972a85 : SUCCESS in 24m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bba0cf777b7a4f66ac72119344b24430 : SUCCESS in 1h 13m 05s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5943851f8a034928a15b18da5808237b : SUCCESS in 1h 31m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fafcb223a3d143f8b793ff2e853c2c50 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6375366606e647a997fb313bba4081f9 : SUCCESS in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b82b75dab1d4e47abb37fcc8ce09507 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3571461bf02425fb7f1627fd2e56200 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b3e890e60de4632817ca2ab54750218 : SUCCESS in 15m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8501725b25440ed895efdcae6f4db3c : SUCCESS in 36m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f5191b2ab1d4aef9c34af60ada1feb6 : SUCCESS in 1h 05m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b8a87ffabc3f477b9ab13c847d5af6bb : SUCCESS in 1h 26m 54s","accounts_in_message":[],"_revision_number":1},{"id":"fef0130ec76646823458aa3d049983c3c169c97d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 21:19:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"21f243d3b61fe0be7fa9410c73f8e10fb6d47715","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 21:22:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f920ccaacca45ab8bea6ddbe82e682f : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d504c1ec9335c020a0f8f6d9a1b8335c5858844","revisions":{"4d504c1ec9335c020a0f8f6d9a1b8335c5858844":{"kind":"REWORK","_number":1,"created":"2020-06-16 13:42:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/18/735918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/735918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/735918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/735918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/735918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/735918/1"}}},"commit":{"parents":[{"commit":"47eecc0180213c7d649243b28bd779eb57ba4333","subject":"Merge \"add libibverbs package for openvswitch\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47eecc0180213c7d649243b28bd779eb57ba4333"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-16 08:11:36.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-16 13:42:49.000000000","tz":0},"subject":"cyborg-api: install proper WSGI for CentOS","message":"cyborg-api: install proper WSGI for CentOS\n\nChange-Id: Icc9d65c4ce22c959145a8d95fc6b823252659ddd\n(cherry picked from commit 14ef775c6ea849964d9afa03575a415687d9b51c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d504c1ec9335c020a0f8f6d9a1b8335c5858844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d504c1ec9335c020a0f8f6d9a1b8335c5858844"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
