)]}'
{"id":"openstack%2Fkolla~736096","triplet_id":"openstack%2Fkolla~master~I9d85f3255047ce445eef7bba6a63fc496bf4a030","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I9d85f3255047ce445eef7bba6a63fc496bf4a030","subject":"Remove the neutron-fwaas image since it retired","status":"ABANDONED","created":"2020-06-17 05:18:01.000000000","updated":"2020-06-30 09:39:07.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"73f1d089bface3527159892fa5b9e6bfcec2dd50","_number":736096,"virtual_id_number":736096,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"date":"2020-06-30 08:30:39.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-29 17:42:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-29 16:42:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2020-06-30 09:39:07.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 19:02:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-20 12:09:51.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-06-29 16:42:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-29 17:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 08:30:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5e27a9e3d75fdb9adc8c995691dd35d5e211c383","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-17 05:18:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8824468eaa3f1a2fb54783137ff974574204a877","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 05:59:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a654243384c2bfe7adadd54246f73bf0fd81b9e7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-17 06:45:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/1de9e7c19d0d4cc89692a76a5fb73f1d : SUCCESS in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9318db4c0d4466a15f2098a4c4b9a4a581ea9f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 06:57:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/dadfc9369c5144aaba04459487c6843d : FAILURE in 17m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8e12a5cacf5c4505a942c5937b6b1f5f : SUCCESS in 27m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ad33c8733231421e834923e885843465 : FAILURE in 12m 45s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c979aabf1b9e4f46a74a58e9f37a0fc6 : SUCCESS in 1h 09m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/795e83007bc84bb6b0dd6e1795b68105 : SUCCESS in 1h 24m 46s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/59ee1722e3ce4eff90220e444a2c2ca6 : SUCCESS in 40m 49s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/83d14d7961014f8aa2daf4dd919d1136 : SUCCESS in 18m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3982a2d83cff4139aa9981c00a0e1624 : SUCCESS in 20m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/daaf9cd676154e41ac55b5b41825828d : SUCCESS in 1h 00m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56feb34b296d4f8cb48e3d77c7253fda : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ec50640c610495d9a60e8a250a211e9 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2f316fd41314b18b5b7f09fe6f4eaf3 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75daba002e8440c4bc5cc6dfad5af46a : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d35ba03a72c420bae3c29a86f60d94d : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ffdc8f1ced6c49c2bd8540a5643632bb : SUCCESS in 22m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3786d771f2ce44f1a51aaaf6862bd9a1 : SUCCESS in 21m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d31263072c0543b7abb67605415e727b : SUCCESS in 1h 09m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8055e92b7d6346d5bf6d5e8c1545f666 : SUCCESS in 1h 05m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/962e048c4c0f4af1b529d11a01e190fe : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7abbd8390c599a4e92753be8d0338ac59857aef7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-17 19:02:15.000000000","message":"Patch Set 1:\n\nreno here too","accounts_in_message":[],"_revision_number":1},{"id":"e96f05b058a87365a44dcc9c174589bc835b2320","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-20 11:30:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfcbc96d89a6bf13c2b11807c62087f7b70f70f5","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-20 11:44:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"198371d934e344e276d779dbb769b88c09e3c5dd","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-20 11:45:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce1a551a348299b1b0895b7fedec1d0dac48628a","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-20 11:46:48.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"fbd8e76068b3990558313ed64cf30787338dd815","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-20 12:04:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ed1253c97a8d3e4705247f20229310f590bf0270","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-20 12:09:51.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6a50b1e9fc74f816b9145e1ba69ed74638bffdb4","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-20 12:11:43.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"70966cebc745e9165412bd8803d53bfb496ce027","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-20 12:42:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7c8d26cf178148ea991cffa896f15a67 : SUCCESS in 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"05a81069c00ebe286bb19becfa8e498690377af9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 13:01:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9e25735ffd03c7008ce1220d674abd6e9a9a1b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 14:25:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/823ce0236a114d28b4b4a601304cc725 : SUCCESS in 59m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6af2572d95d647bda3e4089162cf1ea1 : SUCCESS in 59m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/53d6d9bc2982465393c922cc2bbb476c : TIMED_OUT in 2h 05m 15s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/dd4d2d896a46448eb2d8a823c086379b : TIMED_OUT in 2h 03m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/25fc7271efb04cc1a40658ac7c532a5f : SUCCESS in 2h 12m 53s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/125e6848ba3e4a11bd576f797b33b7e9 : FAILURE in 1h 49m 27s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/af3bb41faa61482e97e2f53f6a83d0e8 : SUCCESS in 2h 08m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/351028eec5eb434c92a57fa4d1dfcf9a : SUCCESS in 1h 30m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/453b27946a97432f83ff1413068af0ef : TIMED_OUT in 2h 02m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac2eece2404546578d3faaf22ec94668 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84fe0e33543453293f538240a21d19b : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a41865bc7544c1b8d776441423e45dc : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68e7601ac4c34691a3c9bf3ebe121354 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd9e5dcebdbf409da4733b94907da2fc : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbaaf89349204dfea6a30c4d663b3a0d : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d325816667b4878a5722a655093fab2 : SUCCESS in 1h 17m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7afe9209a6043a2a56bd4e6ceaf578a : SUCCESS in 1h 46m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/924909ce9e964263bb7cbfd7fa2c126b : SUCCESS in 1h 15m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1f4b40ff6536400185688f158472eaa7 : SUCCESS in 1h 54m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a98bc5285b7b45a8a772d815f5d66132 : SUCCESS in 1h 46m 40s","accounts_in_message":[],"_revision_number":5},{"id":"8027a5a2990a2e03dd8a30a4b7b002eb8a331dca","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-29 16:13:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"dc8175e99e4f0d5b03a4447cabda6bf0529eb3a7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-29 16:42:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/4a9f8852aca1402781ce2184192eede6 : SUCCESS in 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"511ac48cbbe279467b75d557b974a5ff8e91286e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 17:42:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e710c0578ba140169f608d33017f3639 : SUCCESS in 42m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/25bc52120d92410d9d30441ec83498fd : SUCCESS in 32m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/189d151ac1a84ae8a7e27c75ca8012a0 : SUCCESS in 1h 10m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/04b17a197c6c400f9e747c61859327bf : SUCCESS in 1h 08m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c62f3384771d475f85b2b19a0863a872 : SUCCESS in 1h 28m 03s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/51093c999b594561b4e91aa2a04d03b7 : RETRY_LIMIT in 8m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f78b24afd4074b22b56a5de7f1e6d5e1 : SUCCESS in 23m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/60c14b3b4f534b59b7a932018bcfb765 : SUCCESS in 24m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e005571cafb345c5bee673722a871666 : FAILURE in 1h 07m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4cd99d9ee7849e0afe0275d76cd5057 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dacf50cb104c432c892f32b74f3818ca : SUCCESS in 9m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bad6d7e3b4946d69bb7955796f8f290 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8a152051b2c47529e46cc413abdd6d2 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f1a35a5bcd48c29c5d158da900a890 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1211382f430f44bbb9fa54866a2c9a34 : SUCCESS in 5m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/039ff276188943e996e73210550b1f53 : SUCCESS in 25m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc31e74c7c4e4c8b9cfc77fd70dfdcb6 : SUCCESS in 23m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0253c9a99fc4234934927cf517c4f9a : SUCCESS in 55m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/770148a730da43e18d6271658374ffcb : SUCCESS in 1h 03m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7cb8d6fac3b14cd187ed36417301b7d4 : SUCCESS in 1h 14m 05s","accounts_in_message":[],"_revision_number":6},{"id":"e0b1e4645320a2be6439a0ecf71a04dbff457c7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 17:56:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d3495edc8a1469580f6a821f0c13bd4 : SUCCESS in 1h 25m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f3b911f5900e472d82f2bd43d82fe65b : SUCCESS in 1h 15m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/94adcf3ae01c44c2b2b0a7915bf86661 : FAILURE in 1h 35m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7391bd0011d462a9708838628eee904 : SUCCESS in 1h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"849377197985c7bfc24b1d79e53f1e9865d0d4df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-30 08:30:39.000000000","message":"Patch Set 6:\n\nThanks looks like this one was approved first: https://review.opendev.org/#/c/736727","accounts_in_message":[],"_revision_number":6},{"id":"0034603a7aaa9fb18d133be070f024ba3f3c9c57","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-30 09:39:07.000000000","message":"Abandoned\n\nto avoid confusion","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2af039da868252abdebf19d326e329011264b937","revisions":{"ef6cab22b18e063ad88ffd12bf066081f852c156":{"kind":"REWORK","_number":1,"created":"2020-06-17 05:18:01.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/96/736096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/736096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/736096/1"}}},"commit":{"parents":[{"commit":"c07409bfabdd8135316609160e6477d202b46b95","subject":"Merge \"Revert \"[to-revert] Pin ovsdbapp in neutron to 1.1.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c07409bfabdd8135316609160e6477d202b46b95"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:16:03.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:16:03.000000000","tz":480},"subject":"[wip]Remove the neutron-fwaas since it retired","message":"[wip]Remove the neutron-fwaas since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I9d85f3255047ce445eef7bba6a63fc496bf4a030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef6cab22b18e063ad88ffd12bf066081f852c156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef6cab22b18e063ad88ffd12bf066081f852c156"}]},"branch":"refs/heads/master"},"2a433d1e56ad683b3397570e73b049766044debf":{"kind":"REWORK","_number":2,"created":"2020-06-20 11:30:11.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/96/736096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/736096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/736096/2"}}},"commit":{"parents":[{"commit":"e364113adcb529cc06575133b0e377526516f715","subject":"Remove mongodb dockerfile from kolla","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e364113adcb529cc06575133b0e377526516f715"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:16:03.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-20 11:28:56.000000000","tz":480},"subject":"Remove the neutron-fwaas image since it retired","message":"Remove the neutron-fwaas image since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I9d85f3255047ce445eef7bba6a63fc496bf4a030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a433d1e56ad683b3397570e73b049766044debf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a433d1e56ad683b3397570e73b049766044debf"}]},"branch":"refs/heads/master"},"bc85c723a4ef09c0898d797a052d0d465935e116":{"kind":"REWORK","_number":3,"created":"2020-06-20 11:45:50.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/96/736096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/736096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/736096/3"}}},"commit":{"parents":[{"commit":"e364113adcb529cc06575133b0e377526516f715","subject":"Remove mongodb dockerfile from kolla","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e364113adcb529cc06575133b0e377526516f715"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:16:03.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-20 11:44:40.000000000","tz":480},"subject":"Remove the neutron-fwaas image since it retired","message":"Remove the neutron-fwaas image since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I9d85f3255047ce445eef7bba6a63fc496bf4a030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc85c723a4ef09c0898d797a052d0d465935e116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc85c723a4ef09c0898d797a052d0d465935e116"}]},"branch":"refs/heads/master"},"2b53988ade5cada9e0df382bbcc6154bb28bc686":{"kind":"REWORK","_number":4,"created":"2020-06-20 12:04:22.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/96/736096/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/736096/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/736096/4"}}},"commit":{"parents":[{"commit":"e364113adcb529cc06575133b0e377526516f715","subject":"Remove mongodb dockerfile from kolla","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e364113adcb529cc06575133b0e377526516f715"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:16:03.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-20 12:03:21.000000000","tz":480},"subject":"Remove the neutron-fwaas image since it retired","message":"Remove the neutron-fwaas image since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I9d85f3255047ce445eef7bba6a63fc496bf4a030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b53988ade5cada9e0df382bbcc6154bb28bc686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b53988ade5cada9e0df382bbcc6154bb28bc686"}]},"branch":"refs/heads/master"},"c6ef363d3fd65c977068cb33c5bcd612695999f6":{"kind":"REWORK","_number":5,"created":"2020-06-20 12:11:43.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/96/736096/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/736096/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/736096/5"}}},"commit":{"parents":[{"commit":"e364113adcb529cc06575133b0e377526516f715","subject":"Remove mongodb dockerfile from kolla","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e364113adcb529cc06575133b0e377526516f715"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:16:03.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yingjunz@gmail.com","date":"2020-06-20 12:10:10.000000000","tz":0},"subject":"Remove the neutron-fwaas image since it retired","message":"Remove the neutron-fwaas image since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I9d85f3255047ce445eef7bba6a63fc496bf4a030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6ef363d3fd65c977068cb33c5bcd612695999f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6ef363d3fd65c977068cb33c5bcd612695999f6"}]},"branch":"refs/heads/master"},"2af039da868252abdebf19d326e329011264b937":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-06-29 16:13:05.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/96/736096/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/736096/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/736096/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/736096/6"}}},"commit":{"parents":[{"commit":"5b4e00decc28512136cdeb27404b2febb62d8dc7","subject":"Merge \"Change lang package block order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b4e00decc28512136cdeb27404b2febb62d8dc7"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:16:03.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yingjunz@gmail.com","date":"2020-06-29 16:13:05.000000000","tz":0},"subject":"Remove the neutron-fwaas image since it retired","message":"Remove the neutron-fwaas image since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I9d85f3255047ce445eef7bba6a63fc496bf4a030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2af039da868252abdebf19d326e329011264b937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2af039da868252abdebf19d326e329011264b937"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
