)]}'
{"id":"openstack%2Fkolla~736123","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0","subject":"Install Fluentd Prometheus plugin out of the box","status":"MERGED","created":"2020-06-17 08:02:18.000000000","updated":"2020-06-26 16:27:05.000000000","submitted":"2020-06-26 16:23:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736123-1593188626241-a5e95a55","meta_rev_id":"6bc1f79e0a42dc05d6b55e0883ac71103ae080ee","_number":736123,"virtual_id_number":736123,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 16:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-06-26 13:02:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-06-26 13:02:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-26 13:02:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-26 13:02:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-26 14:55:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 09:14:29.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-19 10:05:15.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-06-26 14:55:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-26 16:23:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98df90445ce08feb590964907a6ac125a1ba74d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 08:02:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1053dd8020f273069d58764d00cc93ed364cbf19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 08:44:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45e84afe4080f3e0086153656c3b0bd2dd3b1cb6","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-17 09:14:29.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks - this will be useful.","accounts_in_message":[],"_revision_number":1},{"id":"8b5516c28aa67418126939c6ea2243de52f548f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 09:43:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a8397cc76626479bb7809032c302aef7 : FAILURE in 43m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/51ad7a93f7674671aeabd7a3042bf06c : FAILURE in 26m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d9ba176c059547e0a530fa1ecdaf1504 : FAILURE in 18m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4f0f1f458157479994d64a82ea3ccd9f : FAILURE in 17m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/191ebdbe1fc944b8b3db98019b7dcb34 : FAILURE in 17m 27s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/8fa647e1d29e4334ab9d848a2ca0f414 : SUCCESS in 45m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/846d35657e6f40bb9a5f495d7ebb7481 : SUCCESS in 22m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/981927425c6d4403a8937fd40d7a9a1e : SUCCESS in 22m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fb8bba495fd0405e82191bfab0525781 : SUCCESS in 1h 03m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24663c7ea99a44b5b474be1501a8bc13 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0146492fc424292a3cd2a7749131a0b : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42c100d56b3a4d8ba817a08598e2cbc9 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/136b013b29e04537825303d42bd6510d : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060d2d78570c497c9eb824001c2af93c : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffb43cfbf2014b2382364f252d1951cf : SUCCESS in 6m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a487cbf3b73149af81ce3174a6f7ff54 : SUCCESS in 1h 38m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0759a23d9a624c0896e034b4767ada08 : SUCCESS in 1h 39m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5071d28758b74ca9bf82b6b4bf6f1531 : SUCCESS in 1h 39m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/64f814641a78443a88ab6ede7e6c2251 : SUCCESS in 1h 01m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ebfeb190e19e4a2da1a87e801c5ed3f0 : SUCCESS in 1h 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e11f5c6e7723ed9ce8d98e220bb0c2b60be11e61","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-17 16:01:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a3280ad2dfc5e84e92ee53e2f5f9d4008e35fcb6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-18 22:06:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"83f011b193ba4c41af8a28b4cc4b9bf1aa672c7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 23:35:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b22fb173b9f34875af45ddf9acba9d83 : SUCCESS in 58m 29s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0fd8f09e4661cc80632bc1477ed655245cca6715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 00:02:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b92a8a12edb84e66b109a3c092078aea : SUCCESS in 29m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/03b58a309e674aa68abb3cf31959d912 : SUCCESS in 26m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7d949829242b4fa796e71d594646c41d : SUCCESS in 1h 26m 24s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ec5b0b26dc4d44e28c0551b961f70762 : SUCCESS in 1h 15m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ef01d0f5740f4fcb8ab92fa759ea1f3e : SUCCESS in 1h 19m 08s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/60ced88bf2c74584bf7ba2de7701155d : SUCCESS in 51m 22s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c6a8ef912c0b427fa575d9a3d574452c : SUCCESS in 22m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9b6b9b91e99e4dd7826d835d6d1fdcaa : SUCCESS in 26m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2354939c0a4240cbbc387081e798c256 : SUCCESS in 1h 05m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fec537855d3042f0b302f41f7d8a8ee7 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f9677b7f0e74cd28a3fdd909300d00b : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8e650fe673c4225bb022737bee5e170 : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8b1c008fb3245f1891a3b40df664ac6 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9aab91315d64db6af063809986f63ba : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79079b305c87430cb376f1a05bbd5d74 : SUCCESS in 7m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d9219ad5b56847b5a6718dbcc63b9c75 : SUCCESS in 34m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32883fc6486c4242bf8d82629f561f63 : SUCCESS in 44m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5b9392a063b48829b1d2a383f20605e : SUCCESS in 1h 11m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1533a58d8b14a8fb82b401b55e9f28e : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ace7cd65560f4976ae3e680111d16061 : SUCCESS in 1h 53m 39s","accounts_in_message":[],"_revision_number":2},{"id":"cb3043cfc00fb3f0f7414ebb4ac3070cdc0bb155","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-19 07:46:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e07bb44bc90f1a9a7acf602f7b9e7ea409e95789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 07:47:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b43a4ae4134d583ac2e99ad1eae3e9ae167a1320","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-06-19 10:05:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"52433240b0f814c627efb5cdb3d092cba376996b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:57:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c4c3cb811cb944da81bbc418f29e2745 : SUCCESS in 29m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/630af786577e40078d424fda9714c602 : SUCCESS in 1h 17m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0d927559f66c48398278bfd26d92f262 : SUCCESS in 1h 21m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0f89990b4864b04ba3c968fee88f867 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b5cd2d66364bbdae6575943f50e18a : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b81603e3813842468be78955419a0cb6 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75a1888050774a568ac44c6647d467b5 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb929de0453c43b58365f7374d40c4b1 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acb2ff8f7c32443bac07c28c365a9369 : SUCCESS in 8m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0667945b3cff4b83af4e7549cc11c9ed : SUCCESS in 40m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a83eba50cab4c9e9cbf66e74e75e8fe : FAILURE in 45m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1497c85258954e558d5d65dbe84405e8 : FAILURE in 1h 14m 48s","accounts_in_message":[],"_revision_number":2},{"id":"0f9ac5a302ed28b6b096cd341f6bc4566cf8cfd7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-26 13:02:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b14a600e925559b22e4d6d403ab8eed34fc9d067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:35:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e9144e65f6fe4ea6ab21f5baf7200594 : SUCCESS in 23m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e242688ebd834a6c8f803b3cb3480451 : SUCCESS in 27m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4233b831dad94aeba30172999ac1efd5 : SUCCESS in 1h 22m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f8db0a8bf3f2430fa1fbf4314885688c : SUCCESS in 1h 19m 39s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6e0f2fa4185a4f2cb140a4ba5aa897df : SUCCESS in 1h 27m 18s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/09d9bde270a844cb9e3f123d8b4984e0 : RETRY_LIMIT in 9m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f41619231af24a408110d1e10207be92 : SUCCESS in 33m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f527ebefadc442958e66cc3930510f45 : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf1c89d85f594d7ca281ae66ea4af8ab : SUCCESS in 1h 04m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfabd4f4c55e4099bac68363018d7971 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f09d1674136496387cbdd8643d4953a : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53ae928c742740cd81d3af53450d5f71 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4657f9b710c64f6983fa55ffeac482fe : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b721f1653e748fc92db2f9df0351e93 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/495bdd0ba58644b7a89a224d93e34067 : SUCCESS in 8m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d32c125c0b144fdd90fe354f9d47d805 : SUCCESS in 31m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9de30f5bd2cb448a9abe1e72a13ac3dd : SUCCESS in 35m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d58c02293c249e9a1fb42447f584977 : SUCCESS in 1h 08m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a671f5c74d9473883e06ee5fbd0ffc7 : SUCCESS in 1h 01m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/efbfe0949556441598f5e6e5a00d6866 : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":3},{"id":"4b92af18c46de8595d915359733245a80ec7c2e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:54:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a9a41a8e9d04278b3431cc2f2a3d257 : SUCCESS in 51m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9de29d6b9fd54b41a6303ad095c2296e : SUCCESS in 55m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb51b2ed74424829adb9498e86223619 : SUCCESS in 1h 48m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e68619f682934c97b0a470211592c6f0 : SUCCESS in 57m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5ad4ab0ed9fdfd0e683c1acb95571a4b56b38d06","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 14:55:40.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1b2c179aa7471101c47b6b1028577feb84d6ded3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:55:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"312279d1311e659cfa2f8cce6aa5365826c7fd6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 16:23:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/311f13b462a04c3f8c0c30e7cd7fdd8f : SUCCESS in 26m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/10f65fb4243d4eb89148ea2e5c7e29a4 : SUCCESS in 1h 16m 58s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1c8098e41a13412a8537d48e0f6b2442 : SUCCESS in 1h 21m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a958224b30845fcadd8b1f351b7a1b4 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29ff0bfbaa2941a99972adf23549095b : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b1e7d5ce5bc43079d19af5a4c43f32b : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b8c41cd3e1c4267b5be569a13dfa2c7 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d81973b66c46088963614312cfa76c : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b475b5a8eda9493bbc58ef50ae4be887 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/98b8014af27e40a08e98c73c58ec8cdd : SUCCESS in 41m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a93017a5faab49eeafd092c68311df00 : SUCCESS in 1h 01m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/87afcceab99a4540bf79beb3c74ecadc : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":3},{"id":"de4fb9a242177ce6a7ca7624f578882a90e094dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 16:23:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6bc1f79e0a42dc05d6b55e0883ac71103ae080ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 16:27:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf8c33502667472880e272d38aa22534 : SUCCESS in 2m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab81f982ebc64da1ab4fa10f32ca66ed : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e0175350ae7785becb0807f230a755b0712e624a","revisions":{"b040d72ac167493aa11f63455c01f6c26193f05b":{"kind":"REWORK","_number":1,"created":"2020-06-17 08:02:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/736123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/736123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/736123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/736123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/736123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/736123/1"}}},"commit":{"parents":[{"commit":"47eecc0180213c7d649243b28bd779eb57ba4333","subject":"Merge \"add libibverbs package for openvswitch\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47eecc0180213c7d649243b28bd779eb57ba4333"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-02-05 10:20:14.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-17 08:02:18.000000000","tz":0},"subject":"Install Fluentd Prometheus plugin out of the box","message":"Install Fluentd Prometheus plugin out of the box\n\nThis plugin [1,2] supports a number of use cases which are likely\nto be useful out of the box to users enabling either Monasca or\nPrometheus:\n\n1. As a system admin I want to monitor the status of Fluentd\n   to ensure that it is functioning normally.\n2. As a system admin I want to know the time to response for\n   calls made to endpoints defined in HAProxy.\n\n[1] https://docs.fluentd.org/deployment/monitoring-prometheus\n[2] https://github.com/fluent/fluent-plugin-prometheus\n\nChange-Id: I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0\n(cherry picked from commit e20fe5eaa5eb6c46f815ce7295ecdcf67a4cd042)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b040d72ac167493aa11f63455c01f6c26193f05b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b040d72ac167493aa11f63455c01f6c26193f05b"}]},"branch":"refs/heads/stable/ussuri"},"35173dd564428a1f17b1f89ffc6d9da388f2c90e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-18 22:06:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/736123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/736123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/736123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/736123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/736123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/736123/2"}}},"commit":{"parents":[{"commit":"4d504c1ec9335c020a0f8f6d9a1b8335c5858844","subject":"cyborg-api: install proper WSGI for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d504c1ec9335c020a0f8f6d9a1b8335c5858844"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-02-05 10:20:14.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-18 22:06:23.000000000","tz":0},"subject":"Install Fluentd Prometheus plugin out of the box","message":"Install Fluentd Prometheus plugin out of the box\n\nThis plugin [1,2] supports a number of use cases which are likely\nto be useful out of the box to users enabling either Monasca or\nPrometheus:\n\n1. As a system admin I want to monitor the status of Fluentd\n   to ensure that it is functioning normally.\n2. As a system admin I want to know the time to response for\n   calls made to endpoints defined in HAProxy.\n\n[1] https://docs.fluentd.org/deployment/monitoring-prometheus\n[2] https://github.com/fluent/fluent-plugin-prometheus\n\nChange-Id: I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0\n(cherry picked from commit e20fe5eaa5eb6c46f815ce7295ecdcf67a4cd042)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35173dd564428a1f17b1f89ffc6d9da388f2c90e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35173dd564428a1f17b1f89ffc6d9da388f2c90e"}]},"branch":"refs/heads/stable/ussuri"},"e0175350ae7785becb0807f230a755b0712e624a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-26 13:02:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/736123/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/736123/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/736123/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/736123/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/736123/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/736123/3"}}},"commit":{"parents":[{"commit":"11903c833687449cc05a7bc5c99c6417af4d05d6","subject":"Fix loading of Storm on Centos8 / Python 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11903c833687449cc05a7bc5c99c6417af4d05d6"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-02-05 10:20:14.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-26 13:02:10.000000000","tz":0},"subject":"Install Fluentd Prometheus plugin out of the box","message":"Install Fluentd Prometheus plugin out of the box\n\nThis plugin [1,2] supports a number of use cases which are likely\nto be useful out of the box to users enabling either Monasca or\nPrometheus:\n\n1. As a system admin I want to monitor the status of Fluentd\n   to ensure that it is functioning normally.\n2. As a system admin I want to know the time to response for\n   calls made to endpoints defined in HAProxy.\n\n[1] https://docs.fluentd.org/deployment/monitoring-prometheus\n[2] https://github.com/fluent/fluent-plugin-prometheus\n\nChange-Id: I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0\n(cherry picked from commit e20fe5eaa5eb6c46f815ce7295ecdcf67a4cd042)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0175350ae7785becb0807f230a755b0712e624a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0175350ae7785becb0807f230a755b0712e624a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
