)]}'
{"id":"openstack%2Fkolla~736152","triplet_id":"openstack%2Fkolla~master~If6663b2f6dfadfd6a6db5e4aaca4eb782b87161f","project":"openstack/kolla","branch":"master","topic":"bug/1876461","hashtags":[],"change_id":"If6663b2f6dfadfd6a6db5e4aaca4eb782b87161f","subject":"Fix loading of Storm on Centos8 / Python 3","status":"MERGED","created":"2020-06-17 10:09:56.000000000","updated":"2020-06-25 16:50:04.000000000","submitted":"2020-06-25 16:48:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736152-1593103702273-995c8465","meta_rev_id":"71fd01ee303613067c54e044c37abcf7fe7285b2","_number":736152,"virtual_id_number":736152,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 16:48:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"date":"2020-06-25 14:23:23.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-06-18 22:12:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-18 22:12:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 10:04:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-06-19 07:41:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-06-18 22:12:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 19:24:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-19 07:41:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-19 10:04:15.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-06-25 14:23:23.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-25 16:48:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1f0733e0710357552bf57411508b1b5f3117516","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-17 10:09:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3178d71fc6f64a9444264225151d7055ff6cca21","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-17 10:14:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe4e130dd4553be1ed8a6a0ade3829ac8620cf5e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-17 10:14:37.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"152ab926a70ab10042efb6f6ffa68b5b58267154","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 10:45:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9856eed55db1956b9ffe0c222e7e4b3f74a1029c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:56:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cc8e7ee02dc24ac29ea516d44599bf99 : SUCCESS in 1h 08m 30s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8414900489764f90ac2b41f00486f523 : SUCCESS in 40m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/68b3b4f383d14674a80050597e884d5b : SUCCESS in 1h 19m 45s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ca22cc3fe2af4695b21c31e1a9ecd26e : SUCCESS in 1h 24m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4763dea0282a49f08acccbba3b646051 : SUCCESS in 1h 28m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/75bded371d1c40e9aa5469ea4c63fb47 : SUCCESS in 42m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9435eb5307914119a4895c96dc98682b : SUCCESS in 30m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f782f5b1b04448ffa748e486f0aa2835 : SUCCESS in 28m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c99ca76d00b547c3b12933c56217dc8b : SUCCESS in 1h 03m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22691936f1224a0983021b43f4affb8a : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db2583dd9db4010b509d37568cb4b26 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fea5d44b9a164a529feebd9d1e2de467 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccd170131c014b389263fe4e7aed24cc : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7992a44cea384394a56616eb3cc486c2 : SUCCESS in 14m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a7196214dab4d25b37de3e99b3ac8fd : SUCCESS in 8m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d76c5c59726f413b9f2f9130a77e45a2 : SUCCESS in 1h 07m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/249cb130d4b34eaeb615c745aa673798 : SUCCESS in 1h 09m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/031a5f3fa61848d1ae510227e105deb0 : SUCCESS in 1h 40m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ac59f639180f43159c45ecfd5ba235ca : SUCCESS in 1h 17m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/939435139fd74a2c80385b3bab03168a : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":2},{"id":"e723882ff970df5aa972797578887ff912547902","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-17 12:55:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d00cfdbbcb5740688ad9d8a501de8318 : SUCCESS in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dfe8ee02ab2e399e7fbe6656c9d52d906f053a9c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 14:44:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3b5c086f8267d1d0805c9f31d86329c654581bd2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-17 19:24:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthanks, Doug","accounts_in_message":[],"_revision_number":2},{"id":"fbdeea2d331e7c98cdabdfb822ed32a0ee745eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:25:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"47f8e0fe14fc7788d3dd1c0f63d47de2f4182ff5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 20:47:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e5f802d183f340df8a501e2c63193da3 : SUCCESS in 29m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/82345b6c01fb409faeab7ac560d599c2 : SUCCESS in 1h 17m 26s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/407126c869564a7b8663beb81ccabf73 : FAILURE in 14m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5026584230724c369a9bd367848a8e7f : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293194fe0f2c4f1b88e78760aa4a9eee : SUCCESS in 11m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/567e79bcc6f4424691e7cdea685a2345 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25495328a7954d70998ae5afdefe6583 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d6c1dd284f432ca2938ed8b0f6f096 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bc60bbeb62f40e18601a20f0d1d0aa1 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb7c75e876b74545b548158352aa2812 : SUCCESS in 30m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/788c7d84d10d4ef09d3ed9a49020b780 : SUCCESS in 1h 01m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/09a866b823fb4a079e94e3c4fef7e96c : SUCCESS in 1h 14m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4c6699f17865c384af1bb6f42f31f1b2a4b3b52d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-18 08:06:40.000000000","message":"Patch Set 2:\n\nrecheck unrelated failures, thanks for reviewing","accounts_in_message":[],"_revision_number":2},{"id":"da43d268a0709872202a030d2402678f123e4349","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-18 08:40:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/df6a619f2a2c49c2b98747a4d962de4c : SUCCESS in 27m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dbd3d9ac0cdf4a2996cea3cc82d6302bb3cc7cbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:16:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/42b974cdb63a40cdb81f8d8c0c6ba969 : FAILURE in 16m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c37372f4f630410cbbd8d27b6b415bd2 : SUCCESS in 1h 00m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d1214f8a5cc475d8dcb9de2af598e25 : SUCCESS in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f5fd226311ea052218fc4e9b370add96f9922bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:29:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/af6d17a78fe54e4989f1859ff1c6402b : SUCCESS in 1h 04m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/114ce9a03edc459d8fc860d7c8d2cfab : SUCCESS in 29m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3757a46fe683471984aea0fd25f6e2f8 : SUCCESS in 1h 16m 15s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cd2c84ce9887492cafd5005f0b4f593b : SUCCESS in 1h 17m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b6b6e11cc45f47fcaffe34e56686a93b : FAILURE in 18m 15s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/86c5fdd582e140cb870411487e4a9b80 : SUCCESS in 47m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f0525d5519364fe492f920af5889784b : SUCCESS in 1h 03m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/31d426ecfd2a458898409a0522a575bf : SUCCESS in 26m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7cef5b9a3c8546e8a0a49bd06a55de0a : SUCCESS in 1h 05m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01e10553dfda48a1acf2645a8dd95dc7 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e2e7905b7945cfa890c4caf55eced7 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06da83f06f5e441f9b4c2f9b2a4e334a : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a15f68a78f84d6eadb24977108c5216 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb2504c2b49346758cbf369d0b9db4c2 : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8295eb8ce3645a5878ad541857d7fa5 : SUCCESS in 7m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30fd19b3fcc74864b4d7921e12bba1ca : SUCCESS in 24m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/164b81085b4942eaa21925c67ef2ac23 : SUCCESS in 48m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/09c7ed54bbe54f86a711069ddc547fd3 : SUCCESS in 1h 03m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f1abd7671e504a60bf38c8ec0c5a0022 : SUCCESS in 1h 02m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d16b302cd0f405b8e8741d6c9961d7d : SUCCESS in 1h 08m 44s","accounts_in_message":[],"_revision_number":2},{"id":"bf12081c876eb3d8ec3cbeadcd07a1c5978d11f2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-18 22:12:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1ee7cab102430999fda83a7fbd19213d478699a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:42:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ade72ab65db8bdbca34f1ae4ac1d2fef95fb6bb9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-18 22:42:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/632d015df7f94977903a5175c9096e3a : SUCCESS in 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14ec2c0508eb81e23ad09caf3beced7e6d7c0a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 23:43:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/70c08d63684542618bc504599a679253 : SUCCESS in 29m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ba85170bf1be48f4a71202de7e5be118 : SUCCESS in 1h 04m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/94a09e2d18aa436b94f8501d6d1d7500 : SUCCESS in 1h 08m 41s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1ccbe4ba70b344c381cd4b9fd6da076f : SUCCESS in 1h 06m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2d4d67290f164789bab34263e49f28b5 : SUCCESS in 1h 23m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e2b711a5c73d405c8473952d23431480 : SUCCESS in 44m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1749dbe54327461f8b96de121d57aa97 : SUCCESS in 39m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/24d30fa88d6d4451b4c9c0a8dbb3d538 : SUCCESS in 38m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ee8ff700bab542f1920383891b65f33f : SUCCESS in 1h 01m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/429575b630d5435db6e5ae42e6a33a02 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/444df4336f43450e8acfc998061e373c : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98ff4790039b497f984b2e17d1f15a4c : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1d6dfef5ab44025a30c85eec1d2cd70 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7637eebc5c2455dab8cbb4b888edae5 : SUCCESS in 14m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ab9c22eac6e4cab818479165cb79125 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b5a750cf30294c518ecadf366d46b1be : SUCCESS in 40m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6c0099ffd024f86a7ca185b8e79d991 : SUCCESS in 21m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e541ea3619894a65a2840c1beada97f9 : SUCCESS in 48m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e9966769672847cc991b8f5284a122ab : SUCCESS in 1h 17m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2e20b0e494ed40f9bde95b45325ed0ed : SUCCESS in 1h 13m 31s","accounts_in_message":[],"_revision_number":3},{"id":"6e08f384d5a067c5b9295ab50a8e9e923eec42d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-19 07:41:01.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5b84dd23994f3a9afea2fdec9167bc04712cf3f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 07:41:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cc728fb108f03bbae39f8235d4b74345f9b31474","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-06-19 10:04:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8e358042e6be51e77407312d1db8c0d0884ef45b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:05:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0c1e8f99a7784258ba62cde89042e31b : SUCCESS in 41m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fb82cfeccf0b485d93cfc03eefd2e8c5 : SUCCESS in 1h 21m 07s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/29eb59ff9b034edc8f32ca3dea02dc6e : SUCCESS in 1h 37m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29533f92d6cf4ca58acda34d40f3c03d : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92fb4690bc6747f3851de059c423205d : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/672ff62339854f34981eefa12e0b614a : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c57be585ae3476eab3a99d0a5911b38 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8df785cb2c84e859158c4654e06981c : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e5af2b7ce574f6ba91b8d5951825457 : SUCCESS in 5m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3bc41f0f10b945fea2322975954bbcac : SUCCESS in 24m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bec1152bf9d4233b3d6517f8a40a5c5 : FAILURE in 1h 03m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ea731339e33d4dfba19c92d85abdfa3a : FAILURE in 1h 31m 05s","accounts_in_message":[],"_revision_number":3},{"id":"79f7a81557e7f3a56763a1f092e27d5a3eea56d1","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-25 13:50:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"309705223d5512ee92a74d90712ce9d70bf3495d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-25 14:23:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/45ed8a8b8e344554bf029378de214966 : SUCCESS in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f6505c150e991b55ed3350ab1afb4af261bcb71e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 15:13:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/096e4d6b7d784935932a7f63798117b4 : SUCCESS in 23m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8aa09e864e064eaabea15d421c4aef55 : SUCCESS in 25m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8d84d1643fb5401088f79eff8107a95f : SUCCESS in 1h 12m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d5be22f64d6d40979978682a18143b0f : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c3953c1efffb4ab3a6a15bbb9fbd0b02 : SUCCESS in 1h 16m 07s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6b29c39c935e4c4b82b2295a2406144a : SUCCESS in 51m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/43a7afc7f3644eed99dfbf44de43f0e0 : SUCCESS in 51m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dc2421ec02344d2d946b3814af7bf765 : SUCCESS in 21m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bd509a492e304500928ad8b347797223 : SUCCESS in 1h 10m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cde7a11c7a449389ba46a91ad9046f4 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bccf32aadb274f488561159161e0798d : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aab74de0ccbb436ca3437fc19a4a833d : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fa6f47f97d349f8b9ac7145703ec60d : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b10a5393d7f547cd97092c56acd91d21 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48ba6efbb8c049f4b61072af1b7a98c1 : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e76803632924b268de4b9dcbdf78e70 : SUCCESS in 41m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f67d89836164dfb8bfb11bacbc5ef2b : SUCCESS in 20m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/72f8d69f9d3e449fabcfd058d129f3f1 : SUCCESS in 56m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f847c86ce9b4af6b683f6904044c512 : SUCCESS in 46m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/38961879dcfc4b5b92e287e37d28fec6 : SUCCESS in 1h 14m 18s","accounts_in_message":[],"_revision_number":3},{"id":"59741aa1933ef387357a78bb08c218948a20f734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 15:14:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2d496f8134c243e559ee1d6550c03db134a90458","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 15:27:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5471a48ccce54d1cb73eea9cd3dc4943 : SUCCESS in 55m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/708da9af2c68482f88596af32208ec72 : SUCCESS in 40m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/34ddb7bffe934662b813918dd22f51b2 : SUCCESS in 1h 31m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4a364a964de647b8867a21ef249615cf : SUCCESS in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eddf81ca4172c5bea62330aeff48f979ec5646b3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-25 16:16:48.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 11903c833687449cc05a7bc5c99c6417af4d05d6","accounts_in_message":[],"_revision_number":3},{"id":"3d9f70a9db1eb517985abfad4573a10d28bfa919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 16:48:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c69720ec465549edb47a6200d3128e06 : SUCCESS in 27m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d284064ab5814cb5bed70560bf2b3e15 : SUCCESS in 1h 19m 28s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/04acea29220c41258e0516089fe6d845 : SUCCESS in 1h 28m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9570d10a11a445d88c9480e422be92ce : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a90633d988d40b79d7d8055a0063bbf : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a589175805d845a5850ffd65ba47a9c9 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5b1c8cd20e9450b9638b5ad3cbbeab3 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdfb2380d51741e491dddfe263c952eb : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45d81d181304472da3125846032ebbf1 : SUCCESS in 4m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ba02772b26345fc88c3ed5c97bdd3a9 : SUCCESS in 28m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5b114b23236c41bfaeef504ad4e87f6f : SUCCESS in 59m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9c13b6fd88f4d1fb8560beee2fdb10d : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":3},{"id":"62d5b42cb6a5188e89e4881c8f9956318ac8f5ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 16:48:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"264eb1255cde25aac96909c3f8c4d69b85737c64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 16:50:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2119c05fcfaf4b7e8ba75b5f44c213a5 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57c7647f7cf642b0beb7f421ce3d4a2f : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dcde7ffdd02f62cbb2a8a52aa93de029ccc2a43d","revisions":{"74f6e3d43fd8ae2adc15f09b5991363bce173a26":{"kind":"REWORK","_number":1,"created":"2020-06-17 10:09:56.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/52/736152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/736152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/736152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/736152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/736152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/736152/1"}}},"commit":{"parents":[{"commit":"2b140296fc7511c85740a7480019af5e4abe8332","subject":"Merge \"base/centos: take care of whitespaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b140296fc7511c85740a7480019af5e4abe8332"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-05-23 16:23:09.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-06-17 10:08:54.000000000","tz":60},"subject":"Fix loading of Storm on Centos8 / Python 3","message":"Fix loading of Storm on Centos8 / Python 3\n\nIn the Centos8 images, Storm fails to start since it cannot find\nPython. This patch ensures that it can by setting the PYTHON env\nvar.\n\nChange-Id: If6663b2f6dfadfd6a6db5e4aaca4eb782b87161f\nCloses-Bug: #1876461\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74f6e3d43fd8ae2adc15f09b5991363bce173a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74f6e3d43fd8ae2adc15f09b5991363bce173a26"}]},"branch":"refs/heads/master"},"2aef862eb55db331cb295d8697b68e92e21b577c":{"kind":"REWORK","_number":2,"created":"2020-06-17 10:14:02.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/52/736152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/736152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/736152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/736152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/736152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/736152/2"}}},"commit":{"parents":[{"commit":"2b140296fc7511c85740a7480019af5e4abe8332","subject":"Merge \"base/centos: take care of whitespaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b140296fc7511c85740a7480019af5e4abe8332"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-05-23 16:23:09.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-06-17 10:13:17.000000000","tz":60},"subject":"Fix loading of Storm on Centos8 / Python 3","message":"Fix loading of Storm on Centos8 / Python 3\n\nIn the Centos8 images, Storm fails to start since it cannot find\nPython. This patch ensures that it can by setting the PYTHON env\nvar.\n\nChange-Id: If6663b2f6dfadfd6a6db5e4aaca4eb782b87161f\nCloses-Bug: #1876461\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2aef862eb55db331cb295d8697b68e92e21b577c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2aef862eb55db331cb295d8697b68e92e21b577c"}]},"branch":"refs/heads/master"},"dcde7ffdd02f62cbb2a8a52aa93de029ccc2a43d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-18 22:12:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/52/736152/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/736152/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/736152/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/736152/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/736152/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/736152/3"}}},"commit":{"parents":[{"commit":"039a6d431ae1b2ddfb7b593c1c2d10bb58556b9e","subject":"CI: fix line merge in CentOS builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/039a6d431ae1b2ddfb7b593c1c2d10bb58556b9e"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-05-23 16:23:09.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-18 22:12:43.000000000","tz":0},"subject":"Fix loading of Storm on Centos8 / Python 3","message":"Fix loading of Storm on Centos8 / Python 3\n\nIn the Centos8 images, Storm fails to start since it cannot find\nPython. This patch ensures that it can by setting the PYTHON env\nvar.\n\nChange-Id: If6663b2f6dfadfd6a6db5e4aaca4eb782b87161f\nCloses-Bug: #1876461\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcde7ffdd02f62cbb2a8a52aa93de029ccc2a43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcde7ffdd02f62cbb2a8a52aa93de029ccc2a43d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
