)]}'
{"id":"openstack%2Fkolla~736570","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ieb88fd60d177f19a635ada9366193e57c516f90b","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"Ieb88fd60d177f19a635ada9366193e57c516f90b","subject":"Add collectd-amqp to the collectd image","status":"MERGED","created":"2020-06-18 10:47:17.000000000","updated":"2020-06-22 09:16:24.000000000","submitted":"2020-06-22 09:14:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736570-1592817265945-1172c0ca","meta_rev_id":"12d7907011264502a1d4a802d3b0a931de100219","_number":736570,"virtual_id_number":736570,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-22 06:06:54.000000000","_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 09:14:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-06-19 10:56:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-18 22:05:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2020-06-19 10:56:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 17:40:20.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-19 10:56:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-22 09:14:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b1778e71cda975d0996540575c86357f655c9fb","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-18 10:47:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"17a33a4a7af1ba4755263edf85daeb1c69fe84b4","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-18 10:47:30.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 6e6638706801a78fb274d5bc696aedc8f4f6c981","accounts_in_message":[],"_revision_number":1},{"id":"21c8ab9a783fb103ab310959eeb4f98dfececd9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 11:52:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8acd636d0c56413a9c0fd13ac8440371 : SUCCESS in 57m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0f813f6e934ca045aab571f1989e96e65bf0eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:21:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/665a9ee6c03a499c83ece32b449718b0 : FAILURE in 1h 21m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7d3621f895cc4623a2f1483ba2c3ed4b : FAILURE in 1h 26m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/121c2b9ce93e4abf8caac7620e28a071 : FAILURE in 31m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/48696c9313194ddfad151abfc7f4516a : FAILURE in 18m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/329557dd2d7e48f1b6d8bfc3480538f5 : FAILURE in 22m 56s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3c7ac46376724145834d2625cf06a335 : SUCCESS in 48m 35s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1b53dca73a5d49f4bad3e53548f848b1 : SUCCESS in 31m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/103ed01598bc4c8890ade3301cad15df : SUCCESS in 27m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b4537aca57a14c80b2d14c00cb241c6a : SUCCESS in 1h 10m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/457426d6aec34993a0cfb54b29c64adb : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd623a36b1842a6a1498411a708d8d6 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5610ecff7de94d84a33d037363d20be4 : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74c9e76e85624b0b8dc545b58cc84f49 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bba014231964e0b8988d92c2db20708 : SUCCESS in 8m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0ac67f40af9d458bb2d23f9be7439c67 : SUCCESS in 30m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/55159a9786d141cfa3ec3e1efa8d5d94 : SUCCESS in 32m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/93474c893ba2444284becee04e6017b9 : SUCCESS in 54m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/39e2cb9d9454491c8b8b4ea8ecfc9208 : SUCCESS in 57m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cca54307d7dd4bd8b74ca74e99de6046 : SUCCESS in 1h 30m 40s","accounts_in_message":[],"_revision_number":1},{"id":"665cce7b7f459de316fc389a401e8f316679b55a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-18 13:38:16.000000000","message":"Patch Set 1:\n\nThe build failed on multiple containers, but not for the collectd container. It looks like a repository sync issue. Will retrigger a bit later","accounts_in_message":[],"_revision_number":1},{"id":"e244586714012fc5237b335175ff101c9783fe8b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-18 16:10:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4828585f2b8ba0b8a3c88c403977abc90f07d61c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 16:51:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1700dcbcc2d6fbfcaf50cb099dae3a8288f1eed","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-18 17:40:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"661a29bd42cb46de3b51536979765b1096ee6a56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 17:45:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/22e34ae7972549f7880f77516bc88622 : SUCCESS in 23m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/41868280d40240db852ef787e1e5f089 : FAILURE in 27m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9230b85e947f4fe6a8b7a854dd55a114 : SUCCESS in 1h 11m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b1762bc9c332456a9475ecc91f223133 : SUCCESS in 1h 12m 17s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c97a9e20c223497fa765ff70b31a1af1 : SUCCESS in 1h 25m 57s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/af026c06dcf94f44aaf9d4602d748401 : SUCCESS in 55m 55s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fc9a4b01d53d4782b7eae0e6fe497e4a : SUCCESS in 31m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/769f975359d644f788a11dbb120a121b : SUCCESS in 29m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fe240e8f63cf43c5bd516b5d7fe88876 : SUCCESS in 1h 09m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1853c98f99a0498f9cc175ddfa537302 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f01b3f3133ec4377845e8d826c5f9a02 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae3cc26bda6141f28f472ce8a3095d0b : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b5eb0b75a824200a16ba9d221151358 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81fd6a2d186e472180be80926dd745e6 : SUCCESS in 8m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4aed289e422241238092bc0424469672 : SUCCESS in 27m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6903bf9999e47ebb3b9cde9126f02c3 : SUCCESS in 27m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b76d2251366743979de61f334c945758 : SUCCESS in 50m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/15d4ef963ebc47df883c12f4e7af89ed : SUCCESS in 46m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5089c1172374881b932fe5cdbfaacbc : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0671d66adcd6b4163986c8391b958031cd015b92","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-18 22:05:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ebeef732e72ba9bd7d7ac3c87561f7bc063fcdf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:54:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a270a4886f73b40fe7d4ece75246b73aa3d81ede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 23:36:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e445b5c77f844940875e008423d6617f : SUCCESS in 20m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/880fa2caf2bf49dda7d4c5ab482791b1 : SUCCESS in 25m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cb7f50794b914da2b5534810c6c77964 : SUCCESS in 1h 15m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/09d462205c46416a8f56c86186b315f8 : SUCCESS in 58m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e0a6108690e246a59afcb1df501100ff : SUCCESS in 1h 24m 26s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/9f04f2f3c6404d4193156b2cb18c3efd : SUCCESS in 45m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4518d1250fe24e3fb0394ac2f81480ee : SUCCESS in 19m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eedaa556457c42a28e15f0cfbc216803 : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dbe4e8b675664167880a53cc51a59165 : SUCCESS in 1h 05m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f110df9e84e4d7fa1ca73ff5e10e3f2 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9dedee36ba24a40bc42ff2824deebf4 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c91661c572024c6a8524d085f4f577d4 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbbbdfc110334b4bb702f5495c4d5ddc : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37fa1bd745294d8ab55bceccdfe8edc6 : SUCCESS in 8m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4bef32218064056acecbd54af3e7a77 : SUCCESS in 24m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dbce3839df3e4d4e9980e64be0972db1 : SUCCESS in 23m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07789795253845729e5474ae01105510 : SUCCESS in 1h 25m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/15abcfcab8e44c9fbd00ac66e7599a9c : SUCCESS in 1h 01m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ddd60311796443d19353c0106beae7ee : SUCCESS in 1h 24m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ebd5bf822a97d03ec66d0d7376b57a42583cf9de","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-19 10:56:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7494d487e612fbdb4b129cc43794786f846665a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:57:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1099d8e9aceb28c6927b52ec8f5166a8276ccc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:43:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/752bb00880534004b9efc712c5cebdc5 : SUCCESS in 27m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3acab24ebfd44398be1991b8b1ebd878 : SUCCESS in 1h 14m 26s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/eb2684a096e94a759d5e41d54ce19423 : SUCCESS in 1h 14m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6032955feb914c6fa1e726ee4df97006 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c0f5ac0d5e449229e86ddb6c7a4a8f2 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/163402823ba8400db94dee46ece2ec7e : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/139652c83df84d2d815aabfd26327ef3 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/140eceb2bfb34f0d8b0d911c6bb5294c : SUCCESS in 6m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a4244b5c12d487da5a25073cef109ea : SUCCESS in 24m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b37e8f5a540040c4ad5fd268ab231791 : FAILURE in 40m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b0526bf706e94ce7a94ff6fb89f6f5bf : FAILURE in 1h 05m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e6de58cc097d17a4b5ef0f6f2eaf34ac99a9603d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-22 06:06:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"38a56a3170b810f58e4afbdfdd85d930ed3d5a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 07:48:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9d0b9a21e0c44b888a76025dc8ce05c9 : SUCCESS in 49m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e85140f273634e9f9f65769a50ed7944 : SUCCESS in 52m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f2b77dc48d754f0c95603405ad65e7a8 : SUCCESS in 1h 19m 44s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4a186647513a4ea49c94c2ae09f398a5 : SUCCESS in 1h 29m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/710136c6ed65431190453360783f1d35 : SUCCESS in 1h 32m 09s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/02c14b6fb5c8417aba144f500203bb8d : SUCCESS in 56m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6cb5d3f79c854113991321d9553d144e : SUCCESS in 31m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f25a85256044dc3a6fdbe66c0d19db0 : SUCCESS in 24m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/49c4838d719b433f8cc6674ba76b020f : SUCCESS in 1h 10m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c421a143e9a441a8ad4d008fd608195 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c57f3266eb9847b984ccdc55a431b2c2 : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/242fdf87055e41f68ca2eb21821dc619 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e68a67bf3ce444e79bbb3b6510cbc900 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1496519a7aa24040bc7d6abe2b307e11 : SUCCESS in 6m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f6a297617723477d980158531ec108df : SUCCESS in 44m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e9f285a32514f08bfc9ee5629d5fedb : SUCCESS in 51m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/597b3d0b888747458fccf078007e8838 : SUCCESS in 1h 15m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/89fecdf327764cd8b8d02fdfd5212542 : SUCCESS in 1h 11m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/caf358e0c348498bb13615b692ad4e74 : SUCCESS in 1h 32m 57s","accounts_in_message":[],"_revision_number":2},{"id":"fda88968af4eabfdac3384e15c9727c8cddb9614","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 07:49:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f9b8f1550f7380238491ec26a0a3c009b41cffae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 08:30:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/82be54ab2bad4adebbc31c5884c6f7bd : SUCCESS in 1h 11m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d31d376ed61443993de8f9f366c93e7 : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/57abf0bec7a747b6b3dd64ff39fcf0b7 : SUCCESS in 2h 03m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a14c7e6da39742eaa9ab5b9fb6f9b781 : SUCCESS in 1h 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bcb9be9bbbf9922810eb23cbf3a300c957c996b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 09:14:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8b3e0941ce99482ab47f1ada1a93f742 : SUCCESS in 41m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/774bf4a439f541298f1e979e727dc338 : SUCCESS in 1h 23m 25s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c618efe5a9ba4b1b8aeb86e9aa819c49 : SUCCESS in 1h 22m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f69b3df28f34799b79aa50e4b2f6605 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8610b04d5e641d5b2e10d9e871f45bb : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebc7770fcda241a1bf1d5ffc3bf61f4f : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5df585294a9448ae9dc69f80304bba20 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2d5531a9044fa2948cff7d8b5ba56d : SUCCESS in 6m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/618efe26d4e6438a9a034a06634c7e9f : SUCCESS in 25m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6c54e676fc44a3287a20f180676fc99 : SUCCESS in 1h 02m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/23a0f4e02014476e8ee836a254d93a45 : SUCCESS in 1h 09m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f3970bd053e95d87404e7250e17f8e556ba2373b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 09:14:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"12d7907011264502a1d4a802d3b0a931de100219","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 09:16:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7848d00eb90342cf966016c682a7cda7 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b92446b922da6611274701860c482885eb2a9238","revisions":{"63d689c4806d00ab7ce0ed4ea1f0c01be14ae07d":{"kind":"REWORK","_number":1,"created":"2020-06-18 10:47:17.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/70/736570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/736570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/736570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/736570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/736570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/736570/1"}}},"commit":{"parents":[{"commit":"47eecc0180213c7d649243b28bd779eb57ba4333","subject":"Merge \"add libibverbs package for openvswitch\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47eecc0180213c7d649243b28bd779eb57ba4333"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-16 06:24:29.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-18 10:47:17.000000000","tz":0},"subject":"Add collectd-amqp to the collectd image","message":"Add collectd-amqp to the collectd image\n\nThis allows to write collected data via amqp.\n\nChange-Id: Ieb88fd60d177f19a635ada9366193e57c516f90b\n(cherry picked from commit 8bfa655007e9795ee6b4eba09cba4f403c113b48)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63d689c4806d00ab7ce0ed4ea1f0c01be14ae07d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63d689c4806d00ab7ce0ed4ea1f0c01be14ae07d"}]},"branch":"refs/heads/stable/ussuri"},"b92446b922da6611274701860c482885eb2a9238":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-18 22:05:33.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/70/736570/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/736570/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/736570/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/736570/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/736570/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/736570/2"}}},"commit":{"parents":[{"commit":"4d504c1ec9335c020a0f8f6d9a1b8335c5858844","subject":"cyborg-api: install proper WSGI for CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d504c1ec9335c020a0f8f6d9a1b8335c5858844"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-06-16 06:24:29.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-18 22:05:33.000000000","tz":0},"subject":"Add collectd-amqp to the collectd image","message":"Add collectd-amqp to the collectd image\n\nThis allows to write collected data via amqp.\n\nChange-Id: Ieb88fd60d177f19a635ada9366193e57c516f90b\n(cherry picked from commit 8bfa655007e9795ee6b4eba09cba4f403c113b48)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b92446b922da6611274701860c482885eb2a9238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b92446b922da6611274701860c482885eb2a9238"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
