)]}'
{"id":"openstack%2Fkolla~737051","triplet_id":"openstack%2Fkolla~master~I68e0cde16bf84ca5743c8c8dd6b6460752587e92","project":"openstack/kolla","branch":"master","topic":"curl_macro","hashtags":[],"change_id":"I68e0cde16bf84ca5743c8c8dd6b6460752587e92","subject":"Introduce curl macro - fetch_url","status":"ABANDONED","created":"2020-06-19 17:19:07.000000000","updated":"2020-12-03 09:56:09.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9c8c8a26cbc3e01cc435cb99e738c96aaea7e633","_number":737051,"virtual_id_number":737051,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-05 16:39:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 18:15:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-23 08:19:42.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-07-03 11:44:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-05 16:39:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd072d945168cfd205af2532cb653b38a3a337c6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-19 17:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ad47af739ab6660f181c6add54bc6f9fcf2de54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 18:10:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1fb1439f705b4b63b61d175039f913e0 : FAILURE in 4m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d5009930a1a4432d9e5f4682f10d9fb9 : FAILURE in 7m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9227caee96b74c8eb111933d85593e35 : FAILURE in 10m 13s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a4098de307c34e729b6b4900d35e39ac : FAILURE in 11m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b0467aefa86b47ebb7670494386dc36f : FAILURE in 11m 15s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/77a140ee7af34e1a8b723a7f5d7d821b : FAILURE in 17m 53s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/788f13fc84fb4abeba8f3ba961b71435 : FAILURE in 11m 10s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b4650d10e0494606b0733700d4270a1d : FAILURE in 17m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f2bc8773f4c049b3a3806b3da6a85e45 : FAILURE in 12m 28s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/b68ab467979b4bb6916c669bbffcb62e : FAILURE in 9m 26s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/afedb0dc404649e38dc35e41c26039df : FAILURE in 14m 36s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/b8c42499a7764c38b04920f71f7277eb : FAILURE in 13m 20s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/03f4db4e274741a1bebbcdc1cda5b0ab : FAILURE in 14m 10s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ca152b8c41ab48ffaa29ce4e6387a81b : FAILURE in 5m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1dfbec0886594fd0b23e41d4db76d980 : FAILURE in 6m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3fdcc0e650964ecfa6cc876c103f229c : FAILURE in 12m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf22f14001554248804e4b57cf15bd54 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/342c41df08e549859e32fbd8edec55ff : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eb31353783141a280b24c6df6656003 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b878d88bab642e4b48f7507f99df9ca : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35b2668f9844c1dbd90be33e93e9b51 : SUCCESS in 8m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/da179f067eef41caab522810050a440d : FAILURE in 5m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0822e4729bc8485e9566e937bf115345 : FAILURE in 5m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fab2906bdcd44a45a82ee9422b465b98 : FAILURE in 8m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/944d4a934c6f4b97822410a9f147af3e : FAILURE in 8m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/365640f8a869457b90e9533129ade7fe : FAILURE in 9m 46s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/de3b77c5f7804094865d7a2aa5993685 : FAILURE in 10m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/43cc1cd645b446b2a84d95b7dcdd3409 : FAILURE in 12m 58s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f5eac3f5bdc944c497c9b3d7e22ada72 : FAILURE in 10m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/40b32c114c6441dd84306740a2fc03f9 : FAILURE in 11m 37s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/fb8b892fe542440f9f4633d3411b501c : FAILURE in 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00a4fd37d93314db8d075362b8fb59824c975ae8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 18:11:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c330d7000dc86018600062168af6257098ffce64","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-19 18:39:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/3c52903dd58d41548191ad4c784b809a : FAILURE in 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"560b2742b16d3207fbb6b04875da1b37e4905935","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-22 18:09:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5340c76a2fbef080af0124dfd0a78e0da0a08ebb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 18:25:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e271295d974345f8a0997878d6d68cd6 : FAILURE in 6m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/28a86659d0da44dbb709341b0b9c416f : FAILURE in 4m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f76efcc36b0e4e1ea39665fd7a3233dd : FAILURE in 11m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/377b6a6828cb46af8bb84fc96932bbfc : FAILURE in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"136f8626690881d646bdc78392353ca7a9446b93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 18:35:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9dbe2e0f0c12450f800dfe8a37ee42cf : FAILURE in 5m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c9f622bf728f4e51a1ea76847e910131 : FAILURE in 6m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cc915c65cb274d4a8edcb6f94876997d : FAILURE in 17m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/54decd55807f4104921fc176f86b6dea : FAILURE in 14m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3806219bafe34bcb96d01dd82ec6a861 : FAILURE in 12m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/bc94093addcf4c9c89e974b373d8a795 : FAILURE in 20m 22s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/ddc4f0497e0e414f86f063467eef0027 : FAILURE in 13m 05s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d0229cd82cfe4531983930d413f7ac6a : FAILURE in 13m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/781c62cc4c314a6ca53bba1e90a22388 : FAILURE in 16m 07s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/b9aba16482784f9bb8a17b10adb1df9a : FAILURE in 14m 52s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ee255903f7c54eeaaea04337dd6dd70b : FAILURE in 13m 12s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/f10476614b664cc1afb24fafc41d493a : FAILURE in 14m 25s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/88fed727e0d146f78c5835a90084782b : FAILURE in 14m 06s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/471b70bd78344421861efd094fd33fee : FAILURE in 6m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8179a8e8464f47598b649b919e8e24c5 : FAILURE in 7m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/96c037d499b548ec8d4c82d8114221bc : FAILURE in 9m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c23e43797a9485e8dcd110ab3dad2d4 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b777bf948e482fa513bcc674b071a9 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94bfce22abaa4cb4a074c897a02b28a4 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6781625675144115ba167cbb9a088f6f : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f89cbe34b14fbfb08dddddc162f0f7 : SUCCESS in 11m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c05c023b26a846608fab93b746af74fb : FAILURE in 5m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae5bb9407926424192f525e870a0c8e1 : FAILURE in 6m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cefb2578f2db40e4b0f4734b6c90770d : FAILURE in 9m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b121802bf804b849bb074963b9fe7be : FAILURE in 13m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/05f228e14f2b4fa08fba6a6764d04c2d : FAILURE in 11m 26s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e72e0d8a6f0944738be19760087a7181 : FAILURE in 15m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/67d069837bf741debc3b059a0248ab49 : FAILURE in 12m 02s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/dafdf21081ca40ffb5ef3e10d956ea5c : FAILURE in 12m 07s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/59bc723ad14a450faae615a831404075 : FAILURE in 12m 03s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/525e1750b5e841b88edc00009164290e : FAILURE in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ffe483b2a7061ec9454d04f44cb748572145895","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-22 18:37:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/34179f02f6874ac79ac58aa9c7c8b4cd : FAILURE in 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a18eea04c6a0786d36327c7ccc9d6291fef05ce8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-23 07:59:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"29cb974d8ac1351d404205d21911e1344c17c496","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 08:12:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/14b4c5272c544c4794457bdba0762beb : FAILURE in 7m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ad55cbd2f31491ebb4ddf5145239644 : FAILURE in 8m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f752f5e57b9f41ba9ba7adb5cb38f9da : FAILURE in 12m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c06b3143cd954bcdbc0589cb4ac4aeec : FAILURE in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"efec6b0c818bc9cd927d6b14983e679c94e0e95b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-23 08:19:42.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad1342bc9ab1290e1ae695acc10f88dbb1aa4873","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-23 08:28:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/9f46ba2910aa49f3a1ad88df42d71dc4 : SUCCESS in 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f913bb97900c07c85ef766df9574969c1a5a2e97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 08:41:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/82ff91e1fca84e60a055d573b392ee51 : FAILURE in 7m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cf12001d5270475a87bc4417d545e27a : FAILURE in 6m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2efe1251b305435cb8a16cd467b8551f : FAILURE in 12m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9ddc57a14fe24fb19bdc68d35f945c27 : FAILURE in 10m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ebbf64a5361b44efad7aebaef626a7b6 : FAILURE in 12m 56s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/1935fad861eb4828a0ce4fab6d10fb48 : SUCCESS in 40m 11s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/3e900c2e75244963b07c9177e869a39c : FAILURE in 12m 58s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/dfc43cb2a6fb40e3a97d5e3bb7183ed3 : FAILURE in 13m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/142bbaf32ecc4b3aa49737da7b02dae3 : FAILURE in 13m 24s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/478a420519f14058bfc8f800c5511251 : FAILURE in 11m 12s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f62267e0979b414b95cf3e7b4af3e893 : FAILURE in 12m 47s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/8c0c85b478b5442c946adcec12f84605 : FAILURE in 12m 52s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/35357b261b3f4feb9c19e94c1e3b4f1f : FAILURE in 13m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/19cbec2623614618a4cc6a50370380ef : FAILURE in 8m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dfcf19dbb7ac407ebafa4b26ce257b61 : FAILURE in 5m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4b9508e7667d450a8ba23415b986e2ef : FAILURE in 9m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3ac48a7da534ba992a6e4d5466a5f2b : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fd785eb77149599a9634976e5f4e30 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26ac8e9edf384c3887074d76e6da7d42 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64c94080f424475fa88d1ed642f7154f : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b9350635a24d5fb20ba25d4a566c4d : SUCCESS in 6m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76942b644ebc455d81b20a6fadcda2ad : FAILURE in 7m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1def46e8cfce4a34a99bf2bfb4ca2e5e : FAILURE in 7m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/831391c246564ae4b9bacdc180a36fd4 : FAILURE in 10m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cbe40ec4fd604f51bbf2a48dc7b74cd3 : FAILURE in 11m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/57e59b4722544b38b0084ae4c23a1c5f : FAILURE in 9m 23s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0c8a1fa7114e4bd09f49947f30dc7611 : FAILURE in 12m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8d069a7279d74a7d871c5dcdf39c5122 : FAILURE in 12m 13s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2ebdda1354124f339bb1557bb1417055 : FAILURE in 12m 15s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3b3ad5efce1d42eebca36bc9b0aa6faa : FAILURE in 10m 38s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/0b6ee0bde02447ad9b5ad0232864f0a0 : FAILURE in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c5f5b1cfc92472eb910ff1b5ef5b1ad68ea5ba67","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-23 09:49:12.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"cec60c3f4afaf38ca6401a1c8de1068b21cbff90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 10:07:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1aca6da12d214aefb489f6c9bb73d273 : FAILURE in 8m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bfb6cc8683c34e09a1ea201ba1a9f5ac : FAILURE in 8m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a940eb76f677421caab99f479b2d2bdf : FAILURE in 13m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5d78ee610fb4f04a86dff4ffc7fa39a : FAILURE in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b87a2c3403cf1146327f2ebda3b243dcae46b0fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 10:42:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b2b8b9758d4d46e4b7a3bf71a45e467c : FAILURE in 7m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/818f5169b5c14f288cc1537706597978 : FAILURE in 7m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7d0be2fdd5be4eeea8006d9dc9bff4a6 : FAILURE in 13m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/35a44d4a26dc46549e463ee659799d8e : FAILURE in 12m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7814ed8a5b3c4a86abbf3590029acff9 : FAILURE in 10m 42s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3718eda9b77f481eb41f867c9776f612 : SUCCESS in 38m 47s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/8068e69fd247474596c42e9953b66cd7 : FAILURE in 13m 12s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/076106fa85504cf780817bf4cd0c9dd0 : FAILURE in 13m 04s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3bc9abe7acb248aa951ad70a29521681 : FAILURE in 14m 33s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/8429968ddb614f96bc25b80c3dfdc7e2 : FAILURE in 14m 18s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/b3d04b5726f24cbd9c14236a8637b255 : FAILURE in 13m 09s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/71e3a03009f4462f894307a9f341a2e7 : FAILURE in 17m 15s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/473221ecbe50444fbcefbeb8023fa538 : FAILURE in 14m 58s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/952b541ec5534f87b0e18f7081d4abda : FAILURE in 6m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/04759b95b40c4d4598575e4030447194 : FAILURE in 7m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e4642686fd445e290ffc1544cb16ab4 : FAILURE in 9m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c97650fe8534e6b8ce2bd574f8adcea : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237664f6b895428d983808a0e14f212b : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b758978fec204ce38654726170eb2782 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cacd7683c6ae428e818c0a97cfa0cb9a : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a14baebc0f4478b1758ed90a3862d1 : SUCCESS in 6m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57df19e9edc04287849e762d8c2b5609 : FAILURE in 6m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1954395487a54a49b94bce65f2f402b6 : FAILURE in 5m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c6b4242314f445a89b83714fe6d4d920 : FAILURE in 10m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/827f096fa52c49b487c2a6d97a547791 : FAILURE in 11m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5519fc530e5c469a9e400197da1379b3 : FAILURE in 9m 42s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/eefcf3dfb4454f94ad6d3dd488085b06 : FAILURE in 14m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b2ad7af724634faf8ec089b76273151e : FAILURE in 13m 57s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7681703459404dc6b001936f383b71bf : FAILURE in 14m 56s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/6429adf5abf148129b91f50ad75ce204 : FAILURE in 13m 42s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/12ae9f56ee0245c99e8812afb198c9d0 : FAILURE in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00ce06f1742a42983084cf093d9afcc131efcb85","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-23 10:42:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ee3cd7facb1c4525882f634a7315ca1a : SUCCESS in 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6c75587c8532c9a09c0915fb14c02f2d70003975","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-25 09:49:01.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"1dd58240830a6821902a124ab6dd72b06be866fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:06:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/304c4494f22549fd86694515a1e18450 : FAILURE in 11m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/66f20bdbe74f4d2bb4953cc4ea42b9ba : FAILURE in 7m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed796dc4647343788069623436e7f6ca : FAILURE in 11m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/664fb774034641faa759ffcbe4aef66b : FAILURE in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5c7cc44262e7b0eadcd9027b52fe7bd254a84de3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-25 10:16:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/1f9e426c243e480da696a35f4ea6c604 : SUCCESS in 26m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1347cd9a457598fbb69d3d5c10e629c285605596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:49:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/de6f5a6db76341a293e4125b89d94c8d : FAILURE in 7m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8d6a90f5091c48e983b2b2ba31daf512 : FAILURE in 11m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f2aac48405ac461aab059343981fc078 : FAILURE in 14m 37s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4fb4eafd98644e249edd5da7d6c8624d : FAILURE in 11m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7d40683275f945a394905b2db6401f16 : FAILURE in 13m 08s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ca3448172b3d49b99164a93bbfbccfa4 : SUCCESS in 53m 16s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/c992dbdb6e06493eb769ca326ca48eb5 : FAILURE in 13m 14s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/63dd5baab5a2436f821eb0192cd51c7f : FAILURE in 12m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bb43d9399d804e828195e918903cff1c : FAILURE in 18m 47s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c7d1bbb9c0004fe9810be7b2f1dbf6ba : FAILURE in 13m 49s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/2be2ae28bdda40e6958d0da9b56369f5 : FAILURE in 16m 34s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/c2ad3d9606334e02854a5ff176dd7d74 : FAILURE in 14m 32s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/314f024d763c4213a727231dfeea37c2 : FAILURE in 14m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1067dd5b252344afb6f91f055220ce09 : FAILURE in 6m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e6913971d3ca4979ada2e95bf8bbbe00 : FAILURE in 7m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7047f3432a994a79ab2c13ef3dea430a : FAILURE in 13m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7558b551742e4dca815549472bb794e4 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5648946c294f89b2fa2790aecdebab : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf9df60ca9ec438494430a61fd0c6fb1 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9679aa56257c44e0b783671a474ffa5c : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e905b4f15e47079d17ef6f00f66941 : SUCCESS in 8m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/59884d4fbf30471db6b292b94f83b601 : FAILURE in 4m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f35c01c4ba44fb5a344fe24f772ac90 : FAILURE in 5m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c50088cbe5114dfeb0fa758d3763e868 : FAILURE in 10m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b8073861aa6846caacbdf28b15d43e18 : FAILURE in 10m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9ac616e2656b4aec91166e063e56036b : FAILURE in 10m 01s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/31c0585e77ec4e8b8ee3420be5f34c09 : FAILURE in 12m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8563fdb13b0849c4b174514bd4fb1e32 : FAILURE in 16m 19s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/47fd861cf6754a0f94d42b5368522c6e : FAILURE in 9m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/4a8557a832e54c8396647ede6b9abbaa : FAILURE in 14m 29s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/fcbb9e674f9649518442f3e386883e05 : FAILURE in 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ad2641df5d8a9a0fefdffa798c7d70bf36b99900","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-03 11:15:21.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"a5e3edc1f949c859371f8c33a75de2ab6a83d18f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-03 11:15:31.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"2135750a60a81ce3cc2a77c9946e03882afc3a6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 11:33:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3eb949114d04c28a0a67c80a24298ab : FAILURE in 8m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/53bcbe7e6d7642638b829902f7917d73 : FAILURE in 8m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ebfbd452af0497ba5ca995f560da9c5 : FAILURE in 12m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f943eadbb01b40b6982f38c82940e77e : FAILURE in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"511b3137a79c4c147c0e754e8ef42e088a362cb8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-03 11:44:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/923a7c52d90d49b8a545f87e704fcf6e : SUCCESS in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1996ea9b0982e6fec80b20061894754cbad8d5cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 12:00:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6f94dfb52f1e4294b3bfce8058ef008b : FAILURE in 7m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/01c1ef42cb7e49efa6bfa6350a7de423 : FAILURE in 8m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e80df671fe074c01b890d6ed3ca9f4c5 : FAILURE in 14m 29s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/37931ac3309d4af9bf58db8ee7718c4e : FAILURE in 13m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/437ba583034c4b0286e9a0123747e73a : FAILURE in 12m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/cc2c9f7a94104bae9e6e160aeeb7c2d9 : SUCCESS in 40m 19s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/682eeeadd0f14d1cbd4bce5db300ecfb : FAILURE in 12m 43s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/60c1dbfa36ee46e790dc84d1ce9c7e41 : FAILURE in 14m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/91715c1cccd54af9b046effb5154d367 : FAILURE in 15m 55s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/1eb88493de6246ee9f07e9471093d1c1 : FAILURE in 15m 17s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/aab130a7760d44adac89ca0d44a2c3cc : FAILURE in 13m 10s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/e76f9f0bc9734d1e8f2187ed2a59642e : FAILURE in 11m 51s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/7ef4d2da097f4e959e48af8620129947 : FAILURE in 15m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f44b3cd73b6c42b2bea3f45cad275e37 : FAILURE in 6m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f08d6d64472545eab7bbc62a151fda68 : FAILURE in 6m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e3eebf031fda42919477fdb7efd11692 : FAILURE in 11m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/879e65c118f1420bbdd5fc1d7c7029af : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be0ee0c4a4a4021b75ce7cfd57923ea : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9c37e1886884703b9fa362e83e8c622 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a498fa06e64c4c95b2af97453e37bf13 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844e080634964ef782ca27d5e78e9e27 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb44180b814649f6aa717268c98a314a : FAILURE in 6m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/52d2ddcebb914328aee93497044ffd02 : FAILURE in 6m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8013a649aa4c4cc9904f7a11cba07ac7 : FAILURE in 13m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0169c37571984ca4af0e4e4502fb8653 : FAILURE in 13m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/94e5b3fe9f044cc4bdc98aeef3e7c15d : FAILURE in 12m 42s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f3d1aabb27d6411e861805de933e428b : FAILURE in 13m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/178453b6eaa24f64bb8f024209b396a6 : FAILURE in 15m 32s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/9c089da8cdf44eac8dd3eb7bedfd80d1 : FAILURE in 13m 53s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/86a2231406f44508bb6dd17560786dc5 : FAILURE in 10m 42s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/b306dfc79f4549c6b92d720674def021 : FAILURE in 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"37c9ebb0bfdc8f87197b4c147d73ea10833fff5f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-05 16:39:18.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"6eb49f5679dbc2102569483238ec286a59b956dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 16:39:31.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"5326a1cbc4783d998bd63d05169df157e62bd690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 16:39:39.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"9c8c8a26cbc3e01cc435cb99e738c96aaea7e633","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-03 09:56:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4c8eb33fd41c2915838c3f33fb71fae814a94a2b","revisions":{"66b9ab4c49e5203431d24f848b1a8fe351ba3782":{"kind":"REWORK","_number":1,"created":"2020-06-19 17:19:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/737051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/737051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/737051/1"}}},"commit":{"parents":[{"commit":"c42b882aff3d8c911157b52f705dc755294f23f6","subject":"CI: use wheels from infra mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c42b882aff3d8c911157b52f705dc755294f23f6"}]}],"author":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-19 17:18:09.000000000","tz":0},"committer":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-19 17:18:59.000000000","tz":0},"subject":"WIP: Introduce curl macro - fetch_url","message":"WIP: Introduce curl macro - fetch_url\n\nChange-Id: I68e0cde16bf84ca5743c8c8dd6b6460752587e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66b9ab4c49e5203431d24f848b1a8fe351ba3782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66b9ab4c49e5203431d24f848b1a8fe351ba3782"}]},"branch":"refs/heads/master"},"49f9d27b1e91e6ad5ca384e48fdd047d5eda43f8":{"kind":"REWORK","_number":2,"created":"2020-06-22 18:09:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/737051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/737051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/737051/2"}}},"commit":{"parents":[{"commit":"c778749bc79bc235d58730367e8c8b8448f96512","subject":"Merge \"Fix horizon debian binary docker image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c778749bc79bc235d58730367e8c8b8448f96512"}]}],"author":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-19 17:18:09.000000000","tz":0},"committer":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-22 18:09:12.000000000","tz":0},"subject":"WIP: Introduce curl macro - fetch_url","message":"WIP: Introduce curl macro - fetch_url\n\nChange-Id: I68e0cde16bf84ca5743c8c8dd6b6460752587e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49f9d27b1e91e6ad5ca384e48fdd047d5eda43f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49f9d27b1e91e6ad5ca384e48fdd047d5eda43f8"}]},"branch":"refs/heads/master"},"e0b0fcade42e36149db4fb9c28723c3a1ee2512d":{"kind":"REWORK","_number":3,"created":"2020-06-23 07:59:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/737051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/737051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/737051/3"}}},"commit":{"parents":[{"commit":"c778749bc79bc235d58730367e8c8b8448f96512","subject":"Merge \"Fix horizon debian binary docker image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c778749bc79bc235d58730367e8c8b8448f96512"}]}],"author":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-19 17:18:09.000000000","tz":0},"committer":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-23 07:59:08.000000000","tz":0},"subject":"WIP: Introduce curl macro - fetch_url","message":"WIP: Introduce curl macro - fetch_url\n\nChange-Id: I68e0cde16bf84ca5743c8c8dd6b6460752587e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0b0fcade42e36149db4fb9c28723c3a1ee2512d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0b0fcade42e36149db4fb9c28723c3a1ee2512d"}]},"branch":"refs/heads/master"},"4740cff46b264d41e134d4f959d8b56e6aab6b1c":{"kind":"REWORK","_number":4,"created":"2020-06-23 09:49:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/737051/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/737051/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/737051/4"}}},"commit":{"parents":[{"commit":"c778749bc79bc235d58730367e8c8b8448f96512","subject":"Merge \"Fix horizon debian binary docker image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c778749bc79bc235d58730367e8c8b8448f96512"}]}],"author":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-19 17:18:09.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-23 09:49:07.000000000","tz":0},"subject":"WIP: Introduce curl macro - fetch_url","message":"WIP: Introduce curl macro - fetch_url\n\nChange-Id: I68e0cde16bf84ca5743c8c8dd6b6460752587e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4740cff46b264d41e134d4f959d8b56e6aab6b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4740cff46b264d41e134d4f959d8b56e6aab6b1c"}]},"branch":"refs/heads/master"},"7f3aaa0458232990a3af8a53ba2d24767f1ac17a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-06-25 09:49:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/737051/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/737051/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/737051/5"}}},"commit":{"parents":[{"commit":"589bc5edec0cc5cb36e2e5afd48403e2a4755450","subject":"Move out sources and users to separate files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/589bc5edec0cc5cb36e2e5afd48403e2a4755450"}]}],"author":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-19 17:18:09.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-25 09:49:01.000000000","tz":0},"subject":"WIP: Introduce curl macro - fetch_url","message":"WIP: Introduce curl macro - fetch_url\n\nChange-Id: I68e0cde16bf84ca5743c8c8dd6b6460752587e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f3aaa0458232990a3af8a53ba2d24767f1ac17a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f3aaa0458232990a3af8a53ba2d24767f1ac17a"}]},"branch":"refs/heads/master"},"2b5df23c050b2b4a6f06f873260fb44f99c7c4ff":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-07-03 11:15:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/737051/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/737051/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/737051/6"}}},"commit":{"parents":[{"commit":"f976c92dde327bc07e7e4acd0eb742a6ff65a676","subject":"Move out sources and users to separate files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f976c92dde327bc07e7e4acd0eb742a6ff65a676"}]}],"author":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-19 17:18:09.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-07-03 11:15:21.000000000","tz":0},"subject":"WIP: Introduce curl macro - fetch_url","message":"WIP: Introduce curl macro - fetch_url\n\nChange-Id: I68e0cde16bf84ca5743c8c8dd6b6460752587e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b5df23c050b2b4a6f06f873260fb44f99c7c4ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b5df23c050b2b4a6f06f873260fb44f99c7c4ff"}]},"branch":"refs/heads/master"},"4c8eb33fd41c2915838c3f33fb71fae814a94a2b":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-11-05 16:39:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/737051/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/737051/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/737051/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/737051/7"}}},"commit":{"parents":[{"commit":"f976c92dde327bc07e7e4acd0eb742a6ff65a676","subject":"Move out sources and users to separate files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f976c92dde327bc07e7e4acd0eb742a6ff65a676"}]}],"author":{"name":"mnasiadka","email":"mnasiadka@gmail.com","date":"2020-06-19 17:18:09.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-05 16:39:07.000000000","tz":0},"subject":"Introduce curl macro - fetch_url","message":"Introduce curl macro - fetch_url\n\nChange-Id: I68e0cde16bf84ca5743c8c8dd6b6460752587e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8eb33fd41c2915838c3f33fb71fae814a94a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8eb33fd41c2915838c3f33fb71fae814a94a2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
