)]}'
{"id":"openstack%2Fkolla~737131","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Id05733f31219ab3a788e16954f40600d00298662","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Id05733f31219ab3a788e16954f40600d00298662","subject":"neutron-server: remove version limits from networking-hyperv","status":"ABANDONED","created":"2020-06-20 14:24:21.000000000","updated":"2020-06-26 13:01:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc6f2066530ce0665f4a10f7157efb2e095619bc","_number":737131,"virtual_id_number":737131,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-20 14:24:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2020-06-20 16:39:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-20 17:30:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-20 16:39:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-20 17:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f06c608e186120d501f91103b3711df4f7c57d7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-20 14:24:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b74463e1e6353bde2b5297152b1bb6724294203","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-20 14:24:54.000000000","message":"Patch Set 1:\n\nor we should drop networking-hyperv completely as it shows that no one is using it probably.","accounts_in_message":[],"_revision_number":1},{"id":"363ade11ea3b843977f2aa4bd63822c7d1eba838","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-20 16:39:49.000000000","message":"Patch Set 1:\n\nwe are dropping it in victoria","accounts_in_message":[],"_revision_number":1},{"id":"704c7685f458b7ead294554547971f4b0c2bc958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 17:30:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/596089857b2247a3ae6fa7146c460a34 : SUCCESS in 44m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e485d4de3d234ebab82d53108b71f2b8 : TIMED_OUT in 3h 01m 35s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/345d6c3d086045e3bc720c9c46bef2df : SUCCESS in 2h 59m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/582557814e064c8d95ec41cf4aa41c14 : TIMED_OUT in 3h 01m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e6b4821dc69b48ed9e2339b733733fdd : TIMED_OUT in 2h 04m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7b89c816ff11436c9d08420f5f4914c2 : TIMED_OUT in 2h 04m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dc82027fadc943a48f369a52a19f39dd : TIMED_OUT in 2h 04m 31s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/61b5322bc6bb4b5dbab469a1364bc172 : TIMED_OUT in 2h 03m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e82c23ec091c40e1b33d67999bf3fa72 : TIMED_OUT in 2h 34m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6108fadc0714425a8a996cf73d5916e1 : TIMED_OUT in 2h 04m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe18339e48ef4fb588c25aae04e1c20f : TIMED_OUT in 3h 01m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9287cf577730436ea23f7cf7f40ac089 : SUCCESS in 2h 57m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/51f43e110b5c4ef7881e220533402464 : TIMED_OUT in 2h 03m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d83a93885c824170abc497114ff83e16 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3623f674cc824760b6d6c04d502fdcd6 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f836679608f04e22bc38f2ead4a8a6d2 : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b038e08bdb441c2bc1e24d35f0925a1 : SUCCESS in 8m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01404474725749a4a2343d8ab02db9d8 : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa715e80f6f94c6088cc1e14a1531c6d : SUCCESS in 23m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/08b099d1f1a5422aa0511842676311a9 : TIMED_OUT in 3h 01m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ac0a1aabbca48c792d0287b0ce8c50c : SUCCESS in 3h 01m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1279d534971649ac98abbec643557e42 : TIMED_OUT in 2h 04m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a20f90726744898acc3c2298c8c96dc : TIMED_OUT in 2h 03m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4a5672d0aa7c401fbca9d5d03e9e7bf0 : FAILURE in 1h 18m 44s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"dc6f2066530ce0665f4a10f7157efb2e095619bc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-26 13:01:57.000000000","message":"Abandoned\n\nwe are dropping it in victoria","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a91a619bbd3402bec50bca812d875d3311242c84","revisions":{"a91a619bbd3402bec50bca812d875d3311242c84":{"kind":"REWORK","_number":1,"created":"2020-06-20 14:24:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/31/737131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/737131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/737131/1"}}},"commit":{"parents":[{"commit":"0a6cd606c68281b6ef8301ab7a8e954440a6714a","subject":"kolla-toolbox: use u-c while installing virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a6cd606c68281b6ef8301ab7a8e954440a6714a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-20 14:23:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-20 14:23:23.000000000","tz":120},"subject":"neutron-server: remove version limits from networking-hyperv","message":"neutron-server: remove version limits from networking-hyperv\n\nWe install 5.x.y version from Pike each time. Install latest one then.\n\nChange-Id: Id05733f31219ab3a788e16954f40600d00298662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a91a619bbd3402bec50bca812d875d3311242c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a91a619bbd3402bec50bca812d875d3311242c84"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
