)]}'
{"id":"openstack%2Fkolla~737198","triplet_id":"openstack%2Fkolla~stable%2Frocky~I128fb04a5ddeb77428697d33e2015158bc74738f","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I128fb04a5ddeb77428697d33e2015158bc74738f","subject":"Add /var/lib/kolla/venv/bin to Zun exec_dirs","status":"MERGED","created":"2020-06-21 23:31:10.000000000","updated":"2020-09-04 11:06:38.000000000","submitted":"2020-09-04 11:06:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737198-1599217598340-87da3b05","meta_rev_id":"114508224d5c0116b3c6895108d1af0205dcb06c","_number":737198,"virtual_id_number":737198,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 11:06:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-09-04 07:37:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":2,"date":"2020-09-04 07:37:26.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-09-04 07:37:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-04 07:37:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-09-04 07:37:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 10:31:30.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-08-30 15:50:09.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-08-31 01:16:59.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-09-04 07:37:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-04 11:06:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45def8b9c1b640fc1370bed6afe1725da5b6a22a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-21 23:31:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6066eb6eec2ca92675ddcab66fcfd97dfec4e4e4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-21 23:33:39.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7b58293305ac261aca4b9704166e2116eaa815d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 00:52:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/29391623a97d4129a13c0c7a21f2f5b2 : RETRY_LIMIT in 2m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8603e7c8a8544685b0bc1a0f30ac44d2 : RETRY_LIMIT in 1m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fc7a78ef294d4d7bae6e83aea9636a56 : POST_FAILURE in 3m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6b5e1aa0b8e647bebd2839b0d52141ab : POST_FAILURE in 3m 09s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b490266738a64c27bae565e8da7f2cff : RETRY_LIMIT in 1m 48s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b4992e745a714d35bf38524cfc429d8d : RETRY_LIMIT in 2m 56s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/da912e92c5fb484281b3c1f064885ef3 : POST_FAILURE in 2m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/cc63f9d0ee7e4936906205d4c9c1711d : POST_FAILURE in 3m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22fc8ae58b2f46a697e618b46bb3eaeb : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e3bad412a9e4ada994e836edaf4754d : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/592183e884ad4870bf840ab36070268e : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/75c154c3f5d84312b3e95701c77273b1 : SUCCESS in 4m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/40f521749c644545be57f6d924b4ad3f : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d8a2f2ad639417aa7de5d3680857872 : RETRY_LIMIT in 2m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d79f0a0f8694da49e79764e75087201 : RETRY_LIMIT in 3m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/379bd12a315d42178c760dcc5983a203 : SUCCESS in 57m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b8dfdde96b744d1cb93ca45d5d79ff9d : SUCCESS in 1h 18m 08s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"f0c8adae8e497ef65d2087b84e728f0b4e9bb329","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-22 01:03:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7e2392fdbdd9cb1972c61293e6c8b4f93233e879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 02:58:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/80edf6619c57451f9f4da440e7ec6284 : RETRY_LIMIT in 2m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/632adcd67ebe45199553c4ee3faf3a1c : RETRY_LIMIT in 2m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e522c061aa834dbbb24d45318d532502 : POST_FAILURE in 2m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/48609cc3e1fc4cb9af0a98203d58a28e : POST_FAILURE in 3m 12s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/0b95a885816e44b999bdc92c080105cc : RETRY_LIMIT in 2m 07s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/10299c6fb0bd48e0a5a5851648f3b03c : RETRY_LIMIT in 2m 10s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5edaf9cb467e42fc81feafae9bb92a03 : POST_FAILURE in 2m 50s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9d8ad75559024785a9a09279f00e5162 : POST_FAILURE in 3m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7665d7fd362f49e0be221227176ae0e6 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/605ba2695a694ac9ab7e819fb3402f83 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/287b31af7aef41c49d367cf5613267ff : SUCCESS in 3m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/993a784cdb004f079d66c1581d31358d : SUCCESS in 4m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b82b66521ba04618a5a17dd86197c786 : SUCCESS in 4m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7dffa145e3d84526bcd70f7e124bfa93 : RETRY_LIMIT in 2m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ac196efc2454d579d42912210c83db8 : RETRY_LIMIT in 1m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5974f6e83a6e465cb8a6d5ba28495e83 : SUCCESS in 55m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/01f98feb86cf45d4887b4f385a624e1d : SUCCESS in 1h 49m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"9e4ac9d6735ee2da5e44e30491743cb0aac9a18c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-22 03:34:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6d5f18404d8b115fdaaf5853b2f42757227c359f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 05:29:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a2047689df374a95aeda20342fe2ab72 : RETRY_LIMIT in 2m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5ce0d072da0b471c860945880d8c09a8 : RETRY_LIMIT in 2m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3469ab1a807b45cfa908fa61353038a0 : POST_FAILURE in 5m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/57670b439a86449cba20a0e6d2a5ebdc : POST_FAILURE in 5m 00s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d57f7d533035404684beddde6be0579a : RETRY_LIMIT in 2m 05s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/247f951f3ecd4f2aa578e3902790cec5 : RETRY_LIMIT in 3m 24s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/51fb4d9f7a7d489f917a35f4bab6eb1c : POST_FAILURE in 3m 26s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/59a7d06d3e0f4f5babf818d3ad59087e : POST_FAILURE in 3m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d86bbea412440559641fbefbb3c093f : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/781c60de29f8433fafdb018d7587158a : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0987e8b48854c988422a1d117031285 : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/407d0f81c6b446228479de95ca99906e : SUCCESS in 3m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dcc7cf80194e40279478b81d2c0d05e1 : SUCCESS in 4m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ef58c02462a43d080f2ea018383c30a : RETRY_LIMIT in 2m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2475ff3f373d48eab8cd068b7fc4c334 : RETRY_LIMIT in 2m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7c83e5ef4504b9d9f36d20864993faa : SUCCESS in 57m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73fa782ae93145e0bfa079ea2e26db31 : SUCCESS in 1h 49m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"25c03098c3d432c64734e8d4e54eb149ea97671d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-22 08:25:07.000000000","message":"Patch Set 2: Code-Review+2\n\nbut we need to fix the gate breakage","accounts_in_message":[],"_revision_number":2},{"id":"bac7ef88bb784b61f12c4094ee4f3c636e466d8d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-24 19:35:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4edfdb289e6f543bc0b8fb5415c2e556710b5b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 21:45:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4e0d73cb761c438e934ddd73e6e0deee : RETRY_LIMIT in 2m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/545cc605dd3043978a9e5e607ce8d816 : RETRY_LIMIT in 2m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bb572431a2c5400ab48604492cc10176 : POST_FAILURE in 4m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/adc0e8adfb534e7c97d89092e06ecd98 : POST_FAILURE in 3m 57s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/8bbe0b633a68441180e2a3e03d96a219 : RETRY_LIMIT in 3m 33s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c0b2ff7d9a384c9897b54b25757515a2 : RETRY_LIMIT in 8m 17s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/05a974f1986d406c851e2cb5eefa9702 : POST_FAILURE in 4m 59s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/cb2cc363c75f470f9420fd086e78d8fb : POST_FAILURE in 5m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3ac732428e643e283b1d9c782e532e4 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d563fcc781c436aa7c8c0d0f816bc3f : SUCCESS in 7m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2073105a0984a46b7a5c247904a33a0 : SUCCESS in 9m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2420ea400495485a916339828d36532d : SUCCESS in 8m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e8b97e67bd2047cd81ff8e3bd65e7ca7 : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ff5f887e6e444f73be14a30d2355e66f : RETRY_LIMIT in 3m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c219aa87776c4b7a8f2a6a20654f169c : RETRY_LIMIT in 2m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/482c25ee7c3b44d992fe007b44d4f590 : SUCCESS in 1h 06m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1ef4ed2072d486a9083d8017544f4e9 : TIMED_OUT in 2h 01m 57s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"26a6f4c0ac68a9df5a5a2efcc069a81764d773a9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-26 00:46:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bb074261116b488ce9fbbd0c49f4f280e05ec5e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 02:53:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bf7ffddaabf64e3f9086acee49c33247 : RETRY_LIMIT in 3m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fd8ce3bc01214505940b06c8f2c95a2a : RETRY_LIMIT in 2m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d4ff9c899e364ba39b75b839f6f169d6 : POST_FAILURE in 3m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0bd7474ae8a846288ab69ced8919e885 : POST_FAILURE in 3m 50s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bb0295dcd7244e4da1cd94a9e0057f32 : RETRY_LIMIT in 2m 10s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/42ecb90a1e094c4ea98ffe905dd8d1ef : RETRY_LIMIT in 2m 20s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d31b35a149f64d5ea62acde8d6f79205 : POST_FAILURE in 3m 38s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/37d3a35dcc3f4d2393f9bf764c813875 : POST_FAILURE in 3m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b92218f1791c48be97303c61a8cfb64a : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e14d2ed712419fa76b2d4ace49600b : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f996f0269a79488cb226481efbc8bca5 : SUCCESS in 5m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/018937b9cb194b298dc609937633136f : SUCCESS in 4m 28s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f287d07e93f84a04bfbe770fb5c62a37 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/77f262469e6e4d2a81a921c9f8dfb05f : RETRY_LIMIT in 2m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32f83e782d64419da8d4d4037160d6c5 : RETRY_LIMIT in 2m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4baf53fe5682465bba96c1afb48d3118 : SUCCESS in 1h 01m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45c1b1e9261545cdb6729f1edfd06478 : TIMED_OUT in 2h 01m 25s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"14f035af7384e3c9f7dd6536bd32dec0cc793fda","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-01 17:02:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"25d2f08412932bd800b4fc3c7aaf1a2478caabec","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-01 17:09:24.000000000","message":"Patch Set 2:\n\nCI is fried, will let you know when fixed","accounts_in_message":[],"_revision_number":2},{"id":"d039ba045e702d568bd7d24e1f48daee329bcc11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 20:43:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/446281d4e32c45da83a9349fe7b071dc : RETRY_LIMIT in 3m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/88ae7297282c4652b4a63523d1196e2c : RETRY_LIMIT in 4m 46s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/63b80c3d5d584a3dbaf833095386ea83 : SUCCESS in 47m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/913f9616c286463890e6092bcf24a0d9 : SUCCESS in 1h 13m 42s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/f286858ec58b40ecad650ed64255718f : RETRY_LIMIT in 2m 54s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/51a28c00290540d58cc369f9a9e13521 : RETRY_LIMIT in 2m 33s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/44e488d21092465fa7c94a5d0a9ad788 : SUCCESS in 1h 26m 01s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/71aaf47c64fc414997b5a299c47ab605 : SUCCESS in 1h 12m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b6a1b5c50d4abfa7e7656f1c080c46 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02e50d9b8bd841a2a0d7c930bdf56a3c : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d68bd8d80cd4dc98c7e0e3749c17b1a : SUCCESS in 7m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f4bcf669aeb84533b09a29d046bba975 : SUCCESS in 5m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d7149f54e7c54352a5c4c2059dc5d2a2 : POST_FAILURE in 36m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73e51239710a4adaa09735b1307d1a64 : RETRY_LIMIT in 3m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/29c044ea855b4bdb9ec2fd040f70d564 : RETRY_LIMIT in 3m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc6d3ceb9b4c4490825c6ce70d241911 : POST_FAILURE in 3m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4dbd447a7cb546729cf477e17a5a9d5b : POST_FAILURE in 35m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"a1204d966e1d0c47089c5a0371adb165fe944d76","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-03 10:31:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ca2eaaee82b50fb90d98fb4c3a717b4aa00d3dd1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-03 10:31:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"03febd48c0ec0f3825763980d8063f8dfb083d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 12:43:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d57b571c9d2b404fa504254632e37791 : SUCCESS in 35m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/220f28dfaa094073833d1f3fcb5c2520 : FAILURE in 1h 02m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c14c2b8cdb8a412bbf7cbe9d13ccedae : SUCCESS in 1h 06m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e213815f495f4d1e8e6fc09f542d6e23 : SUCCESS in 1h 21m 01s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/af1a1d51183d47bd9753874193a21bec : SUCCESS in 28m 27s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2d2fd48c516d4fcdb14e69fc6713863a : FAILURE in 41m 32s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8f21ced660db4efc8c6f5b4bfa37f204 : SUCCESS in 1h 13m 48s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/c253809097bd43618aba6798e9181a0c : SUCCESS in 1h 05m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faf82ea35ede4c25aa84d12b0b2d4dcd : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08c61fbefc847b5b52a5338467d0640 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27de46ae797a4cb6a0f8f405c3b38861 : SUCCESS in 4m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/14468b388d5440329e0180adff0f30ba : SUCCESS in 5m 28s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/894fa73c7ede4ab5859d211d19362b08 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c017653d0f2540b3806cf1b8af811045 : SUCCESS in 1h 29m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e1c9e9c800a473a875666a002de483a : FAILURE in 55m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b57cfd68f11447c59abeba3bbf635584 : SUCCESS in 57m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6bfea8665a1643d18780e1e48afd161f : SUCCESS in 1h 53m 42s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"a5e90e59aa80b2813c96c5d0be8e4f67f4b0d194","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-03 16:41:21.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1490d9fc5e08b3653e0d7ad5f2f98a303458fdbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 19:08:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1a4b11b0195540eea79eb16693ee6b9e : SUCCESS in 38m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0be37da455ec4b5da104d932650891b5 : FAILURE in 1h 12m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/63d96e777ef34629916c87f978a6fd0f : SUCCESS in 1h 06m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ea2c6927d5de4cffb0e02b2fb55e39e7 : SUCCESS in 1h 06m 23s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/946a1712417949498872c89301b33df3 : FAILURE in 2h 06m 37s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/9469d32211f448258882ea4f80a8d189 : FAILURE in 50m 00s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/67e3c62c8c3c42d7824100c080a2b072 : SUCCESS in 47m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/997fff68988640528a36d434a382cef7 : SUCCESS in 1h 03m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65d2dcc6af2c44a7813ff5b9a516a9ff : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9b822d96c414a9c889e5b2426e4c422 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a5fb78002934523aa15fa7fb424949b : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/78fb54f879114ca795077c3437b61ac1 : SUCCESS in 4m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8287634205a84ccaa51123600a846f8d : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/548356e4415d4a2eb848b3ba4363a999 : SUCCESS in 46m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32674e454e2e41949e95769c55c0ee31 : FAILURE in 31m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/937d65347b4740539a00b945c324e2fb : SUCCESS in 1h 03m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b904cdcde66b4787917d4f00e6927467 : SUCCESS in 1h 52m 01s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"de5a6bb6289d5bfed03af71a0a054ce9c4d1b91c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-21 16:34:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"04da85badf2e23017039cc61c30eb68ef06b1271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 18:26:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cb1a102aae8641b380ff3c7d58d68a25 : SUCCESS in 43m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1da3d20d156648759338b0e6b79c3907 : FAILURE in 46m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/92ffc3fe9d2a4b6595325fbda8ee523b : SUCCESS in 1h 38m 07s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9113ab9aea554b0b8e1b067b3738e799 : SUCCESS in 57m 21s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/877acabf7be94a878e8cc62e704dfbd2 : SUCCESS in 45m 07s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/82522603d3bf4e88891a24ff528b5547 : FAILURE in 57m 28s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7dbb32a4ec2041f28c49b41398f375ad : SUCCESS in 1h 25m 34s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/185551b443d74a52b23ea160be169d4b : SUCCESS in 1h 04m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/952be466601c440c837a76537c2329cb : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d2237efdac4a1cb79b8ba280062b53 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5f800b7eac2405090807bf9fc6419b2 : SUCCESS in 5m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/67bec52436914a888d4fdfec891288dd : SUCCESS in 5m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4e61f1cdd52445a6bedb9ed6bffb90c2 : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2d35120d60e5470180b81bb1cf58d57c : SUCCESS in 1h 29m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c5d135ace134132bfe99b69a87f8762 : FAILURE in 28m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ac5a53ee4da4240b72291bcd768ec87 : SUCCESS in 57m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3807a4f1cf994a1b8c96e926b8255f69 : SUCCESS in 1h 46m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"5264bf1569ef208f1d4515b7f895ab8fb08c6b96","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-30 15:50:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"57f007d4b24c9cf9ef5be79a6373b9d96a1137cc","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-31 01:16:59.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b7cd79e5dbc7d2abae4b3dc14054066eec053153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 03:10:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d597e353a80c413daae2e2204b056b22 : SUCCESS in 48m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7e7639ec323e47388ee4cf2c9ecedcf9 : FAILURE in 55m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6f97e36cea67427c8cbe31856371e9af : SUCCESS in 1h 11m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/215b6f3cff474817bbb5f98902b79fe6 : POST_FAILURE in 11m 31s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/14860e17be79461bbe62f96f8409f760 : SUCCESS in 43m 18s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/651e1397fd834fe3af899eb358e28701 : FAILURE in 38m 17s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/98933da02865474aabea1ba0436f692b : SUCCESS in 1h 10m 19s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/e37cdd228fb8437cb975484b3de1f2cd : SUCCESS in 1h 08m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59b9ca2b1c5b40668593c12f20d36a18 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1a7e4598604a9d91b5546c74bc4a9a : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dad9aa7289664c4dbd16f84988547ab4 : TIMED_OUT in 40m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d18a93d6fa6b43d5898868ecfa295e73 : POST_FAILURE in 4m 05s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/423ad971ef934679bb1d6f3eba2dd637 : SUCCESS in 4m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e4c5479062145c18fc728548f366221 : SUCCESS in 1h 23m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8bc31690e0bf45f79c96cebb7b8fdb18 : FAILURE in 39m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5cf32be88f954ea9a23725fd51d8d997 : SUCCESS in 59m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e1eee388d564e9ca64a5b823b55b479 : SUCCESS in 1h 49m 34s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"0ccb041e537bd233a36914956c09413dbf06c454","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-04 07:37:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"927f51433381886aafd87d0dcdc9a0a11f13e1de","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-04 07:37:30.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"91b65cce0cb3f43a6e97f98634e9eb9f8dd4da2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:17:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fb2e1f1dfd5a4eb1a69ec9c75fa225fd : SUCCESS in 37m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9960d1df32234c38b54be01f9bf1a230 : SUCCESS in 54m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5a974267a0104a3099534d72405a698d : SUCCESS in 1h 25m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8756ac596d0a4d8d9a87765c4f689c45 : SUCCESS in 1h 08m 10s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/241168f524564a2ea10f862b025782e1 : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c28121cc69864801af65f6742af6566a : SUCCESS in 58m 50s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7d96dca9ea7148df8eb6961387f02e84 : SUCCESS in 1h 16m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bc70cdcf7d2446ba82446aaa8b87ba68 : SUCCESS in 1h 14m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8ef6024d21d4450b76bf7ed3c9fa735 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9bb4c26f0a74ea9a3f46e75f85a1a64 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6cdd13945159499db2ec75bacb934397 : SUCCESS in 6m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/474e8f8b428747549fc3ade7b269c1c5 : SUCCESS in 6m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f996f9ab35fc4f17ab62e252a86478cf : SUCCESS in 6m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c4d511b3f6c4dc3a924801d6524e565 : SUCCESS in 1h 25m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c4f6921a88e43658e720be5603eab7c : SUCCESS in 36m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/39d132703efc4c8196ce2d8cac81151e : SUCCESS in 47m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/93ecad93231845538f6bdb5cc3dd37b4 : SUCCESS in 50m 10s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"8823521385328428ca02d62969f34f3e73a30ebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:34:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2a1f632d0698222a5f39fd205d15e5bf4b5739a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:06:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2d31c9b8da0c4197a19713c19d83b731 : SUCCESS in 47m 01s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/48b58dc992ad4478a763c010cbd2532e : SUCCESS in 56m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/909d7b3412724e3994d5a0bfa2658504 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aaf2425ddd64b5b838bb63afd507a82 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/accd0142f6294305a493fa7fc7fd9d11 : SUCCESS in 4m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5f86db19b3bf414dac66a622b98e64e4 : SUCCESS in 4m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/586e9d2a27da46258b071b18d88f17a6 : SUCCESS in 4m 47s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8081cca03e34d49aca66bd2c915deca : SUCCESS in 32m 01s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"16346fe5157e5c88841d74deeb1eaf24eb2f76cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:06:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dd0069addbfba430d1a8cf15e91005d0f0279c11","revisions":{"07235fb131adbfaa0d3a67d07ec1bc6e7a9501f5":{"kind":"REWORK","_number":1,"created":"2020-06-21 23:31:10.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/98/737198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/737198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/737198/1"}}},"commit":{"parents":[{"commit":"03212577ef09c5d1a613b6277c5d48d6ca4c03c0","subject":"Fix multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03212577ef09c5d1a613b6277c5d48d6ca4c03c0"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2018-09-30 16:14:31.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-21 23:31:10.000000000","tz":0},"subject":"Add /var/lib/kolla/venv/bin to Zun exec_dirs","message":"Add /var/lib/kolla/venv/bin to Zun exec_dirs\n\nThis is necessary for rootwrap/privsep to work properly\n\nChange-Id: I128fb04a5ddeb77428697d33e2015158bc74738f\n(cherry picked from commit 570a6120be48dde70a793be52574a0ae1e0fce6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07235fb131adbfaa0d3a67d07ec1bc6e7a9501f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07235fb131adbfaa0d3a67d07ec1bc6e7a9501f5"}]},"branch":"refs/heads/stable/rocky"},"f6d6e8c64c3aa1e4c49a9e6d1004a48968f8da8b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-21 23:33:39.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/98/737198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/737198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/737198/2"}}},"commit":{"parents":[{"commit":"03212577ef09c5d1a613b6277c5d48d6ca4c03c0","subject":"Fix multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03212577ef09c5d1a613b6277c5d48d6ca4c03c0"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2018-09-30 16:14:31.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-21 23:33:34.000000000","tz":0},"subject":"Add /var/lib/kolla/venv/bin to Zun exec_dirs","message":"Add /var/lib/kolla/venv/bin to Zun exec_dirs\n\nThis is necessary for rootwrap/privsep to work properly\n\nCloses-Bug: #1883604\nChange-Id: I128fb04a5ddeb77428697d33e2015158bc74738f\n(cherry picked from commit 570a6120be48dde70a793be52574a0ae1e0fce6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6d6e8c64c3aa1e4c49a9e6d1004a48968f8da8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6d6e8c64c3aa1e4c49a9e6d1004a48968f8da8b"}]},"branch":"refs/heads/stable/rocky"},"af40df9e6011d65a153f6d31f99bdd90cc726fd6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-03 16:41:21.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/98/737198/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/737198/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/737198/3"}}},"commit":{"parents":[{"commit":"5b64da9f319162c9d17e2894d8f5b0658d661e75","subject":"Fix multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b64da9f319162c9d17e2894d8f5b0658d661e75"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2018-09-30 16:14:31.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-03 16:41:21.000000000","tz":0},"subject":"Add /var/lib/kolla/venv/bin to Zun exec_dirs","message":"Add /var/lib/kolla/venv/bin to Zun exec_dirs\n\nThis is necessary for rootwrap/privsep to work properly\n\nCloses-Bug: #1883604\nChange-Id: I128fb04a5ddeb77428697d33e2015158bc74738f\n(cherry picked from commit 570a6120be48dde70a793be52574a0ae1e0fce6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af40df9e6011d65a153f6d31f99bdd90cc726fd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af40df9e6011d65a153f6d31f99bdd90cc726fd6"}]},"branch":"refs/heads/stable/rocky"},"dd0069addbfba430d1a8cf15e91005d0f0279c11":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-04 07:37:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/98/737198/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/737198/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/737198/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/737198/4"}}},"commit":{"parents":[{"commit":"965a288da2e3c0f170167279990ddeeb5f1de054","subject":"Fix ironic build issues - LANG related","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/965a288da2e3c0f170167279990ddeeb5f1de054"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2018-09-30 16:14:31.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-04 07:37:26.000000000","tz":0},"subject":"Add /var/lib/kolla/venv/bin to Zun exec_dirs","message":"Add /var/lib/kolla/venv/bin to Zun exec_dirs\n\nThis is necessary for rootwrap/privsep to work properly\n\nCloses-Bug: #1883604\nChange-Id: I128fb04a5ddeb77428697d33e2015158bc74738f\n(cherry picked from commit 570a6120be48dde70a793be52574a0ae1e0fce6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd0069addbfba430d1a8cf15e91005d0f0279c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd0069addbfba430d1a8cf15e91005d0f0279c11"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
