)]}'
{"id":"openstack%2Fkolla~737238","triplet_id":"openstack%2Fkolla~master~I036cfe3af3dd96625f986a820c78a10e6558b086","project":"openstack/kolla","branch":"master","topic":"bug/1884484","hashtags":[],"change_id":"I036cfe3af3dd96625f986a820c78a10e6558b086","subject":"Add sysfsutils to nova-compute ubuntu source","status":"MERGED","created":"2020-06-22 09:08:10.000000000","updated":"2020-10-14 09:36:22.000000000","submitted":"2020-06-26 15:10:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737238-1593184201753-ccd9e89b","meta_rev_id":"b193ec39e43f3329a67ccac73cc843277bf2ae41","_number":737238,"virtual_id_number":737238,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 15:10:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-22 12:26:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-06-26 13:42:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-22 09:12:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-06-26 13:42:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 09:12:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-22 12:26:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-26 13:42:39.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-26 15:10:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"404f9866ceb569ec6eb167b7c3fa2d512e1dc182","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-22 09:08:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e6645df51b268ff4ed1c6c42ef8f30c80b76077","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-22 09:12:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6b4fd03b6424755dc70a8e82a29f7786bb0f5ace","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 10:38:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/53157655590c48e3b9856f702089fa84 : SUCCESS in 53m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb8886a7e06c4753ab1e1d31d0c34dfa : SUCCESS in 27m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea9cd677e05c43438fc834a697a382fb : SUCCESS in 1h 25m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bfe9280b0034607bcb3fb2e65032a88 : SUCCESS in 28m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b97ffa49060a26707416e5b30249ed7a21e3dc45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 10:39:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ff5ac7cbccd8492d8e3ca68dcf493b75 : SUCCESS in 30m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/40c46fa7be9a4e938c6db276d23c29c6 : SUCCESS in 38m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e18deb502d1049669c28c03f65d1b830 : SUCCESS in 59m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2fac205d825548428fd9d20a56570853 : SUCCESS in 55m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/933646cd92444cf4b926e2c8374ad44f : SUCCESS in 1h 27m 32s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/90ee241a67294bd688b7e178efff222a : SUCCESS in 42m 56s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1346d99dcda543f1aa24289e34379256 : FAILURE in 20m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3ccfb09f671741ca8e17ebe59cf5ca0b : SUCCESS in 43m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1e6da9f3ce484818ab011c21989650d2 : SUCCESS in 1h 19m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc69d23d4cc242fab92d39d950c9b51c : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c8575f86cf4fca9a45e0ab56345bf6 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69bf273ccf0548f39e2e5091e5c9f36d : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7749e15b18694c1eb94a0386d7fc5518 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876ae38be8034a6a95fd63165b4ede2a : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84d0634c8b014e0eadcf882edcaa6ef1 : SUCCESS in 4m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db467993b03b44fa897104f4abee11c0 : SUCCESS in 26m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/695807c3dbc84ed7842d103a78094d00 : SUCCESS in 24m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c29d898698640bba4e0b2c3a3c37368 : SUCCESS in 1h 00m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68ffa1a10457457ab50e65a8e904c2dd : SUCCESS in 1h 04m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e205bb7bf93440ea446582b1587feac : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4d4a3cffac2a511afd6fa3e439134f002ab792de","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-22 12:26:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d3def39bba1b4536aef3f22ad764d4b4 : SUCCESS in 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fd30fbbfb9b82907d4f53fa82c586564371dc7a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-26 13:42:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37e4af0e20522fc906aa190ae9df565a52a5df90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 13:42:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"279e24ad550d2faef0bad08150d8af4493d0550e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 15:10:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"09783f8846e74dcd30c3c606986497d1f9143ca4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 15:10:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/28a2788704044b51905c55cc67407580 : SUCCESS in 37m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0a4d14b175fd4cd3b0bc9947f967f96a : SUCCESS in 1h 11m 34s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7f226ac2737f4a7f9e652340dfb70756 : SUCCESS in 1h 21m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/274f5abcd6134a9db3d6e45742ff2606 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66357fd6d0724e839ebaf42c510155a1 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77712e12803644aba35cf585e22fb96b : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89168b02eac94ee39c92af299bb2adc5 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8706679233a24b2591d7b5acb6c6847c : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a93759d751d4e86975814905e4fa8f6 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a8ebd2357654d55bed667e10a262b70 : SUCCESS in 31m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/62a2fbf1fb3c4ffd8b118fca72c61260 : SUCCESS in 1h 10m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b5d30ab2ccd7498aa48034764a2df1bb : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0f7c1c2e98527f3bbdc40aed96226032055c134c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 15:11:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f32c10c28549bfae4daa43e681aa8f : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1164213e53d4bda88f5549597229731 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1},{"id":"99d1ab241af7ddc2e50ad6dd45bfcc0f15909fab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 09:36:05.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit de7f3c25918439a97a9ab20b6135327938e74ca1","accounts_in_message":[],"_revision_number":1},{"id":"dd6d426501dce99e0c13ab65eb6d4f6178b4568e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 09:36:14.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit b38ed5d9f1925a3c5dd24c154e66f4dd22d0f9e2","accounts_in_message":[],"_revision_number":1},{"id":"0193e27ec1df6beb6cfc0cf2113dbad8dae8b166","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 09:36:22.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 28df01a8b55370bd512aedbef685e12aaaa4b4b0","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c8702c96f852eae7a1fb784a260c01c15fb86f3","revisions":{"0c8702c96f852eae7a1fb784a260c01c15fb86f3":{"kind":"REWORK","_number":1,"created":"2020-06-22 09:08:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/38/737238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/737238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/737238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/737238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/737238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/737238/1"}}},"commit":{"parents":[{"commit":"e364113adcb529cc06575133b0e377526516f715","subject":"Remove mongodb dockerfile from kolla","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e364113adcb529cc06575133b0e377526516f715"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-22 09:05:32.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-22 09:06:58.000000000","tz":120},"subject":"Add sysfsutils to nova-compute ubuntu source","message":"Add sysfsutils to nova-compute ubuntu source\n\nThis fixes the FC Cinder backend usage in Nova.\n\nChange-Id: I036cfe3af3dd96625f986a820c78a10e6558b086\nCloses-bug: #1884484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c8702c96f852eae7a1fb784a260c01c15fb86f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c8702c96f852eae7a1fb784a260c01c15fb86f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
