)]}'
{"id":"openstack%2Fkolla~737254","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ieca15bdb315d52e9547d798df11641ef36485b26","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"Ieca15bdb315d52e9547d798df11641ef36485b26","subject":"Fix horizon debian binary docker image","status":"MERGED","created":"2020-06-22 10:45:10.000000000","updated":"2020-06-22 15:29:14.000000000","submitted":"2020-06-22 15:27:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737254-1592839628736-4f349c0a","meta_rev_id":"580eb800adca6235ef8d67b238b02bd8539a9ae9","_number":737254,"virtual_id_number":737254,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 15:27:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-22 10:45:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2020-06-22 14:00:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2020-06-22 14:00:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-06-22 10:45:10.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 10:45:10.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2020-06-22 14:00:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-22 15:27:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c866db6811c149104cfe04912fe01f6f03e10a43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-22 10:45:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dffffdc478cb8d69e05249985bbd63ce443f4e5e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-22 10:45:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"477fd0cc389e904bd519fc19c7532b4651aff0c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 12:12:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f3076523c723468b93490c08b8a44b9e : SUCCESS in 17m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/accda7e8da864ca7a4441fba98a2fc6f : SUCCESS in 45m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/74c67941d9204c9d8128c4a2188c8595 : SUCCESS in 1h 15m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bd9cda2d95fd49aa8e4282a025205933 : SUCCESS in 1h 08m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1c6687e9f3bc4f5b8a378653e762db7c : SUCCESS in 1h 19m 55s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ec59b19e89cd4d63a33453c0a7fd0850 : SUCCESS in 1h 09m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0a68a87d3da842e1afab5162b353ee81 : SUCCESS in 34m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a1bf1e8456ca425cbe9a45fb6d691b17 : SUCCESS in 26m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d0d8a099c5cd4cea8c0c3b9ef4921b64 : SUCCESS in 1h 03m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b11dac5050eb4d26a0b2b534118288d4 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d606dda7b4f4e05827b28c77fdc1d65 : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e714a3a4198543e9ac674e9a6bb9ff80 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/096ddb54124e47898e9c3426bd60ebda : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f8e29b949144833a33eba8fb389f0fe : SUCCESS in 8m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e9ebc24b46ee405491daca709dc350db : SUCCESS in 23m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a5633cbf8a443acb6f115478d1d6e39 : SUCCESS in 20m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0903df9220354a20a1368d5c3c3fe3db : SUCCESS in 1h 18m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d2d0893e45c48faafd52dde1d7ebfb0 : SUCCESS in 1h 04m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc7e7db16a5a4a849857205b008885d7 : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":1},{"id":"467821fb1ad46dad9a07e04e1513ecde44a26f0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 12:34:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b389c7f6f034eec84adb7cb2c763c80 : SUCCESS in 46m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e484c04f249b4c5c84b0a4c06cb8bb0e : SUCCESS in 57m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b3613feec2247d6833ff10b662b963a : SUCCESS in 1h 45m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d463feea9f04e36b1cd1b244d96aecb : SUCCESS in 58m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1739f61cd55a92ce0c0e396fabc88b521ad31aab","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-22 14:00:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"adba5e4089656db5ffe4ef88801259daa17cc52c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 14:02:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29d87c49dd5a3a751d19e64e9c049447726c333b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 15:27:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c257273009cbd7c28c343fc7a13862abd25f244e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 15:27:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0d2121f6f95c46ffb74f8eb49d64a292 : SUCCESS in 26m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/76f2392777bf4a8e87e4b7cd71c24647 : SUCCESS in 1h 13m 30s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6b83e62406954f12bb610e0f76df7292 : SUCCESS in 1h 20m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eb805fd310e4f11af504dd31a00319b : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda33af24a9c41f7804f91ec1d344ed3 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/299505479f9a4ff5868777897ee69bb5 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12fa98aa32f7473d83f12b1c06062238 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc33addd50dd4fa2910b752e3de3f73c : SUCCESS in 6m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9486566edc114dbbb18e30cf026f1795 : SUCCESS in 23m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3698f3aaf794286bf693a84f8c1fbd1 : SUCCESS in 1h 01m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4d58871c5a924382b88b4b8e6e6d39ce : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"580eb800adca6235ef8d67b238b02bd8539a9ae9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 15:29:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5b08bddde5d49d792db7275e5eb787e : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4103b0b151af100493c7242b088b42f051bf5485","revisions":{"4103b0b151af100493c7242b088b42f051bf5485":{"kind":"REWORK","_number":1,"created":"2020-06-22 10:45:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/737254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/737254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/737254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/737254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/737254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/737254/1"}}},"commit":{"parents":[{"commit":"21b5b651131303cd928ab728c358a651004d70bd","subject":"Merge \"Add collectd-amqp to the collectd image\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21b5b651131303cd928ab728c358a651004d70bd"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-06-04 13:09:46.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-22 10:45:10.000000000","tz":0},"subject":"Fix horizon debian binary docker image","message":"Fix horizon debian binary docker image\n\nDebian packages are different from ubuntu packages.\nDifferencies in /etc/openstack-dashboard:\n  - Symlinking {{ python_path }}/openstack_dashboard/local/enabled/ -\u003e /etc/openstack-dashboard/enabled\n  - Symlinking {{ python_path }}/openstack_dashboard/local_settings.d/ -\u003e /etc/openstack-dashboard/local_settings.d\n  - Symlinking {{ python_path }}/openstack_dashboard/conf/ -\u003e /etc/openstack-dashboard/policy\n\nEvery dashboard-plugin debian package is copying his policy files, local_settings, enabled to above locations.\nEvery dashboard-plugin is triggering dpkg and collect-static, compress is done by openstack-dashboard package.\nKolla has to remove all these debian package\u0027s configs and provide kolla configs.\nMove also /etc/openstack-dashboard/policy to standard location and delete symlink as kolla-ansible is overriding\ndefault policy files path to /etc/openstack-dashboard/.\n\nChange-Id: Ieca15bdb315d52e9547d798df11641ef36485b26\nDepends-On: https://review.opendev.org/733612\n(cherry picked from commit 46f8eed88a865adbbf564aae6949a0e2a39f3c31)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4103b0b151af100493c7242b088b42f051bf5485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4103b0b151af100493c7242b088b42f051bf5485"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
