)]}'
{"id":"openstack%2Fkolla~737478","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I52094ce9e3d98cbdedfdc23c8a43dd3093ce1771","project":"openstack/kolla","branch":"stable/train","topic":"bug/1884034","hashtags":[],"change_id":"I52094ce9e3d98cbdedfdc23c8a43dd3093ce1771","subject":"Use upstream Erlang on CentOS 8","status":"MERGED","created":"2020-06-23 09:36:58.000000000","updated":"2020-09-04 11:08:43.000000000","submitted":"2020-09-04 11:06:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737478-1599217595828-8a9bf4a0","meta_rev_id":"330041597876b77a7a075f5cd0c58481c7adc12e","_number":737478,"virtual_id_number":737478,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 11:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-09-04 07:13:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-09-04 07:13:58.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-09-04 07:14:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 08:38:45.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-08-05 19:41:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-04 11:06:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1427a7e8d7f272b55f4476a90e424c0f1b304cf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-23 09:36:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c73ac5562916473274c977617c5ea1fe478e721e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-23 09:37:22.000000000","message":"Patch Set 1: Workflow-1\n\nfor testing","accounts_in_message":[],"_revision_number":1},{"id":"0b055081db653171917a9f91e62171d19f01b48f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-23 09:43:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d4902da6bf036145b64b5ee5aaee8680c53d5add","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-23 10:00:42.000000000","message":"Patch Set 1:\n\ni let it settle with the current patch and later touch only x86_64 parts then","accounts_in_message":[],"_revision_number":1},{"id":"d2887b5583fb56cf928d667251a27b4e1e72de07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 11:53:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0038f9ffadf64f429a146f7642ebecd8 : SUCCESS in 54m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/aad622078af348f7961cdfd08801959c : SUCCESS in 44m 17s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9d4051c496fe4ed8bacdaacff8f3cd74 : FAILURE in 25m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3c60cf6a4ce04259b2f0dc121e425391 : FAILURE in 32m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/43774df36d04489496f22644b3bf1746 : SUCCESS in 1h 00m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f620ec8349ff43b39b6d88c6fcc65cf3 : FAILURE in 31m 11s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/678b86d3889344a9ae2b8b1c1d685378 : SUCCESS in 1h 12m 14s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f62d012221ea40f0ba81bdf5420ee83e : FAILURE in 24m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1ad40588363449ce9d567c0630f6fdb8 : SUCCESS in 1h 35m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0188db0d79ea4bc18920ce8a3895c667 : SUCCESS in 48m 54s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/8d18b9d63cc7438bb1e662c8a6bf50c7 : SUCCESS in 46m 17s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/87c90d3870c94966abb3674f066ea444 : SUCCESS in 30m 45s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/d35aaf81aa0a4199924044584c0fda2d : FAILURE in 1h 27m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/30e65d528a1b432fb45e0cacf79e6442 : FAILURE in 1h 32m 22s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/5d010084f85640b79a4b851a7ffa5b56 : SUCCESS in 1h 27m 20s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/1a07cfdedee14db4a588b9d7137f5a70 : FAILURE in 28m 10s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1be27cdd84db42d7abe721f1ece1817a : TIMED_OUT in 2h 03m 38s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/fb1bc7cba22d4a87aa943f7cf0b08b71 : FAILURE in 27m 53s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f9933e5bd9554adf8a674b7b4e5c0202 : SUCCESS in 1h 03m 54s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/5fefdcf5e37947a5bd66249c8fa5a671 : FAILURE in 26m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e4f2bbc99d0748679cf47ae884fcf190 : SUCCESS in 23m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ee0ff80d2e82435188fc86b3fe325124 : SUCCESS in 44m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5e16dd2bb41a4deb978d8797b69e437a : SUCCESS in 1h 39m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5e35156c93e476ab7868d7d7e5b9e24 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e5414bcb051438bb8fccaa11aa8656e : SUCCESS in 6m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f59d466ee50499cada735d0bf13adf9 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d440782322134e9bacf3972ca51d1287 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db77f0f75d5c457f993468700a04ad2e : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a79118e766441c696b4536d0edf95a7 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/520420a9e1f344a0afd672637b5aa6a9 : SUCCESS in 4m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1bcbd2f12bc44188ff3f24c8582490a : SUCCESS in 25m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4ddb97a320b423d9c3bb7e52d6c67f2 : SUCCESS in 26m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/badb8e426e194264a212cfa64fa1f47f : SUCCESS in 57m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db342255694440d3a3e82b58f1316c61 : SUCCESS in 55m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/19530587845e4a38a2297022bcfecdaa : SUCCESS in 1h 06m 56s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/84adff44038e4691b1696d8564669ba8 : SUCCESS in 1h 23m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/94380097230c4e8ea9785bb08c8f2fb0 : SUCCESS in 1h 46m 16s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/62897fbb152f49afa2c0da1fd5452bc3 : SUCCESS in 1h 18m 28s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a4c2dc04e6b2407e964444025e894bf3 : SUCCESS in 1h 01m 36s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"d3504c2370a3de57874dc9cb0f642c1484f9548e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-24 08:17:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c36d7e1087d4e807211d031445ac9543d98d6e4a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-24 08:20:30.000000000","message":"Patch Set 2:\n\nupstream erlang 22 dropped hipe support","accounts_in_message":[],"_revision_number":2},{"id":"60971db39ceded8d8dcd8fb0c4e574ead58b178d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-24 08:26:36.000000000","message":"Patch Set 2: Code-Review-1\n\nThis aarch64 repo is c7 only. You need centos-release-rabbitmq38 on aarch64 - like in https://review.opendev.org/737473","accounts_in_message":[],"_revision_number":2},{"id":"4acbab71047b8bb793a1179eb9cc9723f83fd64c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-24 08:27:40.000000000","message":"Patch Set 2:\n\nyeah, sorry, forgot the other issue","accounts_in_message":[],"_revision_number":2},{"id":"1829b304c02c2707b81b2c7dcb1cc5a5beef6d9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-24 08:30:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b674ac60afd2a75eabd5ac1c77a49cbd54faeb53","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-24 08:31:28.000000000","message":"Patch Set 3:\n\nstopped touching aarch64; we don\u0027t run rmq 3.8 in train either so it\u0027s either distro erlang or none ;d","accounts_in_message":[],"_revision_number":3},{"id":"5dbc6f2ced501da0b578252225db9d854a38ec3d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-24 08:38:45.000000000","message":"Patch Set 3: Code-Review+2\n\nAs I do not plan to work on getting rmq working on aarch64/train I can accept this version.","accounts_in_message":[],"_revision_number":3},{"id":"40a7a8b00f09235d0ebdc3bea4663ec2e0b3dd1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 10:40:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0a6b3d2fa6984d4ca39f920d090d7824 : SUCCESS in 57m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/52e0f96ae0644ba8a2a9722894ee9831 : SUCCESS in 50m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6171b7495a1d4cf497fc4e5776c8ae4f : SUCCESS in 24m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fcf8df0633824105b6db27ddbd23af20 : SUCCESS in 31m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a348801ef970494fbab06e7275603a0d : SUCCESS in 1h 27m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b28746b6ed404a9abcf4391e34cb2886 : SUCCESS in 1h 10m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0e385d99aef84a3c82871a42d532a847 : SUCCESS in 1h 20m 37s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f2cf7b8c2a7f468bb14bd3978f64fa38 : SUCCESS in 1h 30m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2db6660161e34e9f8113017baab527a8 : SUCCESS in 1h 45m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/29d60e138ca84ab58648ed2aed988faa : SUCCESS in 59m 42s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/53ae543ed32a47c086298d3ad286b90b : SUCCESS in 48m 21s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/f292334e74774885b9c8e35e25c095b3 : SUCCESS in 31m 03s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c30f4e9aca32411e98dc3f151fa70541 : FAILURE in 1h 21m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9411c45e30874da29d4aba33a76ff58e : FAILURE in 1h 01m 10s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/b2354b336fd3414aa7da8595b4589551 : TIMED_OUT in 2h 03m 16s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c8b8009a633b4fa9a4a8f50c78ab414c : SUCCESS in 1h 46m 25s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c0f9ad0c4aba404fa4be153c8c2724e2 : SUCCESS in 1h 58m 51s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/eda12d0a24154e5db74d4069a96901fb : FAILURE in 1h 21m 18s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/698243ea8d184b6da262123da19eaa3a : SUCCESS in 1h 43m 30s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/1f77c6393b0a484a967a443cab183c13 : SUCCESS in 1h 34m 00s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ccbe3a7af5c44b1f913f56b4275fffd0 : SUCCESS in 23m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0ffcb98bd3d34266b20bc4f41fd8d836 : SUCCESS in 23m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/72c0947ff0c940c5b2cdcb02e5785588 : SUCCESS in 56m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99657badd3154994bf8622ca6f296cdf : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b3d44dca08479686cebd433f81268b : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3cb3f349caf94a829dc55ff0901db9c9 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12405ea8a3724a8298b568b078ec23c5 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f402f9030ef4fb6b7f1836d96c4cbcb : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be39ffa57ff947649da0215e452b65cb : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9531da07b6f4b76b175702ae19cc090 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/01728dc708f34ba88bcdf5bdddf67c28 : SUCCESS in 25m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cabfaf30c394ec3b66d1e5156ac36ca : SUCCESS in 34m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/29e9137c5d1b439f8a4728c7543c0808 : SUCCESS in 1h 03m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e06592b957e483d844878595ff8d4f6 : SUCCESS in 53m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/474cc736ccb0474d9be0ce567e6ea431 : SUCCESS in 1h 10m 07s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/c577338a37754c16988260dfc5b71672 : SUCCESS in 1h 21m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5f7a1018e0c14d47ba70e5679d310285 : SUCCESS in 1h 44m 36s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/16482ea15f23455fa946bce2b995f7b2 : SUCCESS in 1h 20m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7609aea631a7422eb1f675293bf47d96 : SUCCESS in 1h 06m 45s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"1cd51d6800453508010c32c6e5ab8e95a0bda135","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-24 12:21:36.000000000","message":"Patch Set 3:\n\nrmq seems happy still, wonder if it fixes Stig\u0027s issue","accounts_in_message":[],"_revision_number":3},{"id":"ca1169040490f0be18f4acb5c93e0aa375b1b0a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-10 17:01:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98743c9278db3cfb0e32916bd6c9964175068bbc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-10 17:23:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"78a2c21033b28ae4f699ffee5d8f4b463cf75bcb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 08:32:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"68a298e0ee0f90e6b51c0fc745129e34edf32fe6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 08:35:33.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"e8dc7be7a56d293b19634d48dcdc566343cb2168","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 08:46:14.000000000","message":"Patch Set 3:\n\nProposed a patch for tripleo: https://review.opendev.org/742613","accounts_in_message":[],"_revision_number":3},{"id":"309c554a92940026860db91a39cd3b98fc26f19d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-05 19:41:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9b1d329ba0e271208379bd58cb3511755fc6a667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 19:41:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9e7711a9f53f7f5e87e42d333f68ac0660230545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 21:23:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1bc46c4e217e466bb156156a5c91d90a : FAILURE in 54m 52s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/af5609a8c20b4bafb19b7fcfac315970 : SUCCESS in 32m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9b736d6f69a841769de0865b9f3f530a : SUCCESS in 1h 32m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6dc78a0d80564c6393c7cf19c8d05a67 : SUCCESS in 1h 07m 58s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a4371a1cdadc4969af0d2bafd843f6ad : SUCCESS in 1h 36m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7a94535b58b44d73ba8579c79c984673 : SUCCESS in 55m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b820dfc7950a464fbd5f292d952c699f : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db43d5dbc2354041805619b90413abdf : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8623b22052a743468108270f3ae40e13 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83bad1bcb2034f41bf585ebb8c55d201 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a572b961e87a403fb068e85d94e02737 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d5956f05ba445e869901e34a3d1873 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3f2a76fc6da4603b680f2a03dabece5 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d280ef40b0ca40cca8a93a78f9262574 : SUCCESS in 26m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/32a97f3796c446639dc58f991cba71ff : SUCCESS in 54m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/38356a7b10d443318c5c6e1e29af5095 : SUCCESS in 1h 12m 31s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"bfad0f588f9960c706eb3cabd16782492c5e571b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-06 09:05:07.000000000","message":"Patch Set 3:\n\nINFO:kolla.common.utils.bifrost-base:\u001b[91m: \u0027ascii\u0027 codec can\u0027t decode byte 0xe3 in position 25: ordinal not in range(128)","accounts_in_message":[],"_revision_number":3},{"id":"3debf5e34a08f6f510320fe4865044f9d4b5a37d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 18:14:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1b2cefeeff7ba630e1229f2f971658d8afaff27f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-03 18:14:35.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"437c859cd89c39c098026126309db25015fddab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:41:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/de95418e2fee49d6b11a18fef0d645ad : SUCCESS in 40m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ee7233d3c2574aaeb305091be98ee175 : SUCCESS in 1h 07m 37s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cf4b0b309b934ad896cb9d05133d2fe4 : SUCCESS in 41m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3e9aaa335e4348e5850ddd0e45fe4373 : SUCCESS in 55m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6a53de07531048159808f660917456f6 : SUCCESS in 1h 29m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/451d72349e614ba38f3b98d80602eb14 : SUCCESS in 58m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dbd43347e1bc449a94923a279c7a0f8a : SUCCESS in 1h 16m 28s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bab61c7e319f4786a745b07ebf4ac5a9 : SUCCESS in 1h 04m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/613e20b2edf24681a968d33b4b1c12aa : SUCCESS in 1h 38m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ed09923fb6954e98a89b0be57d26eed2 : SUCCESS in 59m 44s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/8f44059cc48e4f60bb5d6ecdb2c53713 : FAILURE in 42m 55s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/02669492d50e4a3aa1d395659579a5ff : FAILURE in 34m 32s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/f388f9ae4c724bf694c0d51c2807f48c : SUCCESS in 1h 17m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a6bb5d3fe1d64d03956b732535befdbb : FAILURE in 1h 27m 53s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/4e60a256fdee4f0b985a5c49a7138465 : SUCCESS in 1h 58m 27s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/02ff480184cd4a278145c7980450b7f1 : SUCCESS in 1h 16m 52s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/bae2a064d67a44f9b18d946c03d7ff2d : FAILURE in 41m 12s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5d1e22ca004f467aa04b57c9c358c65b : FAILURE in 1h 28m 22s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/ee1f0e7594144604822aa3df48962d12 : SUCCESS in 1h 23m 35s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/505d4d3e94224223b01ac67ca8a886b1 : SUCCESS in 1h 00m 59s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/386568bf42c54e5ca048959918c69168 : SUCCESS in 22m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ea97d9d1d959494c87c5c76afa284c3c : SUCCESS in 18m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fc96735a0d084d90a40a748f3a0aeab2 : SUCCESS in 51m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/431b61a746834d7187c7ade9e80700b5 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfbaf68d685c4da88004c98481119060 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99c8821caed7429696d78f28517c380e : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d4a573ce03b499781742012a93c3c3a : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4499a5de459c4990a6a45e54f1be3801 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da504e4d6dc48d7a2974de0371b63b2 : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/394cb6f2f5f74a2a932f22e850c9896e : SUCCESS in 4m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b36c0e3531f945ca919f475b144bda93 : SUCCESS in 34m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcf7ac46b1a94f7baa105199931651c7 : SUCCESS in 24m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df60ec76796944998ad6da8e22302fe3 : SUCCESS in 52m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7be6f001028e49408cde101b5012ea25 : SUCCESS in 54m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1d2704b08c0544af87499cd799eeda7d : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/8c23c3dc0b76445098d598fe75be5a94 : SUCCESS in 1h 10m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3afdc39b7c7347bfba44789fbc50a822 : SUCCESS in 1h 41m 41s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/1e4b4314e73248f89ee054bb761e2f7d : SUCCESS in 1h 24m 00s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/90b0da07aacb4ccd943efc60fd686aa5 : SUCCESS in 51m 05s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"ed7a04b5d453a9095f166f52fce271c497a5d884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:54:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3314e98e4a398632561e7bfe74b8af319e94eb4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 23:37:13.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"9844453e2c1b235012fd44b30bd02e257f3e787a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-04 07:13:58.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"69fad504021d85153e6ca553a826bc5743030513","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-04 07:14:01.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"81f29916e9edbb7bee6d8aca59d7de30bf87cc6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:21:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fd64ba3181a6453e9c5f66912f90ac71 : SUCCESS in 39m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8e128483245742539733f1d42549c709 : SUCCESS in 43m 50s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ea980db5ade740f4b1a87258c28f0aef : SUCCESS in 21m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e4e6ef3791d9439aa21384d492179d63 : SUCCESS in 35m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/df663de3df834d8c94e507c976c53f21 : SUCCESS in 1h 20m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/403bd88c7d074ad587516277ff45eaaf : SUCCESS in 1h 19m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5d683596e3334367bbfbe418b4d6dc62 : SUCCESS in 1h 08m 30s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/36103f68bdec42439c4dfe12a7b2e1d3 : SUCCESS in 1h 02m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/534f686dba2e420bac936f5e04eac98b : SUCCESS in 1h 22m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fef35066bb084c4a80a05414a0981961 : SUCCESS in 49m 25s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/432bdb5a2a3044aca1dc8a3a63682297 : FAILURE in 33m 12s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/00e5e0e48fbd4bfab3120d18d4ad77db : FAILURE in 23m 47s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/069a33f3e4924c808d6415359c5c2b3e : SUCCESS in 1h 36m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f7a26ac2cc2945188d936a9ab47b8f35 : FAILURE in 1h 26m 54s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/b9df8799ee984eaca8824bd7a9366f35 : SUCCESS in 1h 50m 05s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/4724933e8af1405b9adbf8f9ac5c4c25 : SUCCESS in 1h 27m 30s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c6f2a2e957974aa1abdbf57876616c92 : SUCCESS in 2h 00m 29s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/e1201851a21f4308b5deea8d6bb4ed7b : FAILURE in 1h 51m 33s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/450a075936544fcc9f315deed1f7068b : SUCCESS in 1h 11m 53s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/abed2af09bbe4d57aa8e956c1b0ddc00 : SUCCESS in 1h 21m 46s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/49afe90c81764ea88b04b9f36acd3f1f : SUCCESS in 18m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8f4e2424e344483eb9c592e706624a1e : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f35854c39db044ec925a7663d594ede4 : SUCCESS in 58m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef4ed5cf508b42188145198c095569ce : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30d7140c7424306be5174a28281f72f : SUCCESS in 8m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61560a6d090b472282ede2b90b4dc1b8 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c2f6f2c94a6477e8269c6c961362a91 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9ae525fc3bc4b19abef180f410d94f6 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c7fa757ca842caacfcc16bbcb8202b : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf1f5fbd96994dbbb16e29f4bd536ff9 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4705f472ab046eda0814f300a01f415 : SUCCESS in 29m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7abb57b4f4844dc838b5de34d01bbdc : SUCCESS in 26m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6bfbc40fe2a41728d552a82b954a77a : SUCCESS in 51m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06713af82afe422db79b142e932459e0 : SUCCESS in 59m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cfdc1adaf757428c8611fb675a60f57d : SUCCESS in 1h 01m 03s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/82e29e585b3541c5b515bebb3b9b46f9 : SUCCESS in 1h 18m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b97cd0ff24d94d28be918e692d3b30a7 : SUCCESS in 1h 43m 16s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d92c825c24794b9aa4550f4b6c23e2b4 : SUCCESS in 1h 02m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/e062190f37d942e2bb1d82705957f64d : SUCCESS in 1h 05m 23s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"72a23925726b77bb4966a8f7321aa240b6d9e99b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:21:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b05bf37b9036d606787d1a5ad7df5042f22fbcfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:06:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"582d757ccf02234db7aa66a1c375c9e1c358407e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:06:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1520f0e3c9a24814bd2b8d0432d28ef3 : SUCCESS in 1h 43m 11s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a99b569574dc41b79ba2aa3a45e0edad : SUCCESS in 55m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fd4195666533454fbe63c12109ab8df7 : SUCCESS in 1h 06m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7688633974b247e589343564d2cad06c : SUCCESS in 1h 01m 16s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0898b096b1564254836c1db446827ca1 : SUCCESS in 1h 34m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f9595fd80efe4e0f88575853d01f84dc : SUCCESS in 47m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9df2e3d74374480aa1db1889c15f4e22 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa027a22d906459a9d1cac033f0622d5 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f112edd8eb9f49378719dc3e3aa92c72 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df38214bf4044120b2e1fc8335f85c2f : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9ea1f13405f445cb8ed758c4faa4020 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991744bce3be4a039fb5fdd8270e9fa3 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf491baf8721452e97f7b69efaaa9444 : SUCCESS in 4m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f42b6e140d7f4574b9c8dca352fb9040 : SUCCESS in 26m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36b6b85b5f4d45fb9135a7dda2802702 : SUCCESS in 1h 12m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/06f7341abc544872ae68fea54e013db9 : SUCCESS in 1h 25m 55s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"8a53e445169ef1d3a12cacf160e7d5ee5ea6e718","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:08:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862c3afa91044457939f906d04ac7040 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b5afe302d814427b599836db4ff67ac : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fec2c701dc36ea8d246b69290f5546bf37d414b3","revisions":{"6d8b0b5a2e9a21b386fe9698edb4f436ae393425":{"kind":"REWORK","_number":1,"created":"2020-06-23 09:36:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/78/737478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/737478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/737478/1"}}},"commit":{"parents":[{"commit":"381637bbaca5b00e369e50558f90c1fcd4f77114","subject":"Merge \"Bump versions for Train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/381637bbaca5b00e369e50558f90c1fcd4f77114"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-23 09:34:09.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-23 09:34:09.000000000","tz":120},"subject":"Use upstream Erlang on CentOS 8","message":"Use upstream Erlang on CentOS 8\n\nTrain only.\n\nIn CentOS 8 Train we had a mix of rmq upstream and rdo packages,\nerlang was from RDO and rmq (3.7) from rmq.\nAs this is not really tested by any other party, I propose to\nunify the sources of both.\n\nChange-Id: I52094ce9e3d98cbdedfdc23c8a43dd3093ce1771\nCloses-bug: #1884034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d8b0b5a2e9a21b386fe9698edb4f436ae393425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d8b0b5a2e9a21b386fe9698edb4f436ae393425"}]},"branch":"refs/heads/stable/train"},"0d9bf85aea28f9c0a87409d051f4dd95fa7dc451":{"kind":"REWORK","_number":2,"created":"2020-06-24 08:17:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/78/737478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/737478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/737478/2"}}},"commit":{"parents":[{"commit":"381637bbaca5b00e369e50558f90c1fcd4f77114","subject":"Merge \"Bump versions for Train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/381637bbaca5b00e369e50558f90c1fcd4f77114"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-23 09:34:09.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-24 08:17:47.000000000","tz":120},"subject":"Use upstream Erlang on CentOS 8","message":"Use upstream Erlang on CentOS 8\n\nTrain only.\n\nIn CentOS 8 Train we had a mix of rmq upstream and rdo packages,\nerlang was from RDO and rmq (3.7) from rmq.\nAs this is not really tested by any other party, I propose to\nunify the sources of both.\n\nChange-Id: I52094ce9e3d98cbdedfdc23c8a43dd3093ce1771\nCloses-bug: #1884034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d9bf85aea28f9c0a87409d051f4dd95fa7dc451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d9bf85aea28f9c0a87409d051f4dd95fa7dc451"}]},"branch":"refs/heads/stable/train"},"3cb44bd14e84d96f696094a0a688508a712d6714":{"kind":"REWORK","_number":3,"created":"2020-06-24 08:30:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/78/737478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/737478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/737478/3"}}},"commit":{"parents":[{"commit":"381637bbaca5b00e369e50558f90c1fcd4f77114","subject":"Merge \"Bump versions for Train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/381637bbaca5b00e369e50558f90c1fcd4f77114"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-23 09:34:09.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-24 08:30:24.000000000","tz":120},"subject":"Use upstream Erlang on CentOS 8","message":"Use upstream Erlang on CentOS 8\n\nTrain only.\n\nIn CentOS 8 Train we had a mix of rmq upstream and rdo packages,\nerlang was from RDO and rmq (3.7) from rmq.\nAs this is not really tested by any other party, I propose to\nunify the sources of both.\n\nChange-Id: I52094ce9e3d98cbdedfdc23c8a43dd3093ce1771\nCloses-bug: #1884034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cb44bd14e84d96f696094a0a688508a712d6714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cb44bd14e84d96f696094a0a688508a712d6714"}]},"branch":"refs/heads/stable/train"},"b689b60746605d3cefb655afc5f7b8976441fce8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-03 18:14:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/78/737478/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/737478/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/737478/4"}}},"commit":{"parents":[{"commit":"bfc8ce085b338df91eabd55b833490e3ec2a02e2","subject":"Upgrade Fluentd Monasca plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfc8ce085b338df91eabd55b833490e3ec2a02e2"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-23 09:34:09.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-03 18:14:31.000000000","tz":0},"subject":"Use upstream Erlang on CentOS 8","message":"Use upstream Erlang on CentOS 8\n\nTrain only.\n\nIn CentOS 8 Train we had a mix of rmq upstream and rdo packages,\nerlang was from RDO and rmq (3.7) from rmq.\nAs this is not really tested by any other party, I propose to\nunify the sources of both.\n\nChange-Id: I52094ce9e3d98cbdedfdc23c8a43dd3093ce1771\nCloses-bug: #1884034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b689b60746605d3cefb655afc5f7b8976441fce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b689b60746605d3cefb655afc5f7b8976441fce8"}]},"branch":"refs/heads/stable/train"},"fec2c701dc36ea8d246b69290f5546bf37d414b3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-04 07:13:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/78/737478/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/737478/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/737478/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/737478/5"}}},"commit":{"parents":[{"commit":"2ff242c4a270258fed0a0dbdbd3b66a402cadece","subject":"Bump up versions for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ff242c4a270258fed0a0dbdbd3b66a402cadece"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-06-23 09:34:09.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-09-04 07:13:58.000000000","tz":0},"subject":"Use upstream Erlang on CentOS 8","message":"Use upstream Erlang on CentOS 8\n\nTrain only.\n\nIn CentOS 8 Train we had a mix of rmq upstream and rdo packages,\nerlang was from RDO and rmq (3.7) from rmq.\nAs this is not really tested by any other party, I propose to\nunify the sources of both.\n\nChange-Id: I52094ce9e3d98cbdedfdc23c8a43dd3093ce1771\nCloses-bug: #1884034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fec2c701dc36ea8d246b69290f5546bf37d414b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fec2c701dc36ea8d246b69290f5546bf37d414b3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
