)]}'
{"id":"openstack%2Fkolla~737831","triplet_id":"openstack%2Fkolla~stable%2Fstein~I6068abcbc0e73eeab77b0aa99983a95adc6defd3","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I6068abcbc0e73eeab77b0aa99983a95adc6defd3","subject":"CI: Install python deps, mark oraclelinux rsyslog as skipped","status":"MERGED","created":"2020-06-24 15:27:29.000000000","updated":"2020-06-26 08:59:33.000000000","submitted":"2020-06-26 08:57:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737831-1593161854654-a308ddd4","meta_rev_id":"7e0c3e7b6a70b54b613942e908c57fc5c10df9cd","_number":737831,"virtual_id_number":737831,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 08:57:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-06-25 13:30:12.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-26 07:51:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-26 07:51:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-06-25 13:30:12.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 13:30:12.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-26 07:51:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-26 08:57:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"825857533ef7ebbc099173a2ec2ca2fecf93614f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-24 15:27:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b3fff34800b181aaa4921471e07487221f3471a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 15:54:47.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e3c5fb6533cad5887ef56b2eef48dd5379078989","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-24 17:42:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93ec018cc9d19c25443a14e5890e875b98404ec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 19:12:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ef3b95a1e98149bea2ea60525b82af11 : SUCCESS in 55m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d6ace90491334ce7ad6555edc2ba5360 : SUCCESS in 49m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b02a4514e8ac492fa954e45072da4d76 : SUCCESS in 1h 20m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/659c45b70b9c44b8b667b437d9c3f89c : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/33e5743fa202438196a74e76c44c7dae : SUCCESS in 1h 21m 13s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/32bee3693aa54a759c14907487f21c9c : FAILURE in 1h 00m 59s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d43a96fe4e3a410f8a129831b487920a : SUCCESS in 54m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/959cdef0edc140d89ee66322e7dc5aa0 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27231434587841f3b665316288d608fe : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ace6722b95f94616880a6cf77b495b14 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/252ac53bed434dfc8a488566ed644ffb : SUCCESS in 7m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/958e03246850479f9c251f07161ca396 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2289fe107e6493db9752ebf203e45c4 : SUCCESS in 4m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f71244bf2a44b41a441a1af9d969da3 : SUCCESS in 23m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/47a0f5f282954310acdd6e9603fc8a4f : SUCCESS in 34m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/26b68ec50b1b4fb9a2b8c2e2acc049ec : SUCCESS in 1h 10m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3835a00757454cb58bb8c2729de4e824 : FAILURE in 32m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/97205c88ccc04bd68a206e7c68883d15 : SUCCESS in 1h 01m 36s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"67bb43999684540b1c73a6ad1f9c84073b4781eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-25 08:48:19.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"50aef129bed332e92c9409e592a6ace6e5c6abc2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-25 09:06:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"695a49124425f6d9b8e5e0a7add1c82bbb1dac19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:39:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d09b3487d6f24855917eb6781bd0639a : SUCCESS in 49m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ca41ba151c31407c8244a1845ea9dc40 : SUCCESS in 57m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/693954145dec45fc88d2c80bab9b9fe4 : SUCCESS in 1h 29m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/91dc8716902e4b2b9e02290998d133d6 : SUCCESS in 1h 08m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/356d47eccf70446cba5d59d8589232be : SUCCESS in 1h 49m 21s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/78486ffbc77a4cfdac05aa73d6fd371f : FAILURE in 1h 12m 38s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/49ad525b1d01483e9030c4773cf4e770 : FAILURE in 58m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64a33fcc85174abdb03df8e0a455169c : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b6a108895d44903b0915b66492f2862 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4bd97fe4628e413a818f6636657f1c71 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9433773e7a074390ad99c7f23005615f : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4df3cf65ef0d4a09943f0d68b04155ec : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900b7b0f67634ea592b9d1ba32fdd57c : SUCCESS in 6m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b409c6d7afb40dc925f6f8660e4bef8 : SUCCESS in 1h 05m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/006d7b5640094010be7199f7e22bdcfc : SUCCESS in 31m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/33fa0b38af9e481b950d02d677346cf0 : SUCCESS in 1h 06m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/00e3f23f93dc43a39cbd42ef64d52ff6 : FAILURE in 40m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dcfe788c5b0f4e9f8d2ad3839e43336f : SUCCESS in 1h 01m 00s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"9bdf8fe2429463171d5732bed6eb6a295c476eb3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-25 11:03:23.000000000","message":"Patch Set 2: Code-Review-1\n\nOh good, oracle linux issues","accounts_in_message":[],"_revision_number":2},{"id":"07ee4bdba30441608b4cde5eb48113d7827d697f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-25 11:19:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8a5853d44ed60dac78a7dc1b0a1b5fc5672fc1b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 13:16:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/06520e1d6cd34d748b589ce17ea653dc : SUCCESS in 34m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/45b6fe73f04740ac82c067d50ace6502 : SUCCESS in 35m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c92761dcb2ae4e179aee30d25d54fe03 : SUCCESS in 1h 31m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b955592481ab4fb29ebbb544cfd9ff64 : SUCCESS in 1h 07m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9be4920da6434bae8547cc6417ac8a57 : SUCCESS in 1h 40m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/55513566d344476180f7ae760e8c2b67 : SUCCESS in 1h 05m 35s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bc4de99b3ad44141a284c8cf3dc09d3d : SUCCESS in 35m 43s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/83cbc61ee84c4821bca2ec21b85e4bed : SUCCESS in 1h 12m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7269d44db2c548cab5374fb8a3f75de2 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082775538afc4944bad32ff8cfb5862d : SUCCESS in 6m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a60a1898e26945058668b51eb62cbcbf : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d179714645d74c6181c03b63cbce4188 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6fe479621ae545fa8f76f33e91ce37ce : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59b11daa35434be28641acf91c510555 : SUCCESS in 4m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a0c3831de2a42eb8dc6a2d3b0752302 : SUCCESS in 24m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4cd86c5d90640f79c7a5fab98112174 : SUCCESS in 34m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e3e6822857f4fe59453bc4c5378407b : SUCCESS in 1h 02m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3334cf6eb0a48ebb3483b7b45adee8b : FAILURE in 33m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2f629241ed7246db858c18da8851cb88 : SUCCESS in 59m 45s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"dbe7116be3722c95f3af411e0ec3ca1a08fb2ced","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-25 13:30:12.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"0e088d70197d4f6542e0ff9c746f46609e9cfba1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 07:51:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"87f7aa32a2f198d1f012cfa4d7785135a9535fb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 07:53:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3994220d4ddc0254dd42736b886c9baf77ba19d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:57:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"be8f83e847d26a2d3797c62fda3305f10329106d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:57:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5eb3e74179d54723aa1b761ad7fffa8f : SUCCESS in 30m 05s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/bd0ea158e58c4f2d95e978445eaccd1c : SUCCESS in 30m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d35eb3e86c742f3880a9cedf3c8e205 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c96b7397ad24455b4aaa7b6bc73b4f7 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b561d3e2746641e18d7c054220e01c19 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/091da4b018a244c18dbf7134e36c15b8 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83513d951fa94d488a07ab6d3c65fd5c : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b05b0666e2d64148b7abddd07f97a36f : SUCCESS in 4m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fbb250435b7e42649d3e8520f5367fa6 : SUCCESS in 28m 08s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"7e0c3e7b6a70b54b613942e908c57fc5c10df9cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:59:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9309481eb340098dba0318eb7b836e : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d12137ce64fab7e8436aa1c2e32e3bc6ed47c1a1","revisions":{"62e46d23eaed074f1639232626ea8fe299e4866f":{"kind":"REWORK","_number":1,"created":"2020-06-24 15:27:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/31/737831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/737831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/737831/1"}}},"commit":{"parents":[{"commit":"8cce9293a42afca7a99f32090d4f55b1cf7310a7","subject":"Bump versions for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cce9293a42afca7a99f32090d4f55b1cf7310a7"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-17 11:13:09.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-24 15:27:26.000000000","tz":120},"subject":"CI: Install python deps, fix mod_wsgi in cyborg-api, disable masakari-monitors","message":"CI: Install python deps, fix mod_wsgi in cyborg-api, disable masakari-monitors\n\nsetuptools is missing from infra image - so it needs to be installed.\n\nCopy cat of a kolla-ansible change: https://review.opendev.org/#/c/735808/\n\nFixes also cyborg-api to use py3 mod_wsgi in centos8\n\nTemporarily disables masakari-monitors build, because it fails on centos8\ndue to problems with libvirt-python\u003d\u003d\u003d5.7.0 (upper-constraints):\n\nMissing type converters:\nvirTypedParameterPtr:1\nERROR: failed virConnectSetIdentity\n\nDepends-On: https://review.opendev.org/737829\nChange-Id: I6068abcbc0e73eeab77b0aa99983a95adc6defd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62e46d23eaed074f1639232626ea8fe299e4866f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62e46d23eaed074f1639232626ea8fe299e4866f"}]},"branch":"refs/heads/stable/stein"},"e8bc3de316371bdcc416f99e71352c38b96eb5a7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-25 08:48:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/31/737831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/737831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/737831/2"}}},"commit":{"parents":[{"commit":"8cce9293a42afca7a99f32090d4f55b1cf7310a7","subject":"Bump versions for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cce9293a42afca7a99f32090d4f55b1cf7310a7"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-17 11:13:09.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-25 08:48:14.000000000","tz":0},"subject":"CI: Install python deps","message":"CI: Install python deps\n\nsetuptools is missing from infra image - so it needs to be installed.\n\nCopy cat of a kolla-ansible change: https://review.opendev.org/#/c/735808/\n\nDepends-On: https://review.opendev.org/737829\nChange-Id: I6068abcbc0e73eeab77b0aa99983a95adc6defd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8bc3de316371bdcc416f99e71352c38b96eb5a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8bc3de316371bdcc416f99e71352c38b96eb5a7"}]},"branch":"refs/heads/stable/stein"},"d12137ce64fab7e8436aa1c2e32e3bc6ed47c1a1":{"kind":"REWORK","_number":3,"created":"2020-06-25 11:19:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/31/737831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/737831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/737831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/737831/3"}}},"commit":{"parents":[{"commit":"8cce9293a42afca7a99f32090d4f55b1cf7310a7","subject":"Bump versions for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cce9293a42afca7a99f32090d4f55b1cf7310a7"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-17 11:13:09.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-25 11:19:19.000000000","tz":120},"subject":"CI: Install python deps, mark oraclelinux rsyslog as skipped","message":"CI: Install python deps, mark oraclelinux rsyslog as skipped\n\nsetuptools is missing from infra image - so it needs to be installed.\n\nCopy cat of a kolla-ansible change: https://review.opendev.org/#/c/735808/\n\nDisables rsyslog build on oraclelinux (both source and binary) and\ntripleoclient (binary).\n\nDepends-On: https://review.opendev.org/737829\nChange-Id: I6068abcbc0e73eeab77b0aa99983a95adc6defd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d12137ce64fab7e8436aa1c2e32e3bc6ed47c1a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d12137ce64fab7e8436aa1c2e32e3bc6ed47c1a1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
