)]}'
{"id":"openstack%2Fkolla~737927","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I014c9a52b2f61a86e3a8adafdf4783c3f58148ef","project":"openstack/kolla","branch":"stable/train","topic":"bug/1885104","hashtags":[],"change_id":"I014c9a52b2f61a86e3a8adafdf4783c3f58148ef","subject":"Fix Monasca Agent Statsd service under Py3","status":"MERGED","created":"2020-06-25 08:51:43.000000000","updated":"2020-06-26 12:35:30.000000000","submitted":"2020-06-26 12:33:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737927-1593174804491-995c8465","meta_rev_id":"ac2af73d36e22562edaedd8bcaa248505b11d7d3","_number":737927,"virtual_id_number":737927,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 12:33:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-26 10:47:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-26 07:50:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-26 10:47:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 07:50:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-26 10:47:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-26 12:33:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c70f1127407b568326613aaec02e1289f6a541b4","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-25 08:51:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"239ebae81b09ccd0d91d1754d5658dd3ca7b6502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:34:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/dd95cf23d2a248ab89f9803ecdfc54da : SUCCESS in 57m 00s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/21fa29a70e864ed3be4cba237dde02ae : SUCCESS in 39m 55s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3760209f07e245189879339e4a61bad7 : SUCCESS in 37m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4ce8a73c4c264539a74ed0c2b7b3cc56 : SUCCESS in 40m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d3f5ddc7708f479ead18f24df625d243 : SUCCESS in 1h 28m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4a0196af2bb3425d85b3895e84fb921e : SUCCESS in 1h 25m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c590f36ca7d74448b077fc5d5530b47c : SUCCESS in 1h 23m 27s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b39b70389ba9434c8c97c807a67f05f7 : SUCCESS in 1h 14m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/77f5417dde394fa7885726ec17555d92 : SUCCESS in 1h 36m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/182e2778833e4a10be75cb834e6a6341 : SUCCESS in 1h 08m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2bb9d6b78b854b7eaa0ba6931c07a264 : SUCCESS in 21m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5aaa4d3419ab40119212139f8c42502f : SUCCESS in 30m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/52043601bf574b4b8ba6e6b4c0d8ff1e : SUCCESS in 1h 07m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d41ee1bdce34e0786fe180e6058a298 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50dda00e95a043b4a6e50f499f1a8be2 : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6b88b435147446c878607533572ae90 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/022246e9c1d34fd9baf39ae60943b5c1 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/502efa7f7cec437786f97094dc9a5410 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a1deca0a3b4465b3f06e6edd9f4d23 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d001c222135c45c5b431f4a2caac8be9 : SUCCESS in 4m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/11aea3c7e05c443885762f2d4c659563 : SUCCESS in 21m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ebc747376eeb42d1ab7857f6d5f69e58 : SUCCESS in 49m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fed9a8021af543f4875acdb6824c383f : SUCCESS in 1h 05m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2b2c4e487d449ba88e2022228c25680 : SUCCESS in 59m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ae6871a82ebc4f7c9ce29eec9983b9cc : SUCCESS in 1h 13m 45s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"124b4f56a4bddd213120ade489e5513a53bb45bd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 07:50:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff6a3c1465540c33f61f69cc5da5182ca013fd15","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-26 10:47:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04c8da04b9c462b721df9a1ef1591b92e3b19e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 10:47:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41fa63edffb0ce80612a54cf2330806682198656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 12:33:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/844b1dbc51b549c6b7ecdd2c46888b1f : SUCCESS in 58m 27s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/53f6586dc4bb4270b104b8c7258cdaa3 : SUCCESS in 32m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/59e92633b3f6451f87423156c26ff220 : SUCCESS in 1h 32m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bc309fce99534a8084ec9ecff1e42b3a : SUCCESS in 1h 21m 56s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9f0060fadaf146fc966151fd0f79afba : SUCCESS in 1h 40m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cc80074612aa4d1aaf6d678a71ebe15e : SUCCESS in 43m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0452b785c0b44963a369389cd6522ac9 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d103ad190fb48f8aa2968f74fe340f2 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3774711b37c343cfa02ce749a14ba7d7 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0245ebc12f947208f2314f02da46e4d : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/704c7d5da59f417ca66bc2f6c6bf6e58 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0967bc9cdb41388abec895c5fb2703 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5db737849f384597b30865548691e3af : SUCCESS in 5m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d85067e68f3479c86b361a5c34f1db1 : SUCCESS in 28m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4672142225d4054b166ffd070e44611 : SUCCESS in 1h 04m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5136b65d24dd4e3ab3f441c3dc60963e : SUCCESS in 1h 20m 16s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"e8f0f54dc09069892d71fdb33d844949180fe6e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 12:33:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f23a113ae63e9138b4f8feec15473938f15b11e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 12:35:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754f78b457194056a0cdd34b22ae5e97 : SUCCESS in 1m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baa67a8698dd42c989718664fb484965 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"151a0e15ff31cfa751bb4c3f8ecc21345ba54851","revisions":{"151a0e15ff31cfa751bb4c3f8ecc21345ba54851":{"kind":"REWORK","_number":1,"created":"2020-06-25 08:51:43.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/27/737927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/737927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/737927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/737927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/737927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/737927/1"}}},"commit":{"parents":[{"commit":"0a6cd606c68281b6ef8301ab7a8e954440a6714a","subject":"kolla-toolbox: use u-c while installing virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a6cd606c68281b6ef8301ab7a8e954440a6714a"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-06-25 08:41:45.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-06-25 08:46:26.000000000","tz":60},"subject":"Fix Monasca Agent Statsd service under Py3","message":"Fix Monasca Agent Statsd service under Py3\n\nPulls in this fix:\nhttps://review.opendev.org/#/c/736610/\n\nPartial-Bug: #1885104\nChange-Id: I014c9a52b2f61a86e3a8adafdf4783c3f58148ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/151a0e15ff31cfa751bb4c3f8ecc21345ba54851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/151a0e15ff31cfa751bb4c3f8ecc21345ba54851"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
